)]}'
{"id":"openstack%2Fkolla~785569","triplet_id":"openstack%2Fkolla~master~I27dfe34ade088228d71f6857ec4f3ab28ee63915","project":"openstack/kolla","branch":"master","hashtags":["cleanup","drop"],"change_id":"I27dfe34ade088228d71f6857ec4f3ab28ee63915","subject":"drop leftovers of RHEL support","status":"MERGED","created":"2021-04-09 07:37:26.000000000","updated":"2021-08-04 13:55:03.000000000","submitted":"2021-08-04 13:53:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"785569","meta_rev_id":"b48544a6911b1a05c766b0208d1eb02cabdf12a6","_number":785569,"virtual_id_number":785569,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-04 13:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-08-04 12:39:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-04 11:19:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-08-04 11:18:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-08-04 12:39:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2021-08-04 13:55:03.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 08:12:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 09:29:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-06-29 18:51:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-08-04 11:18:41.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"885663eb62520089c284970afe058d02c869db77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 07:37:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef396b4a9b1308eb39b33af51a23d9b9c95ff138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 07:48:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7fe678ca5f40f70360e01a58a162fed5f49974cb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 07:53:54.000000000","message":"Patch Set 2: Review-Priority-1\n\nThis will be discussed during Xena PTG.","accounts_in_message":[],"_revision_number":2},{"id":"c5bcccd4f62daab10bae90f2f0a37f2ceab605ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 08:12:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/243be40d7dca4842954a526a99d7ddcb : FAILURE in 13m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7ad635fe192e4042a443f9baffef298b : FAILURE in 16m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/319381ac5dac49ac958ffc4add8013f7 : FAILURE in 17m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/47b1abd90ca84fe7990ed3b25f2b96de : FAILURE in 12m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7b6675a3accb4246b2a23c05cb2e9400 : FAILURE in 12m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d24466130e245bfb740300c21193261 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81502bff0ae043bdb5cedf727a573c45 : FAILURE in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cea8e821b9bd4ab4a5f2c122034fcfc6 : FAILURE in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b32f85645b5d47a18cf67c84873696d8 : FAILURE in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cb3bf7727084513b9c1a47240ff8c47 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e947dc405454e2eb66075ef4a595a14 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7515200d06824d0d88010cca79819a06 : FAILURE in 16m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7ffada4fb1045d3a4349d6cb2a4e94a : FAILURE in 18m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2d1c15c251447b497ff4e094241d842 : FAILURE in 13m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ac7926b0ed22479782f359ac4353e788 : FAILURE in 14m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0702c08cffd945ddb62ed856b1d8d68f : FAILURE in 12m 59s","accounts_in_message":[],"_revision_number":2},{"id":"9c7632386b167454874132609c6295b65a2e3f37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 08:27:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/10ae320caaf14bbda7822ed53924c853 : FAILURE in 31m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e0c17085af449378cdba39cab8abfd8 : FAILURE in 26m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f8e159659bf4d7ebe0d906414e39dfa : FAILURE in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"309a32c5ab7a180095425c143dd86cb337c8f21d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 08:44:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f3904fbf970951d3c52c83a813e38bb4481311e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 09:16:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"12ff2355396ec73dadcd7ae21ef74ab122e159a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 09:17:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e81c1f0e6860c65f093fa8708c9d2a82ee68890c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 10:37:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/791a0c49b89e40a3913f0e38fc81167c : SUCCESS in 25m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2bc34e646200438ea45cf0f5ca7d25a5 : SUCCESS in 43m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/718c6b939ead43c4857ba7a8b31ada0d : SUCCESS in 32m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ef1256b6f4f34a5a9dc5303badfa3e77 : SUCCESS in 21m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bec57540714445248912d7a1acaeb34c : SUCCESS in 56m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd8afdbedd0441a8936026ca8a00aa20 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f8eca976952441ba017fe6c3c8bf390 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d107d3ace57947b7807588af650928f1 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/995eb582725b4da198e21df60151fa47 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d8ceb89e234b8aa056bee566ab1067 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd1189c23145453db97d078d63c6e348 : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/451e0df4b7a2487eb1e4ffe02d5a5de4 : SUCCESS in 56m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d882e74d7a2d4833a7ed5a335d3acdfe : SUCCESS in 1h 13m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/125d163b0dbe4b03b8a6278f6ce4ddf9 : SUCCESS in 1h 15m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c5a74224a7154084abb549abb5305ba5 : SUCCESS in 59m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/60d2a9c2827a49558ac830152c8f216f : SUCCESS in 1h 11m 49s","accounts_in_message":[],"_revision_number":5},{"id":"1611c21298e4de0bc43ba7b313e14a80874f5e1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:50:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3dfdb551744640d16222458e97a581e29311f97d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 09:19:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0ca006f04e5336529f23b273d8c625ba2e1010f1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 09:20:19.000000000","message":"Patch Set 6: Backport-Candidate-2 -Review-Priority\n\nXena","accounts_in_message":[],"_revision_number":6},{"id":"9ec2cbffd8b1fac04cbd6620af89545318d1f731","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-26 09:29:25.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nneed to drop metatype from docs as well","accounts_in_message":[],"_revision_number":6},{"id":"1f7ae020c2ae857527d72014adc53116ec4241d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 09:44:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8c8a11e89ea46a5d18026deab60aefaca7d08344","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 09:44:22.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"81f5b96025c3bab0aa99726dc7ef0766ba3ac06a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 09:47:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"29a9476b885aca3f20eb330b91cf78f04cab5a1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 10:20:09.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"cf158e6860dda67490c3ab2c05bc5a2a3840d12e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 11:44:43.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b4e6de0ed98f4d1bab7753b1d8429ce4 : SUCCESS in 32m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2711c6d41a474b6f9697bf36009aa4fc : SUCCESS in 42m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c6e1938c7eac4cc6a9522861f05dde71 : SUCCESS in 1h 15m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ae227e8961b54c82bc3110b8d747c0fa : SUCCESS in 1h 10m 35s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bf588013d08d47dea65223bb3312ac8c : SUCCESS in 1h 13m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/94750635724c4064bcdfe1274617a093 : SUCCESS in 49m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3a20f50a0c474f1b86c0cab3e3895f74 : FAILURE in 27m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c10567a0f0cf477aa228253562b94df8 : SUCCESS in 58m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727b6e53074640dd83073ab78ab394ab : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5875adf822c446981b5fce251a8dc62 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e228a87484b74532b1f831e764269b22 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24925ea001f84c079e23a8e9a3ee9265 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a750fe4aab0d474ea55498c0c9c4251c : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf480bb669fb4a2fb3841eb2dc8e55e6 : SUCCESS in 4m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b5096d424faf462882229a095e36f555 : SUCCESS in 20m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a92c88e7c664c0eb9467e673e374312 : SUCCESS in 45m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9dca388761fc481b9bfd1967d03b19a9 : SUCCESS in 48m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e51bf11cbd104ceaa65a2ae11d32e15d : SUCCESS in 1h 01m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4159fc5abd4c4e90a5365841a92f4471 : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":9},{"id":"b43381e54a1a11e163cab7f71764ed7a93e4f28a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:32:07.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/93f30387c0da49a8a37a3e216c3e9cf3 : SUCCESS in 1h 33m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/99cf26d3783f4b8986741293ad499824 : SUCCESS in 2h 08m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1eaa2f26419d277a02bcc4ac012de4e61b841bb4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-26 15:13:48.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"47e04048cad79a4351720d80486cc35fd1ed065a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 15:24:11.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c871fab4611bf95918a5c8121bec9e01d3cbf773","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-26 15:26:19.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"49d08e264ddd0617cfbf16d4ad30c9151ebb40c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-27 14:13:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5133b5ed2e32e1668fa92ddc855dcdfaaa204fff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 15:50:24.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/49ad44753b2642068e85edf5d1680d5b : SUCCESS in 18m 44s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f4d20c08ceb24929bcfc4fc4b3eec9b2 : SUCCESS in 40m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/645ab0fb13934bc996962a021c9be57e : SUCCESS in 1h 06m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f22e265c75984912b6d29f461396da1e : SUCCESS in 1h 27m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3de1193e99044b3a95feac82c2ca09da : SUCCESS in 1h 06m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c520efbdefbe42fc82c21c0eddaec448 : SUCCESS in 49m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7db21c77f1794cff95a1011178a09bb3 : SUCCESS in 17m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9ff3d8a295ce458798fdefa68eff3bc9 : SUCCESS in 1h 26m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/693a1b27c7a24f3fa853f012227093e2 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b764dc06bf642578177293e692869af : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/548e2562c5fe401cbbb254fd26d9f0cb : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afb3ca0f9964492380e65e7109c77747 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/405c5bec0d7a4419b78e12b001d01442 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9904041b3e7b4357aa20d5cb250cfc73 : SUCCESS in 4m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/203cd043e409430e8f5c1a2c06c4ee6b : SUCCESS in 21m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2069fc43fa0478baeae54668e4f1590 : SUCCESS in 42m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/48cb8f0f3e0a4c2f80595534178e1d0b : SUCCESS in 49m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87f81f55c90c429bba51ff5d81791e79 : SUCCESS in 48m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/acd5f4fd3e1746169a9971f2f0c9d701 : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":10},{"id":"052f86333f0ea9e1554f420496a1420a869cf1e2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-27 15:54:20.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"8eea549b0f5f6665fe6b112ec04e74059f30476b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 16:38:29.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/621bd7d67e4d4b178bb0126c788d4ad4 : SUCCESS in 1h 24m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb0ddc0c08db4cb6af418fd61e98bc81 : SUCCESS in 2h 16m 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"bad7d0c2de283b3862ee5e2649c1ecdd6267c2eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-08 09:00:32.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"45a56d09efc5ec6b794ad2a7ee7b64262d78b10d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 10:07:20.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1a6f7b26c5f94f6b8b4e57f815994191 : SUCCESS in 19m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ff9332a14a7442d9b08b9b5bb5e0fa65 : SUCCESS in 29m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/41a8eb2663644b7abced3f24ff8ccbcf : SUCCESS in 52m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e83d6f5a357647c2853fb78212bd45f3 : SUCCESS in 50m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/aad78971f1e94738b5ab8380786b3e4d : SUCCESS in 54m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ed202fa58022488e9c37b84017cde325 : SUCCESS in 31m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d2cb5cf0c84f45d3860e9b5684ae699a : SUCCESS in 17m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e55ca159eee34635a4b5101dad1ae00d : SUCCESS in 50m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f151b03e55f44c59f06445bb43ddf6a : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/637125d953a24f7baf79109072c64b21 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6f0b441ef9446e48a1f858b0a856bbc : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7aad711e7bae47af8b255a5e817ed249 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1cc9d4a8b0941c7974a25743d99545b : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/246a4d35d26b425296f9d93572ae10bd : SUCCESS in 5m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d055f36b2cc64a2bb40790b9ede09666 : SUCCESS in 21m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e7e6396c0b5443ebdd1190dde852713 : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ece688b11244f63be522a1e8f6a60b8 : SUCCESS in 55m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c54b9c6ba8f49db9929f26ee5334e4e : SUCCESS in 48m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/95b4608018cb40f7859f7d298cdf37b3 : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":11},{"id":"0310fbb7cc06f6cafa9ccac634c7343d163823b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 11:28:01.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/26995b537f6c40a48053f20e77e126ee : SUCCESS in 2h 22m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/111e7e41a9f8460cad4857e56c9da026 : SUCCESS in 1h 14m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3181dcc6baf04080b97617c4dda168ae : SUCCESS in 2h 25m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4e140f13a2e04d1fafc8c867395dd669 : SUCCESS in 1h 54m 34s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a21cc26d91c511f70592511eb3f263198bfd73b6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-10 17:54:56.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"db532fe057fe3197b3af8588ff864ba8f70b8cac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 06:47:46.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"cb5f23d1571433ae7772d585b567e77829a39349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:03:50.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5a97feb8ec0b444ea59a25125acbdbb2 : SUCCESS in 40m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a5eb7b301db5463d8f2c1c46293659a1 : SUCCESS in 1h 53m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/208e24e19764435ebb2199a85bfdc526 : SUCCESS in 53m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/89010cc1301443dfb2d5fab12d0b86f2 : TIMED_OUT in 2h 01m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/49edc249a5824307b81e22293bae1dc6 : SUCCESS in 1h 38m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/17d16a28bbb847af9df6345e0eafbfce : SUCCESS in 1h 47m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ba407ec532cf4ab38ad501a92e128c67 : FAILURE in 2h 13m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/61a5276f064b4da1aa12d3f91a9a0901 : SUCCESS in 45m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c340645d56254d8a9394ef5cb7fc2337 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28d81f63b96a492188defdaab8d580f3 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16e09968ab1d4919bdcb0d416ac1ac7a : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43f536a897fc40a99df4d98164d9eacc : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f3623d61f964cd5b4b0fdb730da21c1 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13ed03cdea8241cda65ca40266fe90cc : SUCCESS in 4m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e068eaf38bf43b0b26f6878076cfbd1 : SUCCESS in 23m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ed49d034e4e40579549fc88745f18da : SUCCESS in 32m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/849a2a64883a4abe9773d5cbcfd2b5d3 : SUCCESS in 51m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/61c80b21ebd249aa9802a1364c5c9ea7 : SUCCESS in 43m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/74a8005548d64cf590de1a266315c2fb : SUCCESS in 1h 09m 16s","accounts_in_message":[],"_revision_number":12},{"id":"5f6f9bcfb59350156fc94a4e5e13da7143775470","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:47:00.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/df3a3c70e0c34edc8a4955c027112f76 : SUCCESS in 2h 09m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5973b257e7574a9abdd119cbca3aa9f4 : SUCCESS in 1h 55m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/da9a9e25fcf94261a792893e0557aa5f : POST_FAILURE in 2h 52m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/38182c8e27784fabb748fcfa2e90ec9b : SUCCESS in 1h 47m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c188578fe4ac8ff37f1eb827bfe41cfd26fcca35","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-17 10:01:28.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"8085e3beb7f2b2540882495f27fe01b88c516d16","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-17 10:02:59.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ab919c47ea39ba8558c7300b1118a572b4a0b557","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 10:20:44.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"2c7b43b1b05afe3756debf6ef04d05e81d68af29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 11:46:33.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ff0e194af4974ef492f743666827c0ad : SUCCESS in 22m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1b87e016633a418a8708f97c16aac2c1 : SUCCESS in 34m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fa33802a864149f6b94987bc4b4b3e28 : SUCCESS in 1h 12m 07s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/eb0215faefb14d8797c70953f0078adc : SUCCESS in 1h 01m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/76482b36972d411bbe823cd01d2cd644 : SUCCESS in 1h 20m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6cde07eb959d4e9ca835294769a21ff8 : SUCCESS in 35m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6dd8f57c06dc46d9820d1cee4606889f : FAILURE in 18m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d6775d939b54435eaa6140acad60cfd3 : SUCCESS in 54m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee1fb3a58c5c4b87872036498063ad57 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d09c6d4f98642c8839abb55dffe0ac2 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83661617f71342a3801054513dee72f9 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e62dc762e3194819b0eda3d65d3fe738 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b8c377a5f8c411f9e6c1d2dd9eb34a4 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8508d1d1b1a84c6bb62aa17605b6bf09 : SUCCESS in 4m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/706b4ba95ba245629533e210117cb71c : SUCCESS in 19m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6dafe395672840c585da4674ea3d5d21 : SUCCESS in 45m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2612c04c9dfb43c790b1b732e69e0ced : SUCCESS in 46m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ed1c2418c6144ed8580d90d11ff104a : SUCCESS in 52m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6bc31a1570f144ea8da79ecf3c3a0af8 : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":13},{"id":"c3a9b30fac0984c7f6c3a64b3148c242fbb7aba9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 12:37:46.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"3b92d960a9d82adb0e89fa9fe239bbafd91f8e61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 12:37:52.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fddcb226a04e41b9aaff219fe9801bb5 : SUCCESS in 1h 44m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0cf459ca1ef4e5194aa5ca6125ae372 : SUCCESS in 45m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/584bb8af14454a77b6e72f172d4863ef : SUCCESS in 2h 15m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a028267331b141d6bd18304164a8c384 : SUCCESS in 1h 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"43c624150498d3496f2b19d5cda97b829c58f264","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-17 13:25:45.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"e5f22427b0f0f65a5c3a57c6cf8a935c6f23ed63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:03:35.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9ba3a263f7d64608b63701e7a43c88a6 : SUCCESS in 43m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5149660bcbe946038f077d8cdc4b282f : SUCCESS in 42m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e927c72eca874450a7c4691c6e06de9e : SUCCESS in 53m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/004f3eb247d449e9896960ec0ad9b859 : SUCCESS in 56m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/53f0071d6aef4a3c83959e6246a66f0a : SUCCESS in 1h 11m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c1f06a4f077a4197b1bf125a019d6294 : SUCCESS in 41m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/739cfd49428c47eeaa1c9edb9dc2fc16 : FAILURE in 15m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/16c4f450957843b7be6764a038f23c98 : SUCCESS in 1h 04m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f7a35e252854779a5ed9dbadb531b4c : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d090ff975a7f403a886b963fb481642c : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81af297d4cc44875a0aa0cd15f1df04f : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f384ecf4dcd64065a42b94bf2a890773 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e9a17fcbed47aa823da73d7e8575a9 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/083d25a6ec1e4bc6b39942485a755810 : SUCCESS in 4m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a7ceffbc726f4ded8c2795f1ef73605a : SUCCESS in 24m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b96b9364219042b3a8a2cd07179f2bd0 : SUCCESS in 30m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d258faf6b99c409db35798fb1b5599a1 : SUCCESS in 36m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b4703babeca54a1291a0cf6ada69fbf3 : SUCCESS in 1h 11m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2312364528e14d5ba369e5af59643c05 : SUCCESS in 1h 24m 41s","accounts_in_message":[],"_revision_number":14},{"id":"cd011fa11aba0648b869b06328a357a5081a3525","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 16:12:12.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/358d16f6394e4479bef199f7e0343483 : TIMED_OUT in 3h 09m 57s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a953a1b1554e4899a07b73fb7a99e593 : TIMED_OUT in 3h 14m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7398b1c8f18944f0bf7d30b715b359b6 : TIMED_OUT in 3h 01m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/105e1a577b0b44a497dbd808fe1d02cb : TIMED_OUT in 3h 07m 33s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0bb1aa193632c3f9a402f2f497f49c834b887025","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-06-15 10:30:50.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"200a1bbda4911d1c7f0f79c6fe7733dd928fdb97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 11:45:21.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/abddbfa1cc904ba49d88037b14bb52f3 : SUCCESS in 27m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/54887bf838b04dd4a6c8ece9dbf59d10 : SUCCESS in 46m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/73dfd3a0b5f247d88869222e08bcdc54 : SUCCESS in 1h 05m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/11865351e57243a394a0c4e1594676c8 : FAILURE in 10m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/054beb56393f47deae548a1f994d9519 : FAILURE in 7m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d79e139aa0684949bf44baecbbc4217a : SUCCESS in 30m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/175b223da99e4744be89c9c315c377a7 : SUCCESS in 19m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e00543fcacc4a7e8548531c2421e0f9 : SUCCESS in 1h 01m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7dc1d644621441f8dc5828f40183ca7 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8065a6b7f6df471a8ccd0713157e52f7 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f777522d006e48648b83301dc1b05702 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2ed509a88da40d089fbdba64ff30df4 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ef63c40a81d46b2a6ff84a5ac21434b : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47dd45de08c3478b8b8e7d3db1d1bbb6 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed5fe175b26442aea8d2fa9bfd5cb873 : SUCCESS in 20m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf383edac96048b88dfc07621a9e77fe : SUCCESS in 28m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4896c8fb456941df90fff739039f0e5c : SUCCESS in 50m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1bd52fb5e1944c17b8eae23ee3b0c106 : SUCCESS in 59m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/13b99731835442fc91dcfd4b1add6f7d : SUCCESS in 1h 08m 40s","accounts_in_message":[],"_revision_number":15},{"id":"c0a9912b111374afaf6f79f6fd099513ad5a29ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 12:38:37.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc5eb1ed3e4d451a9c7693caf73ebb10 : FAILURE in 1h 22m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f910ee9fd4fa4c2abc26e2a4e306bdb7 : FAILURE in 1h 11m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58859470ed514f34a7bd33e7ee439e3a : SUCCESS in 1h 47m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08ddb9d9324e415bbf75f38162dd2c7f : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae2d54109e5c440687ff2384d0336290 : SUCCESS in 7m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/38e0eef7bb2f402f983580c659e35674 : FAILURE in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1348adee96d13b762daf0804d2473dbf3b2bc2b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-06-28 09:04:50.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"3d55effb86cfd16954b245688933b0c19bb062cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 10:24:08.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/13873b62a3da4790941de0bbf601a17a : SUCCESS in 1h 02m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/eb44aa49700849f19bd983b98720ab17 : SUCCESS in 1h 02m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/86ae6a3fb5014fc38ef52c3da462b6a5 : SUCCESS in 1h 14m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/af12a3d3b00d4f15b2e4f423c4747c11 : FAILURE in 16m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ba175b2e68444e871466a01e73eaa1 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1043cc3b75545de859df24298d553fa : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58b931af31ae4fff95eaec9ab28c52b5 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4653e19102b480c9ff813e039d380ec : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d61887cd3094985aff2ea029067fc01 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e02c94b4d5374dc7bd97b6f37dc9e15d : SUCCESS in 7m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a78b83a4ebc4aaeb2c5974aedc2507a : SUCCESS in 1h 05m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/472fc04080ff4505be93c1538cd4d845 : SUCCESS in 54m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f6a6b7087f394a90b6d1f886e7ba4dd8 : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":16},{"id":"ae0e06c034cff919a9449c08f80748ac6e8d41c7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-29 14:56:37.000000000","message":"Patch Set 16:\n\nduh, merge conflict again","accounts_in_message":[],"_revision_number":16},{"id":"98632e3d477000e12cb135a5bf3fe2f16f6adc83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-29 14:59:05.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"82ab08da0fda68d3ba8fcf0fa6b312804925e12f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-29 14:59:18.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"09a5af1113c47e6b80c222ba82ae4c7d8df7a365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 16:21:25.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/59159ec0bf2e43adbe6b6595184eeaff : SUCCESS in 23m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/636586fd6ccb4738b1fde194daabbc16 : SUCCESS in 41m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0c711d49979a4257b9d6aab11fcd9a0c : SUCCESS in 1h 07m 35s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cc256007be1d4aefac6108303bcbadd9 : SUCCESS in 1h 08m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a01f9b34df084d20b897aa6a241ae190 : SUCCESS in 1h 16m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cde07b44f3e34b098986cd292c3dbad7 : SUCCESS in 28m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b6f3eae059954189919019e7a6079926 : SUCCESS in 22m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d4ae8d5eb0a744d3aca32068646de8d2 : SUCCESS in 54m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c682329747b44790960d304158bd4869 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cb67d9f25f14553a77955be301bc761 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9553367ca583446c8c84aba91905c106 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fca46301e1cc4a249ee5c714d2b753c2 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95fd5852a5d4c6d86bc1889a7ff65c3 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd440776423d44c9bf2ec081aef0fa30 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5a836900eee64209bb6d9f9fcd604795 : SUCCESS in 20m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f73372ab171a4f848dac7e39d51e6bce : SUCCESS in 35m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6f52b3943e9453685ca0b1947ecfc72 : SUCCESS in 50m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c9cd87defb254333a504c98af42a5a08 : SUCCESS in 49m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6f02c8e449674d90959f40575b8be1e5 : SUCCESS in 1h 06m 56s","accounts_in_message":[],"_revision_number":17},{"id":"ea9a398c46e6a8c375f7d67e958fe538ac7355f6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-29 18:51:06.000000000","message":"Patch Set 17: Code-Review-1\n\nSome hits for \u0027git grep -i rhel\u0027:\n\ndoc/source/admin/image-building.rst:- rhel (deprecated)\ndoc/source/admin/image-building.rst:RHEL containers and RHN\ndoc/source/admin/image-building.rst:To build RHEL containers, it is necessary to include registration with RHN\ndocker/base/copy_cacerts.sh:        # CentOS, RHEL\nkolla/common/config.py:                     \u0027the packaging type is set to \"rpm\" if a RHEL based \u0027\ntests/files/process_build_logs.py:if args[\u0027base\u0027] not in [\u0027centos\u0027, \u0027rhel\u0027]:\ntools/validate-binary-build.sh:for distro in debian ubuntu centos rhel; do","accounts_in_message":[],"_revision_number":17},{"id":"13db1988a8a3c1302fc2aa9e9ba8c9fc651360ac","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-29 18:52:05.000000000","message":"Patch Set 17: -Code-Review","accounts_in_message":[],"_revision_number":17},{"id":"5bf7998bf4be0ab143511d774533d579469f3a24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-06 11:04:42.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"ac00ec7c32c154513815941579f27f20b4c1b1a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 12:33:43.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1beac1c15b3a4405b7c5c845abd3f4f1 : SUCCESS in 24m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3a7994121e574d40a573821729f3efb1 : SUCCESS in 37m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1eb2d28b3c9a41878b02530324d2ff16 : SUCCESS in 59m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/75a92be230124ea68ab2113972d0d7e4 : SUCCESS in 1h 21m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0a9dcbd60da345b2ac1d9a6ca34ca650 : SUCCESS in 1h 18m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5dcc1b75206a454fb3bd7c5920a153d0 : SUCCESS in 50m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dd2696b971e747cc91017847d806a707 : SUCCESS in 17m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b8c7d75a6a4349f1a6e0f7c566cb6fc0 : SUCCESS in 52m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c780a17d46489b91fc2accd2784bfb : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1138a281fa9841408fb1cc51534d4e6b : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/301f22da44ef47fca30a9c2ef6dc9ea8 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfbe628a2cbc4567a00533d03406a140 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e7cdea7e1a4fbaa690b38f8bed5eec : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0f39c5fa2214612a8fe724601823d6c : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/976e4d03be3648669fb773b83675131a : SUCCESS in 39m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ac4e5213c014fb19253c6cc05b56121 : SUCCESS in 27m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6fe2dc8c6cb477fb65ce18da9cc70fd : SUCCESS in 49m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/64a3f5ca76414594b6e596bef75e21df : SUCCESS in 54m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee85e8e2bc7948289d251369160217b3 : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":18},{"id":"ac9e7d0df4595002ecfd671fbb914d8dade6a90f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 16:41:23.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/58db74f5c9994c86bebd061e7e6c21ac : SUCCESS in 2h 45m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3740cbf04fa442b8bd211f85601da44d : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44e3e868ef344902934a5126650864f1 : SUCCESS in 6m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e10f9ca448d64672843d57bbdd1174fe : SUCCESS in 2h 00m 02s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"5f88ce2e592d0ecf995cfadd272b25004f72df04","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 15:42:36.000000000","message":"Hashtag added: cleanup #drop","accounts_in_message":[],"_revision_number":18},{"id":"24da004d01069e3143857e29fa69b13f02a84d54","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 15:52:49.000000000","message":"Hashtag removed: cleanup #drop","accounts_in_message":[],"_revision_number":18},{"id":"1d78ababbd5643859d905fe733734ab78bab748b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 15:52:53.000000000","message":"Hashtag added: cleanup","accounts_in_message":[],"_revision_number":18},{"id":"6cc7d13d60ee1e8e1ba3ba98f64c131cdf2a3e9d","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 15:52:56.000000000","message":"Hashtag added: drop","accounts_in_message":[],"_revision_number":18},{"id":"1be5678dd386047e0e13ab2a89382a6a26b10af1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 16:04:21.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"e3450bb4aba4eae21428920fa6868fbbdb78cc62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 17:26:37.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/383adcb60bed4e3bbaad864fbe557bd5 : SUCCESS in 25m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8207f6d0e91f421d91b31d292044cca9 : SUCCESS in 42m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c18287d8df2f4d36b896e5f7f6857a2d : SUCCESS in 1h 08m 05s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/69f1ac47c71442c98e04de55e73c8191 : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f65a3ed34ae44639a7c42b06f8bc730a : SUCCESS in 1h 20m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7bbb3db5ab704b469b2b9f52dc64efcc : SUCCESS in 35m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8dd6772cc2654d4982c22bc363d47fcf : SUCCESS in 17m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cb2183cb328644bc97dfed844de4deb7 : SUCCESS in 1h 01m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bbbdae345184411851d2a5fb537861c : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c50a4640df774c8fb888e291207dc1a8 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6b8d7bcc4024d1d9cb7df520f815f0c : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff64568272c24937bee5b8a3b78a2ade : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8603c39f7d644d2b2eb24d1c1bdf937 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/175ad7c5fda14492801a45e928b73158 : SUCCESS in 4m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed48da9f6a09496899438da2de80b9af : SUCCESS in 19m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a11663f781949e3a67f69865ca2b659 : SUCCESS in 22m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28a6b7ea07474e0cbe1953ad4938f0d9 : SUCCESS in 54m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2d4abe6f23d4a5681e711e58977d670 : SUCCESS in 53m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/40c7b21ed2e34f08990f487c97dd7bea : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":19},{"id":"f94d7a3891c81fc269bfee7be61b85886d6c7f9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 18:16:36.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/73cf0dc4fa204e9e8b7f7351a0e8e9ab : SUCCESS in 2h 03m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d24c17e89304c2ca41cfc6ee98a444f : SUCCESS in 49m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0857bf5e5d35444297c8a3fd3c395701 : SUCCESS in 2h 03m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d48dde5294fb4e9883ed56d0f6a83b7d : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df9430014197479d85d39225a3784772 : SUCCESS in 6m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b988f239d93e48cca9027a857364e060 : SUCCESS in 1h 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"8e9b91db672d7bd9aaa7442baccc7fbab97146c4","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-04 11:18:41.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"8ed6663d872ff0df41561856df008627427af933","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-04 11:19:32.000000000","message":"Patch Set 19: Code-Review+2\n\nMark\u0027s comments are done","accounts_in_message":[],"_revision_number":19},{"id":"f0872b7d92660e534da602ad64b7437c9f46e465","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-04 12:39:24.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\nThis may not be the last we see of RHEL, but let\u0027s start with a clean sheet if it comes to it...","accounts_in_message":[],"_revision_number":19},{"id":"df9aaee1c719423eb72ef3bdb7bb3d99fced820d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 12:39:51.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"c46a50e9bc8df16c6e2fb5bd6bce3a816c34a33c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 13:53:41.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/21c9063d77b942adae4fa1707913b7b8 : SUCCESS in 33m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/64a4070d04af47bbbf47e1f355672e66 : SUCCESS in 57m 19s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/66fd79df737a4893b408d726a3b289d4 : SUCCESS in 1h 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa15b3ec38b417984ee2a82e76ba5b0 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ee28c5f5c1e42c09013a1f9152c069f : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d8ec068dde540e498345033721c2c3d : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473260ba02cf4fa1baa24a72febc0b75 : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f17831d4ece4c8fba9d76485bbaca53 : SUCCESS in 5m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0575074e0aa844b48a7797c9b37696d7 : SUCCESS in 24m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3b4257c421d42ddb33e9ed9bd346e18 : SUCCESS in 46m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ddaa0ee34e454e12863df044c0f126d7 : SUCCESS in 1h 01m 36s","accounts_in_message":[],"_revision_number":19},{"id":"f3c2d3022991236ded92b7db7fd8870ce38566d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 13:53:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19},{"id":"f08a9d0c91fd4fc877240dfc0fd173e2f255a374","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 13:55:03.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a175d3eb294543aa56ccfba9b23cad : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/530801506b4a4237944c5ae3ec9c4683 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"bd90b02708d8c6b88692ec7308a41229fdcdaf22","revisions":{"793f8d0b8c6092d1dce6b34522769fd360d14765":{"kind":"REWORK","_number":1,"created":"2021-04-09 07:37:26.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/1"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:37:14.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/793f8d0b8c6092d1dce6b34522769fd360d14765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/793f8d0b8c6092d1dce6b34522769fd360d14765"}]},"branch":"refs/heads/master"},"baa92385c56e37f4c9482361db36d66aea62980c":{"kind":"REWORK","_number":2,"created":"2021-04-09 07:48:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/2"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:48:30.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baa92385c56e37f4c9482361db36d66aea62980c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baa92385c56e37f4c9482361db36d66aea62980c"}]},"branch":"refs/heads/master"},"9bc6c4eebdf2451fa6e46aed2f8ff976a628f82e":{"kind":"REWORK","_number":3,"created":"2021-04-09 08:44:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/3"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 08:44:36.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc6c4eebdf2451fa6e46aed2f8ff976a628f82e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bc6c4eebdf2451fa6e46aed2f8ff976a628f82e"}]},"branch":"refs/heads/master"},"71f57a08bac0d3d25b08daca6792196d69dc5a54":{"kind":"REWORK","_number":4,"created":"2021-04-09 09:16:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/4"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 09:16:37.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71f57a08bac0d3d25b08daca6792196d69dc5a54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71f57a08bac0d3d25b08daca6792196d69dc5a54"}]},"branch":"refs/heads/master"},"a5a4506f89640782e49c8b2067c7c91fed92f583":{"kind":"REWORK","_number":5,"created":"2021-04-09 09:17:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/5"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 09:17:06.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5a4506f89640782e49c8b2067c7c91fed92f583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5a4506f89640782e49c8b2067c7c91fed92f583"}]},"branch":"refs/heads/master"},"de5f53962cc77a6f32d4d83e2fb6a0b974a9b867":{"kind":"REWORK","_number":6,"created":"2021-04-26 09:19:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/6"}}},"commit":{"parents":[{"commit":"5f2eb06b89bd11f7b8ae001f6ede17a452a502ba","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f2eb06b89bd11f7b8ae001f6ede17a452a502ba"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-26 09:19:24.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de5f53962cc77a6f32d4d83e2fb6a0b974a9b867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/de5f53962cc77a6f32d4d83e2fb6a0b974a9b867"}]},"branch":"refs/heads/master"},"c52d180469a595fa0f2b3eeff6cc560859165a85":{"kind":"REWORK","_number":7,"created":"2021-04-26 09:44:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/7"}}},"commit":{"parents":[{"commit":"5f2eb06b89bd11f7b8ae001f6ede17a452a502ba","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f2eb06b89bd11f7b8ae001f6ede17a452a502ba"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-26 09:43:56.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c52d180469a595fa0f2b3eeff6cc560859165a85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c52d180469a595fa0f2b3eeff6cc560859165a85"}]},"branch":"refs/heads/master"},"e698c8ba909ce5cdf2a1294bfd45799d724b1938":{"kind":"REWORK","_number":8,"created":"2021-04-26 09:47:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/8"}}},"commit":{"parents":[{"commit":"5f2eb06b89bd11f7b8ae001f6ede17a452a502ba","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f2eb06b89bd11f7b8ae001f6ede17a452a502ba"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-26 09:46:48.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e698c8ba909ce5cdf2a1294bfd45799d724b1938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e698c8ba909ce5cdf2a1294bfd45799d724b1938"}]},"branch":"refs/heads/master"},"20916cf983f93f9f38b2732cd6209c07ef389597":{"kind":"REWORK","_number":9,"created":"2021-04-26 10:20:09.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/9"}}},"commit":{"parents":[{"commit":"5f2eb06b89bd11f7b8ae001f6ede17a452a502ba","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f2eb06b89bd11f7b8ae001f6ede17a452a502ba"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-26 10:19:42.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20916cf983f93f9f38b2732cd6209c07ef389597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20916cf983f93f9f38b2732cd6209c07ef389597"}]},"branch":"refs/heads/master"},"6a8de5bab13985cfa488fcfdd03792c54533a141":{"kind":"REWORK","_number":10,"created":"2021-04-27 14:13:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/10"}}},"commit":{"parents":[{"commit":"edc14536d0544c81e2e5f3d011a9828177800547","subject":"Merge \"releasenotes: we move to Debian bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/edc14536d0544c81e2e5f3d011a9828177800547"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-27 14:11:40.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a8de5bab13985cfa488fcfdd03792c54533a141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a8de5bab13985cfa488fcfdd03792c54533a141"}]},"branch":"refs/heads/master"},"2d4e431c43f63e170379792d630c613d3e5c486a":{"kind":"REWORK","_number":11,"created":"2021-05-08 09:00:32.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/11"}}},"commit":{"parents":[{"commit":"6dfae74813289575e77aaa7dfe43f7e6a81a5f26","subject":"Merge \"[release] Use UCA Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6dfae74813289575e77aaa7dfe43f7e6a81a5f26"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-08 08:58:58.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d4e431c43f63e170379792d630c613d3e5c486a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d4e431c43f63e170379792d630c613d3e5c486a"}]},"branch":"refs/heads/master"},"300aa8927f5ee7418848ce003a68ed87930bf11a":{"kind":"REWORK","_number":12,"created":"2021-05-17 06:47:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/12"}}},"commit":{"parents":[{"commit":"73c456f85b4e861e4be12282a67ca59f5f319357","subject":"Merge \"docs: update supported OS distros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73c456f85b4e861e4be12282a67ca59f5f319357"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 06:47:24.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/300aa8927f5ee7418848ce003a68ed87930bf11a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/300aa8927f5ee7418848ce003a68ed87930bf11a"}]},"branch":"refs/heads/master"},"a13845a602b0db579fd01bd89357daeaaa482955":{"kind":"REWORK","_number":13,"created":"2021-05-17 10:20:44.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/13"}}},"commit":{"parents":[{"commit":"73c456f85b4e861e4be12282a67ca59f5f319357","subject":"Merge \"docs: update supported OS distros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73c456f85b4e861e4be12282a67ca59f5f319357"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 10:20:41.000000000","tz":0},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a13845a602b0db579fd01bd89357daeaaa482955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a13845a602b0db579fd01bd89357daeaaa482955"}]},"branch":"refs/heads/master"},"e76d91a97d2b5f5fa545e2c482efe3f6a9ffd029":{"kind":"REWORK","_number":14,"created":"2021-05-17 12:37:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/14"}}},"commit":{"parents":[{"commit":"0506ede84f97159a8156ad51aea395d0868e6141","subject":"Drop ppc64le support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0506ede84f97159a8156ad51aea395d0868e6141"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 12:37:35.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e76d91a97d2b5f5fa545e2c482efe3f6a9ffd029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e76d91a97d2b5f5fa545e2c482efe3f6a9ffd029"}]},"branch":"refs/heads/master"},"d1e94ff5ecd65a5d1f29b9c541a61cce2f592fc3":{"kind":"REWORK","_number":15,"created":"2021-06-15 10:30:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/15"}}},"commit":{"parents":[{"commit":"3110cb136731c1d2b1598c6b3006069a90e464fe","subject":"[docs] Fix Debian release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3110cb136731c1d2b1598c6b3006069a90e464fe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-06-15 10:29:26.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1e94ff5ecd65a5d1f29b9c541a61cce2f592fc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1e94ff5ecd65a5d1f29b9c541a61cce2f592fc3"}]},"branch":"refs/heads/master"},"cc591a834dfdef820d9555d6d3da7a909a9f4d8b":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-06-28 09:04:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/16"}}},"commit":{"parents":[{"commit":"8e020c5f86e033d9baba2c80c39f800bbd3cf2de","subject":"Merge \"tox: Use allowlist_externals instead of whitelist_externals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e020c5f86e033d9baba2c80c39f800bbd3cf2de"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-06-28 09:04:50.000000000","tz":0},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc591a834dfdef820d9555d6d3da7a909a9f4d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc591a834dfdef820d9555d6d3da7a909a9f4d8b"}]},"branch":"refs/heads/master","description":"Rebase"},"c25a03ca15acd063c69eb9a67b340e4e0be0955c":{"kind":"REWORK","_number":17,"created":"2021-06-29 14:59:05.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/69/785569/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/17"}}},"commit":{"parents":[{"commit":"314362ff07035b9dbfdcdd1461552f687ca990a2","subject":"Merge \"Fix build of debian images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/314362ff07035b9dbfdcdd1461552f687ca990a2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-29 14:58:56.000000000","tz":0},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c25a03ca15acd063c69eb9a67b340e4e0be0955c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c25a03ca15acd063c69eb9a67b340e4e0be0955c"}]},"branch":"refs/heads/master"},"fb80b1433e44ffed1ba6f505328757f2f2bc87f2":{"kind":"REWORK","_number":18,"created":"2021-07-06 11:04:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/18"}}},"commit":{"parents":[{"commit":"5e0b238b01250392fa9f92c069a7253902d6b325","subject":"Merge \"Bump up cAdvisor\u0027s version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e0b238b01250392fa9f92c069a7253902d6b325"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-07-06 11:03:26.000000000","tz":120},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb80b1433e44ffed1ba6f505328757f2f2bc87f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb80b1433e44ffed1ba6f505328757f2f2bc87f2"}]},"branch":"refs/heads/master"},"bd90b02708d8c6b88692ec7308a41229fdcdaf22":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2021-07-28 16:04:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/69/785569/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/785569/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/785569/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/785569/19"}}},"commit":{"parents":[{"commit":"53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1","subject":"Add boto3 as s3 dependency for Glance container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 07:33:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-07-28 16:04:21.000000000","tz":0},"subject":"drop leftovers of RHEL support","message":"drop leftovers of RHEL support\n\nAs a project we never built RHEL based images so support for them\nwas not tested.\n\nChange-Id: I27dfe34ade088228d71f6857ec4f3ab28ee63915\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd90b02708d8c6b88692ec7308a41229fdcdaf22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd90b02708d8c6b88692ec7308a41229fdcdaf22"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
