)]}'
{"id":"openstack%2Fkolla~785622","triplet_id":"openstack%2Fkolla~master~I635964a84f0cd5cadccd7ce8040fa36e1300c765","project":"openstack/kolla","branch":"master","topic":"prepare-for-kolla-wallaby-release","hashtags":[],"change_id":"I635964a84f0cd5cadccd7ce8040fa36e1300c765","subject":"centos: switch to RDO final release of Wallaby","status":"ABANDONED","created":"2021-04-09 11:21:38.000000000","updated":"2021-04-28 15:06:12.000000000","total_comment_count":7,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"3dcee9aece6d50a16cab7bbaa1886533135768b2","_number":785622,"virtual_id_number":785622,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-28 15:06:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-28 15:06:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-1,"date":"2021-04-28 15:06:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-28 15:06:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-04-28 15:06:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-04-28 15:06:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 11:29:29.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-04-09 11:36:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-04-09 14:07:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-23 12:38:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-23 12:38:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-04-23 12:38:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-26 06:47:49.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"6ad7a245969430daca4262330c4cfbf10a7a7b8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 11:21:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3fb44ae9552db2da6f931d2caa3bd002fb4ef56","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 11:26:55.000000000","message":"Patch Set 1: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2fc727ca946635347e750876b1996f3a110746e2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 11:27:28.000000000","message":"Patch Set 1: Workflow-1\n\nw-1 as it needs review but not yet ready for merge","accounts_in_message":[],"_revision_number":1},{"id":"c31dd4f46752b9a90bc5a395faf2cf64bf30da55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 11:34:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6bf45fe83289cc0dfc34820c32a9694b65b9f6fb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-09 11:36:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ac2057e89a43cce3a70c462838a8b4497410d59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 12:00:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fefcc79ef87260a23f5eabc72892256db225780f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 12:51:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9dede6ac9b14095528bce140f8256720ab631c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 14:07:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b9b77947ada34e46941e8fa14571bfff : SUCCESS in 36m 36s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9cc31bf31bc24e06b00ad43301e60ef0 : SUCCESS in 38m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/231d8fcc871a44ef804c9ca9def4fe76 : SUCCESS in 38m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ab200b07edd542168d3480a08860275b : SUCCESS in 21m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/da206a13dabd4f5ab0cb80f972bbb5fa : SUCCESS in 1h 08m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3258fcd0545f4517b16bc15b817c6138 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcccb95ae40b4039aa120c6c66fc4c34 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7e075b5a3ce4839a75d63a87805c5bb : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/392351f37533467c81fe6b1874875617 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2276a5319144acbedef5ba5d293b70 : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f1cf27a8a225498ab479aa5db9ab90e0 : SUCCESS in 31m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7f7e053b5e847b0b69cfe654a41b195 : SUCCESS in 41m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fce6e21a058e4a8184c7ad57a0c4e9cd : SUCCESS in 1h 09m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/34d9d50236d14967beee6bd450e49e45 : SUCCESS in 53m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4de2b8bb8ca34bdda87ac9acf9d78b48 : SUCCESS in 1h 12m 39s","accounts_in_message":[],"_revision_number":4},{"id":"e6d06de9487964120bf29bba6f7f7dd37852fc36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 14:23:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2d7997db5db54b5bbff3a3f9bcf9e789 : FAILURE in 52m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb5e31140c224f07864d60feba7e1464 : SUCCESS in 49m 20s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1aba5e9e9a7446b850b22810e5edde4 : SUCCESS in 1h 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"498721ee6ff4303c0d72e5e029495e88718bd0d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 15:02:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"64d22a42675d7ce401247ca9955e82247d1b0796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:00:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8bfe533774f94dbbb55cf2a5f163a0be : SUCCESS in 21m 49s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1332abfe545743b697f1b5471cb4d03b : SUCCESS in 44m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eb45962288604233af11b1733503a196 : SUCCESS in 39m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/985eef7bee73438da7bbac298f19aade : SUCCESS in 26m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/502401e5d46f4081b8e4b2daae76509d : SUCCESS in 56m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9cd86f2b03d4d5985d3efa688367c86 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/490cf4a430c64b8e8a4603ea59fa3fbc : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edd2ebc3884745db894cb2597a57943e : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/444ad14538a04e41bf1b521298342323 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c4e1ef28f5b4e2889542d6f3e517aef : SUCCESS in 8m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7cbea7e4098242dc884f27aa8b9b377c : SUCCESS in 28m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fdefbdb31e4e4033b16236ee6c555ebd : SUCCESS in 35m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/79b53b9ff17140479c2972ee99ae384b : SUCCESS in 1h 09m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf71bdb2143a49f58c631cc9b3f552ff : SUCCESS in 1h 25m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/337465f3db974e81848d2e56fb6cd3b8 : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":5},{"id":"d4e30ea87b1f99f8fc03cfd3b004b9fdbe4f824a","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 17:03:27.000000000","message":"Hashtag added: prepare-for-kolla-wallaby-release","accounts_in_message":[],"_revision_number":5},{"id":"6c795e5ad4ec30ff44b30b2901bb49921c53f527","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 17:04:33.000000000","message":"Hashtag removed: prepare-for-kolla-wallaby-release","accounts_in_message":[],"_revision_number":5},{"id":"d4b3e2e7df597d325fb57bf28e532ca7e1a897d0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-09 17:04:36.000000000","message":"Topic set to prepare-for-kolla-wallaby-release","accounts_in_message":[],"_revision_number":5},{"id":"82266540c9b9a2159a56dafd2ee504538987ce62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:50:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cf2546f3e7d86e73156558674eef5aedec21adde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-12 13:17:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"84d91585e13b6b797908af11caccd7191fc1d8a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-12 14:04:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fd1f18efd709196f1ae2448931360ccb911d988f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 15:22:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bae96b1e979144f1bf7028910692102b : FAILURE in 20m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/da505c2dfe084cd69bbadc49c512c218 : FAILURE in 46m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/97d0c9ccb0844c0692a9b70096b61f10 : FAILURE in 47m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ec1bd213011f4e31a864b4a96cd18af1 : FAILURE in 21m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/22ff85ad2273434cacfac7627dd3dcc4 : FAILURE in 19m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354719feab2d4443b3180d70a8318ee1 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ec20124559f4126b94ee9540cafa4d2 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90e65203e76c40d081365edfb4f25879 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b52d55e222434db888929edc1aaf0ca1 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff8573b194f24367ae2f0c85bb392bf0 : SUCCESS in 7m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e3925592bbfc425e9060e4666e74da61 : SUCCESS in 21m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/16b9df67383e48c3a1dda451a389b812 : FAILURE in 1h 17m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e96bc0199f94261919443943dba447c : FAILURE in 16m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9871c5a66f684c21a23afdf4ffa84fde : FAILURE in 19m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aec90e8466da4caab4f26b2d3b1aa2a9 : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":7},{"id":"e1c9df7547336d14a0daffa7bbc33fe94ca9e4d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 15:51:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/414c610fb7d74fe9b0bd1b4880e533b2 : FAILURE in 1h 38m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e0799bf415d747ccb0c2479fa2137bf9 : FAILURE in 1h 28m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dd5d90ae22604d13af13b74d4fa800ce : FAILURE in 1h 04m 23s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b555793f5cc8070f87c661016bf3caebd615ace7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-13 07:54:13.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"93ec848eb632270d7e8b2fee12656114f7b79ea0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-13 07:57:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"448210fb6e3bb6bfbf6f91f381461c094962b77a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 09:07:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c80bdce554d24cd1b6302fcd59bc782a : FAILURE in 25m 16s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0f03f65e3814465b9be7c8de29e0260b : SUCCESS in 1h 00m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/aef72fb864f24016b6c3a57cc6f9f631 : SUCCESS in 53m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f2b4711f959e4b378b397ec876e7f948 : SUCCESS in 39m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/161245db98974da580f9d05ec716234a : SUCCESS in 1h 03m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/123afff9d4124abeaa29d3c1f3d7e653 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e5844e764f344a6ac80ed91d738074c : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3df484b27a4c4264be010611cf773e09 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69ac34a9620348a5b03736b3818ef95b : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95753863b6fd4cdab21889ca1191d73f : SUCCESS in 7m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/901a01a1bba14f6ea24e7ab2ca2f0736 : SUCCESS in 27m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c143418ead5b4b89b5a4f1e5b6a44064 : SUCCESS in 42m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7f8a1edc48d41248e1db1bfb8bb721b : SUCCESS in 56m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ac02ec4c2e74065a2aee30831f32efd : SUCCESS in 56m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/04faef4bbdee435da05ac3e9cdc0dccd : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":7},{"id":"f3e2e0830db1afc7697d424a04eb17f385282ee8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 09:46:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c21e58c4c46a47f6b4bdb521e17e2c7a : SUCCESS in 1h 14m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/292c668a7ad448918fd00491d208fe0f : SUCCESS in 1h 09m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c09480c98f6c4eaeaa84bf1b2fe038cd : SUCCESS in 1h 46m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9e8adc3c9e84d4c7d6dd1c0f20a9857da6859f6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-14 16:41:52.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"e56183269848ee0c737f50d2e9db4057f27dd5ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 17:58:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/68bcbb6b091a4bb4923bb7e8a3797656 : FAILURE in 25m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/039e561fe1ba4c35beb97d3d539c7f5c : SUCCESS in 39m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/27bc8f0837014a92b9dce6a29b0b7230 : SUCCESS in 30m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/05aaa4e9ef2142b9a850ea3861535965 : SUCCESS in 1h 12m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f8333815f7004faa8f7a937c07710800 : SUCCESS in 1h 02m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b857eec3c8bf4fc39c011b4b933491d7 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97b8ebe77a814f9d80113df84ba1cb85 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/520de40991e841dca4435a1a5ed1f147 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d36f504247de4ff584d9528fa9d800ab : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d5392e931a46eb9b6d944c1ccca254 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/36fc5a7ece984a8ca546ffdc5d5eba8c : SUCCESS in 23m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/78eb6d1f5b5244d9a3a2d03cfe4b8f9a : SUCCESS in 37m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f78b1ca159140f2b0b48358449199a3 : SUCCESS in 53m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/825fef3ab9394572b8df29d0e21d0208 : SUCCESS in 52m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cd5de6ceca9f481c978e6b6e6811eee2 : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":8},{"id":"989a58408bfe8ef4d4249a272d1a109a19be9623","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-14 19:30:48.000000000","message":"Patch Set 8:\n\nINFO:kolla.common.utils.neutron-base:No match for argument: python3-networking-vmware-nsx","accounts_in_message":[],"_revision_number":8},{"id":"6636e3c5e3df02ee63f4e06b7f8ef890b8b16cd0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-14 19:31:10.000000000","message":"Patch Set 8: Workflow-1\n\nW-1 so no one would merge it by accident","accounts_in_message":[],"_revision_number":8},{"id":"4556085f0b9795309155fc72df5ad2b9b207f69d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 20:04:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6f8f59ad3284defa01a16395a56f0d0 : SUCCESS in 1h 15m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c8bf7e0641b43fe986d5ebc0f12cda1 : SUCCESS in 1h 14m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/32cf71b768ec43a48c9f29893fb2a9c1 : SUCCESS in 1h 55m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"16a47b27250dbe1ab924361726a0b66eaf2ab0ea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-16 08:30:21.000000000","message":"Patch Set 8:\n\nrecheck (python3-networking-vmware-nsx is in RDO trunk repo)","accounts_in_message":[],"_revision_number":8},{"id":"b16277a913ab45b7c90f380d53327f582dc16a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 09:37:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/385c52ba967a4a89b8f40984e4f93fc4 : FAILURE in 23m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5f236adeae7848bc9a3053ecc8704097 : SUCCESS in 43m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/12328468477e4267b0007e986f8625ce : SUCCESS in 33m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/919fef25f55b46eaa2ea5e55c5ad793d : SUCCESS in 23m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ddc1a61e904d4ee79ed49969f4b702f5 : SUCCESS in 57m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ef81515f2764e27b3615610039fcfed : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0795e8f2b9f49a3adc0e7fd2e8b580a : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b843859ab9544d35b5fe2099909ae93f : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fdf832ff04e480a85c3b9a99c3ca2b5 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f4c3941ceb4e529622dd37f80f9d3f : SUCCESS in 6m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b946b0ba36b4a10b2048890fc3764de : SUCCESS in 24m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/82ef07e2e4814c9b81b6dff54bff32be : SUCCESS in 40m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6448b65b551544028704ddaac55e186d : SUCCESS in 54m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b851b7f9125a4934bae1224236e4fa86 : SUCCESS in 54m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/120a7c7eb4024e1b8ef8b00fee740763 : FAILURE in 1h 01m 48s","accounts_in_message":[],"_revision_number":8},{"id":"d4db1990e494a05ba072fb84285197225da6b4df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:10:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b634e1503d114f1398b4fe785ba02c40 : FAILURE in 57m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c8a724a8d2b4aafbb941bb0e76458f5 : SUCCESS in 1h 36m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8054ad50211445a1843d466e07ab945c : FAILURE in 30m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"af1c375e58c35abce09d6cf36f75a5a7341241ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 10:51:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e22cdf1f34ab4f29d5c3c9612084ad6014de3094","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 10:51:51.000000000","message":"Patch Set 9: Workflow-1\n\nwaiting for centos-release-openstack-wallaby package","accounts_in_message":[],"_revision_number":9},{"id":"1faf88735cb4b2537b07744e9579f2eb5adaa11d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 10:51:57.000000000","message":"Patch Set 9: Review-Priority+1","accounts_in_message":[],"_revision_number":9},{"id":"30ad98a0cd1c209d2195557b709c96bd35658ce2","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-04-22 11:16:27.000000000","message":"Patch Set 9: Code-Review+1\n\n\u003e Patch Set 9: Workflow-1\n\u003e \n\u003e waiting for centos-release-openstack-wallaby package\n\nThe package should be already available in centos repos.","accounts_in_message":[],"_revision_number":9},{"id":"3bec0c1a8056696a40d12a547c3a5eb5c96d8ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 11:57:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d577501d75934b85ada6f3b43d6d56dc : FAILURE in 25m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7cd3c44ec6904f83bdd86417a58f3fed : SUCCESS in 57m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0ccb79225917417fb66daafa33363614 : SUCCESS in 56m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d91160e545244deeb6cbeb9a1bae7ab6 : FAILURE in 22m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6c2b92aaee7a4ae6ba2d2432c0b45185 : SUCCESS in 1h 01m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/63b08ee656a047f7b44d999885757a40 : SUCCESS in 42m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/91ebf6aead82467c96b81b662f5b8b73 : FAILURE in 24m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5ecb15b852d8403abedf5297a43894e5 : SUCCESS in 58m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5808474f2094b4bb3711e1eb9f6e90d : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd5a5b461896479eaf0d5aa66dcd8ce2 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1743c6b3f843d4ab1c4a21d4d05024 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/894c411257044ce1be2ecef9e747c4a3 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb931b4a43ff46c6928d3ff22f43f085 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6e68a8c234f46efbf29e82025d45d36 : SUCCESS in 23m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5363b59fec714e09b5765fa07c460b5d : SUCCESS in 53m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf0a01a2d21047a1abc7f4fb39ae9cea : SUCCESS in 44m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bd15dfa888b243ab8e2efd5047100f54 : SUCCESS in 54m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c27a4f7a1c114a13bdf172b1ba59235e : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":9},{"id":"4e38aec88565ed39336aa3e3a6d37621f12a3c58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 12:04:50.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"07d958a95632f0a2589b94aee19ba97274a6374d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 12:05:17.000000000","message":"Patch Set 10: Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"47e36d2c6deaaba1cd65d030fc2d1382d2cbfd0b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 12:06:10.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0bea20d0a6af0d0ef805483757b2adc602bf58c1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-22 12:08:21.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6ff3907b6db07ec2e129c92240398c1d8da51bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 13:20:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9f5ae5cd1ebd4216b3457d970f316fcc : SUCCESS in 31m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/69ae9f56071c4783840406051c10b6d8 : SUCCESS in 48m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c433d4a74ac44b9ca11f3dfaee49a17a : SUCCESS in 1h 13m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0d0ebb55e90b4e8399643d769e579206 : SUCCESS in 1h 09m 54s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e1b0c412157d452f8bb83b7c7a8b8316 : SUCCESS in 1h 13m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e13e6c28a6f8410d94a4346d4adc0560 : SUCCESS in 41m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b463bad52b9149a99cd363b64ddf0015 : FAILURE in 22m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f0127cab93c145b48c2c430b6e793fa3 : SUCCESS in 52m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ddbb15cbd9949bb902a82745cc9756e : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efb36175a93c4ae7aa3d3bca147f8a5e : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5320df5414014ae684f133d00a23d2dd : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79c4861a9a6f4529af39be234aa10074 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc549052f084cb58437a4f63a3e9219 : SUCCESS in 6m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e616f726c42a4c4eab75222660817997 : SUCCESS in 25m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/34c1357c3c984d28bb0a2cc01c253bb5 : SUCCESS in 39m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9b648bef10f4ffe89b8d42ae9f64f40 : SUCCESS in 52m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a317ef78d6e94fccb4f37b754a36216a : SUCCESS in 50m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/34d90edb5376434d9d127d453d5762e5 : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":10},{"id":"444ea3d5a613c099bae54e092ab3ee8238dd9533","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 13:51:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f0e2eb69d6947c3bb653f438abc9aeb : SUCCESS in 40m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fae0e0d37b1c4b0aaf1d4b7a1249734c : FAILURE in 1h 41m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7cf03a242fbd1cea8a242c08c6c4e0f7ba800469","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-23 12:38:14.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"ca73e6fedeb792a5fc0f7c71bcbba2650f4b328d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-23 18:30:18.000000000","message":"Patch Set 10: Backport-Candidate+2","accounts_in_message":[],"_revision_number":10},{"id":"d2d0987d99d9ab073f64bd1bbf9136d8923a65af","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-04-26 06:47:49.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"b0077881ed385e39b90d2fc8942975a369b90c79","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-26 08:59:23.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)\n\nand add reno that vmware-nsx is no more","accounts_in_message":[],"_revision_number":10},{"id":"4f6a8802c462610771f7d29157bc655e224b8c79","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-27 09:01:26.000000000","message":"Patch Set 10:\n\n\u003e Patch Set 10: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e and add reno that vmware-nsx is no more\n\nThis patch gets reverted later in a cycle which would remove such reno.","accounts_in_message":[],"_revision_number":10},{"id":"df3d3955bac24f21c253006cfe76cd518a525a33","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-27 15:50:11.000000000","message":"Patch Set 10:\n\nlet\u0027s redo on stable/wallaby after plugins drop merges","accounts_in_message":[],"_revision_number":10},{"id":"08efc470d24db2a3017a21a280805cb0b428415e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-28 13:32:52.000000000","message":"Uploaded patch set 11: New patch set was added with same tree, parent, and commit message as Patch Set 10.","accounts_in_message":[],"_revision_number":11},{"id":"a499dbdc355acb0a8996ecc0d086b4ceb4956558","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-28 13:36:49.000000000","message":"Patch Set 11:\n\nhttps://review.opendev.org/c/openstack/kolla/+/788490 - stable/wallaby one, will abandon this one.","accounts_in_message":[],"_revision_number":11},{"id":"4e5fdc3448dbbfc2aac2b0657ceac9b71ce1eaad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 14:55:30.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fa9b175db71e488a8fab0e5cb8e6f4fd : SUCCESS in 52m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d806a9572ec6468e8d16b96505f876a8 : SUCCESS in 58m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1849470df263468783010517a3bd71d4 : SUCCESS in 52m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ba2ec51209a841ad965b8449e0dcb685 : SUCCESS in 55m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/86e2f7552e874dcd83ae8f0c2080efde : SUCCESS in 1h 02m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/51648c4fdef546d58e3b66e5df8b686c : SUCCESS in 26m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b0284111600f4a569992f5bae51f53d9 : SUCCESS in 24m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9c074086af96470b904a94ee5c59c22f : SUCCESS in 48m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfcff120c8ad474eb56cef54cf9d1044 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef41c3d532cd467887a6a13a0786062c : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa8cdeccdf804392a6be57450d917ea6 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c35fab7452b5472da04d2a7232f40ce4 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c764a9e41167472b81ede51e334ef69a : SUCCESS in 6m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef0906c7b4fc40a7b2802cd2b1b15428 : SUCCESS in 23m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd03fd97d9c54b7583ac623524d564e7 : SUCCESS in 38m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cdedb3b661ba44b1a1b4c8fb93579007 : SUCCESS in 50m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/904b10072a12455d9f9da988577b0814 : SUCCESS in 48m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9355e9af16114c928bccf37cb5e600ef : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":11},{"id":"f8f8252dc072c53e92c457e88c363741e9eed6c6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 15:06:12.000000000","message":"Abandoned\n\nas planned","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"632e91bf890d9a96b761afc63f0a923facd55fe5","revisions":{"c0ff87b96e307c9a8800cd0ec2fbcbc8c23a5197":{"kind":"REWORK","_number":1,"created":"2021-04-09 11:21:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/785622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/1"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 11:04:28.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 11:21:15.000000000","tz":120},"subject":"[WIP] centos: switch to RDO final release of Wallaby","message":"[WIP] centos: switch to RDO final release of Wallaby\n\nRDO is close to Wallaby release so it is time to move and use it instead\nof delorean repos.\n\nv0409:\n\n- rdo-release-wallaby package is temporary as it will be replaced with\n  centos-release-openstack-wallaby once final packages will be built\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0ff87b96e307c9a8800cd0ec2fbcbc8c23a5197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0ff87b96e307c9a8800cd0ec2fbcbc8c23a5197"}]},"branch":"refs/heads/master"},"a6d76a8061bc5f62ba6a812f7f58b46bbb5b85e3":{"kind":"REWORK","_number":2,"created":"2021-04-09 11:34:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/785622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/2"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 11:04:28.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 11:34:45.000000000","tz":120},"subject":"[WIP] centos: switch to RDO final release of Wallaby","message":"[WIP] centos: switch to RDO final release of Wallaby\n\nRDO is close to Wallaby release so it is time to move and use it instead\nof delorean repos.\n\nv0409:\n\n- rdo-release-wallaby package is temporary as it will be replaced with\n  centos-release-openstack-wallaby once final packages will be built\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6d76a8061bc5f62ba6a812f7f58b46bbb5b85e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6d76a8061bc5f62ba6a812f7f58b46bbb5b85e3"}]},"branch":"refs/heads/master"},"8c5a92c0bd62a440ebd9c75dc96cee8e81aa3e00":{"kind":"REWORK","_number":3,"created":"2021-04-09 12:00:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/785622/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/3"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 11:04:28.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 11:59:53.000000000","tz":120},"subject":"[WIP] centos: switch to RDO final release of Wallaby","message":"[WIP] centos: switch to RDO final release of Wallaby\n\nRDO is close to Wallaby release so it is time to move and use it instead\nof delorean repos.\n\nv0409:\n\n- rdo-release-wallaby package is temporary as it will be replaced with\n  centos-release-openstack-wallaby once final packages will be built\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c5a92c0bd62a440ebd9c75dc96cee8e81aa3e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c5a92c0bd62a440ebd9c75dc96cee8e81aa3e00"}]},"branch":"refs/heads/master"},"6ed181dd63d34d83e6ec5052f9706f1fe4684249":{"kind":"REWORK","_number":4,"created":"2021-04-09 12:51:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/785622/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/4"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 11:04:28.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 12:51:33.000000000","tz":120},"subject":"[WIP] centos: switch to RDO final release of Wallaby","message":"[WIP] centos: switch to RDO final release of Wallaby\n\nRDO is close to Wallaby release so it is time to move and use it instead\nof delorean repos.\n\nv0409:\n\n- rdo-release-wallaby package is temporary as it will be replaced with\n  centos-release-openstack-wallaby once final packages will be built\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ed181dd63d34d83e6ec5052f9706f1fe4684249"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ed181dd63d34d83e6ec5052f9706f1fe4684249"}]},"branch":"refs/heads/master"},"4d7e2a4737c1c2b75fdc6997c9497b18e1abe6e6":{"kind":"REWORK","_number":5,"created":"2021-04-09 15:02:01.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/785622/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/5"}}},"commit":{"parents":[{"commit":"d50832066d090ff672b54e4170a56f5584d31c14","subject":"Merge \"[CI] Publish images to quay.io on a daily basis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d50832066d090ff672b54e4170a56f5584d31c14"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 11:04:28.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-09 14:54:35.000000000","tz":120},"subject":"[WIP] centos: switch to RDO final release of Wallaby","message":"[WIP] centos: switch to RDO final release of Wallaby\n\nRDO is close to Wallaby release so it is time to move and use it instead\nof delorean repos.\n\nv0409:\n\n- rdo-release-wallaby package is temporary as it will be replaced with\n  centos-release-openstack-wallaby once final packages will be built\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d7e2a4737c1c2b75fdc6997c9497b18e1abe6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d7e2a4737c1c2b75fdc6997c9497b18e1abe6e6"}]},"branch":"refs/heads/master"},"8709882f267e79c1d37f012bacc6670201a56078":{"kind":"REWORK","_number":6,"created":"2021-04-12 13:17:44.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/785622/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/6"}}},"commit":{"parents":[{"commit":"a90d09450eeb3f6893bcc633167f9c4a6494f88f","subject":"centos: enable repos where needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a90d09450eeb3f6893bcc633167f9c4a6494f88f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-04-12 12:32:47.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-04-12 13:17:13.000000000","tz":120},"subject":"[WIP] centos: switch to RDO final release of Wallaby","message":"[WIP] centos: switch to RDO final release of Wallaby\n\nRDO is close to Wallaby release so it is time to move and use it instead\nof delorean repos.\n\nv0412:\n\n- rdo-release-wallaby package is temporary as it will be replaced with\n  centos-release-openstack-wallaby once final packages will be built\n- image enablement moved to separate patch\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8709882f267e79c1d37f012bacc6670201a56078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8709882f267e79c1d37f012bacc6670201a56078"}]},"branch":"refs/heads/master"},"5b2d0534fdc4ed30dd8f5feaf484343fa8b6ab7a":{"kind":"REWORK","_number":7,"created":"2021-04-12 14:04:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/785622/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/7"}}},"commit":{"parents":[{"commit":"a90d09450eeb3f6893bcc633167f9c4a6494f88f","subject":"centos: enable repos where needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a90d09450eeb3f6893bcc633167f9c4a6494f88f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-04-12 12:32:47.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-04-12 14:03:53.000000000","tz":120},"subject":"[WIP] centos: switch to RDO final release of Wallaby","message":"[WIP] centos: switch to RDO final release of Wallaby\n\nRDO is close to Wallaby release so it is time to move and use it instead\nof delorean repos.\n\nv0412:\n\n- rdo-release-wallaby package is temporary as it will be replaced with\n  centos-release-openstack-wallaby once final packages will be built\n- image enablement moved to separate patch\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b2d0534fdc4ed30dd8f5feaf484343fa8b6ab7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b2d0534fdc4ed30dd8f5feaf484343fa8b6ab7a"}]},"branch":"refs/heads/master"},"3165c91c58ed9d669696305f94d488271e4781d9":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-04-14 16:41:52.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/785622/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/8"}}},"commit":{"parents":[{"commit":"3330f624f0355b6d53d1e4f34b5bad72624f0a1d","subject":"Merge \"deprecate monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3330f624f0355b6d53d1e4f34b5bad72624f0a1d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-04-12 12:32:47.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-14 16:40:44.000000000","tz":120},"subject":"[WIP] centos: switch to RDO final release of Wallaby","message":"[WIP] centos: switch to RDO final release of Wallaby\n\nRDO is close to Wallaby release so it is time to move and use it instead\nof delorean repos.\n\nv0412:\n\n- rdo-release-wallaby package is temporary as it will be replaced with\n  centos-release-openstack-wallaby once final packages will be built\n- image enablement moved to separate patch\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3165c91c58ed9d669696305f94d488271e4781d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3165c91c58ed9d669696305f94d488271e4781d9"}]},"branch":"refs/heads/master"},"7985bd22a9e8e99b608ee30057ed66cfae27463c":{"kind":"REWORK","_number":9,"created":"2021-04-22 10:51:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/785622/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/9"}}},"commit":{"parents":[{"commit":"a18056fefb7fda569febbfbbf98bb60f25f8301d","subject":"Merge \"[release] Use RDO\u0027s Wallaby Delorean packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a18056fefb7fda569febbfbbf98bb60f25f8301d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-04-12 12:32:47.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-22 10:49:04.000000000","tz":120},"subject":"centos: switch to RDO final release of Wallaby","message":"centos: switch to RDO final release of Wallaby\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7985bd22a9e8e99b608ee30057ed66cfae27463c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7985bd22a9e8e99b608ee30057ed66cfae27463c"}]},"branch":"refs/heads/master"},"27eb322f09b4dcf1d55fd1e35a46e5a225676cf7":{"kind":"REWORK","_number":10,"created":"2021-04-22 12:04:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/22/785622/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/10"}}},"commit":{"parents":[{"commit":"940baeff65b8e75fed8a77e56cb385920d72ed01","subject":"Merge \"[TrivialFix] Remove unused aarch64 repo mentions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/940baeff65b8e75fed8a77e56cb385920d72ed01"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-04-12 12:32:47.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-22 12:04:36.000000000","tz":120},"subject":"centos: switch to RDO final release of Wallaby","message":"centos: switch to RDO final release of Wallaby\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27eb322f09b4dcf1d55fd1e35a46e5a225676cf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27eb322f09b4dcf1d55fd1e35a46e5a225676cf7"}]},"branch":"refs/heads/master"},"632e91bf890d9a96b761afc63f0a923facd55fe5":{"kind":"NO_CHANGE","_number":11,"created":"2021-04-28 13:32:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/785622/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/785622/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/785622/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/785622/11"}}},"commit":{"parents":[{"commit":"940baeff65b8e75fed8a77e56cb385920d72ed01","subject":"Merge \"[TrivialFix] Remove unused aarch64 repo mentions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/940baeff65b8e75fed8a77e56cb385920d72ed01"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-04-12 12:32:47.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-28 13:32:49.000000000","tz":120},"subject":"centos: switch to RDO final release of Wallaby","message":"centos: switch to RDO final release of Wallaby\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/632e91bf890d9a96b761afc63f0a923facd55fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/632e91bf890d9a96b761afc63f0a923facd55fe5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
