)]}'
{"id":"openstack%2Fkolla~786390","triplet_id":"openstack%2Fkolla~master~I499251d13ca77aea2215f8a879b7bb8dd51f1c9f","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I499251d13ca77aea2215f8a879b7bb8dd51f1c9f","subject":"rabbitmq: Move to packagecloud","status":"MERGED","created":"2021-04-15 11:17:34.000000000","updated":"2021-04-22 08:47:35.000000000","submitted":"2021-04-21 18:06:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":17,"has_review_started":true,"submission_id":"786390","meta_rev_id":"14ccea57288df3dddb8f73e7a146a9d9c9ba10d9","_number":786390,"virtual_id_number":786390,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-21 18:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-04-21 16:39:08.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-21 16:44:03.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-21 16:22:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-04-21 16:39:08.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-04-22 08:47:35.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-04-22 08:47:35.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-22 08:47:35.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-15 12:30:37.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-15 13:33:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-15 18:56:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-20 16:47:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-04-21 16:44:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"dc4c26bd92c5af1abc123528b2932c7ba0842b17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-15 11:17:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff4f11453caf27008fbedc8636d74da48a7bd772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-15 12:25:32.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b4b7310281668ba5d82b9261f348603178c2f664","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-15 12:30:37.000000000","message":"Patch Set 2: Code-Review+1 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"209e693162644536f106aab56f4c6d6b9fe92357","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-15 12:32:38.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a70011d1d7990988b3b3204eda03f048fa45ac96","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-15 12:34:05.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"39587abcc50df0bc097870af2b8483998a221832","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-15 12:43:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"897d20bcd7841e12c22c656a6f33b1aae26980b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 13:33:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/55ba741d737b4f70a418a6cabc3536b2 : SUCCESS in 21m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e5fe5fa4027b4c8fa06f1c7c2d26e8f2 : SUCCESS in 44m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cf1badd3b5324ed08f80df9d2219172e : SUCCESS in 30m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5e34f81db0954cf39cc1dcfc1fda2f81 : SUCCESS in 21m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9b3acf6800df4d7086aa121488a0e681 : SUCCESS in 1h 00m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ffecc1bf93e43668b58945daa8f06c5 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52b5d9bb5044420cabc2bed32903762d : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9684ac93faa34d9d98145f40231da7ee : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/088884cc8abf405fa8729f37f0479ef4 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e7bd0cfac64716ace522075b0d3d49 : SUCCESS in 6m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0eacb52eb5ad4cc1ab1425d137b78f58 : SUCCESS in 19m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b168ef7c9d54ec08f201358726f73e3 : SUCCESS in 42m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d14760346a394dab8728bcbc77db6fc8 : SUCCESS in 43m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1cb559e51f664063b16bd62a58385695 : SUCCESS in 53m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8207b3bb8bbb48479d8e6b3830f816b1 : SUCCESS in 49m 54s","accounts_in_message":[],"_revision_number":3},{"id":"6d8a8f02c8f2d6d59547db4388e693895637f75f","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-15 13:59:28.000000000","message":"Removed Code-Review+2 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"58f50918166f5ba898d3138016dadf1f3d1eebf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 14:55:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2365df1390d540d8b95e5e6f446d64e0 : SUCCESS in 52m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/477bd4fa670c4e7cacbb9a3013afcb24 : SUCCESS in 46m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a65caf72d62f4bd88f05989ceffa9314 : SUCCESS in 1h 43m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"38f6300a73f27f99697a6d98d6c6b01fbd835d23","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-15 16:32:10.000000000","message":"Patch Set 3: Code-Review+2\n\nlooks like it works","accounts_in_message":[],"_revision_number":3},{"id":"a696851b988d19b2f9795a0f203d9855251d1b80","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-15 18:56:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nneeds proper pin on debuntu (it\u0027s in official rmq docs)\n\nalso, let\u0027s use cloudsmith exclusively (it can cover all distros and both projects)\n\nfinally, please add a reno\n\nremember we will be backporting this","accounts_in_message":[],"_revision_number":3},{"id":"78909b9cd736bb364212b25e1b62b724964dc53b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-15 18:57:46.000000000","message":"Patch Set 3: Backport-Candidate+2 Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"2b8c8ea5b1462f5ede25ca6ca310c1477027d544","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-15 19:02:51.000000000","message":"Patch Set 3:\n\n@Marcin: you said something about moving back to distro-provided rmq - would be worthwile to check how it behaves in deploys and upgrades in CI - I checked that, with bullseye, we would have 3.8 in all distros - that\u0027s what we wanted so we can try that path and perhaps be happier","accounts_in_message":[],"_revision_number":3},{"id":"9f751dd09f129d031e624bfe17217390cf013fca","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-15 19:38:49.000000000","message":"Patch Set 3:\n\nI am fine with both external rabbitmq and erlang as long as it is as same as possible version everywhere.\n\nIn-distro version of rabbitmq is older than external in all cases (3.8.[239]).\n\nThe other thing is with Erlang. CentOS has 22.3, Debian/bullseye has 23.2, Ubuntu has 22.2 version. We want 23.x as RabbitMQ will abandon support for Erlang 22 in May 2021.\n\nSo for now we need external RabbitMQ for all distros/archs (it is arch:all). External Erlang is needed for CentOS/Ubuntu - Debian \u0027bullseye\u0027 can use either external or distro one (\u0027buster\u0027 needs external).\n\nWe probably need to pin down rabbitmq in all branches starting with Wallaby to have 3.8.14 maximum.","accounts_in_message":[],"_revision_number":3},{"id":"927def7a5ef8f4a389ce775b86bc8a0904dc403a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-16 07:37:30.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"eaba435ab715f4afb6cb7f0a073ee6954769931c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-16 08:06:23.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"e424443803dfc1c474276030d2b2edb018c1e3c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 08:44:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c50ba2aa2750437ba350c031849e90ec : FAILURE in 24m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/be4033226a974621bdb7373e91b9d935 : FAILURE in 42m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0eb832ae7c9f49b2b12b488f9cce68fb : SUCCESS in 28m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f1a77b94fef1413298fb466906cb448f : SUCCESS in 19m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bd8e79774ad149f38e2af3c69451ebeb : SUCCESS in 55m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30bb7039b66c4607a514694c86eccdae : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6d07ca7a8ec4354958f4e2f98ac4baf : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2fda111854b442593710542d3ccd236 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3f3bc7aabaa45259a01fd36800d37dd : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7dcb040dc4e4df69226412b95889c93 : SUCCESS in 7m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/deca4ba8406341ed84560107c0e38fde : SUCCESS in 23m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc0eedc13f76401a9b9943ca184c364a : SUCCESS in 40m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/63eceb89b02540dfb7cee508b88ef6c3 : SUCCESS in 53m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3a3f08a6507c4c94b522f9f015da63e4 : SUCCESS in 54m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/14f427cf323f40028557d1a44d0002a4 : FAILURE in 1h 02m 18s","accounts_in_message":[],"_revision_number":4},{"id":"510ec26ae9c397f820785b13556b1488d7cfd77f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-16 09:14:31.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"91561a24deefc080be95abdb21abb0d8cd9a2bb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:19:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a90c1ac16088491e8f7a30e3b06a2102 : SUCCESS in 22m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/aa7cf2c3e2d44750a0f1af6cd627a3b9 : SUCCESS in 32m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/78addf5a12ba4dc0a9d9c985fae600c3 : SUCCESS in 42m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/043e39d5dd4c4a4a9a319fa6e9a2140f : SUCCESS in 18m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5e6b10aaa03d4a28b08c723ca878a818 : SUCCESS in 1h 03m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/880e2da2f9d34fccb0b83ede80297473 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fdaaacc596243c4a7feb462739d1b15 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18f4ccfb3c904fd9bbe1c32a6340acde : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f71f1feae1a4ca09fdf1551aebf8c6d : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463e0851cc334030a81cebbf223a718a : SUCCESS in 5m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37905978bfaa45c0b721bc03cfa24781 : SUCCESS in 36m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c191f513b2042c5aafb473886635a54 : SUCCESS in 30m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0d95e2a944e41fcb520e754492ffe14 : SUCCESS in 50m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/11b4aa960acc42d6b7abaf9cec2fd705 : SUCCESS in 50m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/79ab2299c4564377a1c0c27a73ed4467 : FAILURE in 1h 00m 07s","accounts_in_message":[],"_revision_number":5},{"id":"2f5c0551213c8496f2b99afb0e25f3fc42c6066e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:48:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4a67b250b15c45d48f976693c8b60a7a : SUCCESS in 1h 28m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/106c6290ff104bf1a74ac3de6198f8dd : FAILURE in 59m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/015f28e721114dfcbd91c1d53cb8042b : FAILURE in 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e4e3eec929e0df75a4a01582d1ea7674b61032fb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-19 07:32:09.000000000","message":"Patch Set 5:\n\nUpgrade CI is fried because Nova merged an X cycle change to allow only Wallaby computes... so master is Xena, not Wallaby.","accounts_in_message":[],"_revision_number":5},{"id":"ad5f7dcd60242e88a19671faab36b3ded869d5e9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-20 06:57:58.000000000","message":"Patch Set 5:\n\nRabbitMQ team suggests using Erlang from either Launchpad PPA (support for Ubuntu and Debian) or cloudsmith. \n\nPlease consider move to using PPA - we will get aarch64 packages as well.","accounts_in_message":[],"_revision_number":5},{"id":"a2889cfd141be230cc427febe6192c205f1ef776","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-20 07:56:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6aa9b04dc5c37959b51c48539d30c7a26c332233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-20 07:58:33.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"a2bf0f81d87c150f68b8967b3db0e2c5a6d93df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 08:11:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7964536173e34b82a21f54bd56339229 : FAILURE in 6m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7f5b08ada1a6444795b230f5cb75328d : FAILURE in 9m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c38a2477fa414d51a5a15683d21ed8fd : FAILURE in 6m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/83e1b754c9cc4d8f9177c5e58e87223c : FAILURE in 7m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/696742f378c54c928f1d9c35b6e681c8 : FAILURE in 7m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d093bbc8e4104273a0d4eba959a59012 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5be51c32bc94f3e804e35989ccbc8a7 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d2682faad3c4c3fa9e5e621984cac1b : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d59550793a0d4b8190926fddc49ddd78 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/685fffacdde740c08b1c787e7e79eb91 : SUCCESS in 6m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b23401e15fce43d9a430a1855fd8a6ce : FAILURE in 4m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2060311052a04657913375ab0019657d : FAILURE in 4m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a1ceb3ad3c94705ac9c4e95bbaa3c62 : FAILURE in 11m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/095e79ad7c6c4a37a5845a281057411f : FAILURE in 9m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a6da99e9da0046da84260990699c9b5c : FAILURE in 9m 25s","accounts_in_message":[],"_revision_number":7},{"id":"ec64e7111cdb68c5f422bf8cd53850c557400c12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 08:32:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b1bf504bcbf4aefb5e8a4ade45b1e60 : RETRY_LIMIT in 1m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f590a3fa2d6b470eb642041a702f7074 : FAILURE in 5m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/68dfb6c3346f47b49e493d175926faad : FAILURE in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8e11b7ca6d8ae5dbd7301807796b65b66f599d8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-20 08:35:50.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"55465717aa98c96608857f8a23505b06aed7fa2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 09:49:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7d513cb39a074f3b9556ccaf06c64f49 : SUCCESS in 25m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/621a0548f39a4f4caf671a0d98e01d29 : SUCCESS in 42m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8b39aafb289d493f84e58d7a7c4a544a : SUCCESS in 50m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dca37da97b9a452f9a0c61964a595ee9 : SUCCESS in 25m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b4ec7ab90bfa42349d976219d0a4555c : SUCCESS in 53m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c084007bca46a8a49e09ab1bbe878a : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0588356cbda441acabe9e4840d98a9ff : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc5490afe26c421ebfa498c60cf7a155 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71aaa4e33f304d5b9d6ffe6779c28e40 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9621195d84de4c4c92bf97c42bf5f096 : SUCCESS in 6m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afe0c8dfae824c2db1e64c0a3c90f8be : SUCCESS in 33m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/09a7571cd22f4876897a6048206cb773 : SUCCESS in 41m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b58e828896654093b1142be2d2b1a7a0 : SUCCESS in 56m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/989a754b24ee47caa0e5b9e9bd16f79c : SUCCESS in 51m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4331e625973c42bc9b8973001fe325ea : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":8},{"id":"944a60f8797c518fa6a61400b9bca791d1b4a1d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 10:35:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6e8c6156c8b14541ad12b4c3b3199c60 : RETRY_LIMIT in 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ea70b7379be42b7b6423eac1e4e1fe2 : SUCCESS in 1h 54m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a72716cf15d4cefa328297fd21b3668 : SUCCESS in 1h 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6933f9a1b036ae008392ec81c83ba53e4aad66a8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-20 13:52:19.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"c724f27d98f0a93c5f6c63313ec68584a0ce77d0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-20 16:42:42.000000000","message":"Patch Set 8:\n\nadd a reno for backports - the urls are changing drastically","accounts_in_message":[],"_revision_number":8},{"id":"27997e2e4b3a9b30d51e92a310edf56c2fd68178","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-20 16:44:47.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"efa2f982193005fc66debcef8d21cd7eea26c226","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-20 16:47:04.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8b94f598c19bc93e0498abe7c32b6537d06bfadd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-20 16:49:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"51dda51cb46284ed73c040ad5cdd7962e677f787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-20 16:51:19.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bc6c489ebb8d35c654507bac14e4a852ffaf6fbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-20 16:54:29.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"75411843772c1e36dc4af62987ef982cf3866de6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-20 16:54:44.000000000","message":"Patch Set 10: Code-Review-1\n\nnice reno, but comments remain","accounts_in_message":[],"_revision_number":10},{"id":"885a8c4de480da86ba4ff4b7c00d97f16c5b5606","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-20 16:55:37.000000000","message":"Patch Set 10: Code-Review+2\n\noh, that was for #9","accounts_in_message":[],"_revision_number":10},{"id":"9b85d1481fbf3e15dcf02e15327d775f264ebb02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-20 16:55:40.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"be5f4456bb6c46f8a1f9331f2ae49485ed6454d8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-20 16:57:23.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"d27994cb43d11881407ba737650fdac6c56968e7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-20 16:57:54.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"7f5955e1a70b80e70a2b10ccac87d7410874ba23","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-20 16:59:42.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"a716c4d2783309d673c3ccb153d23684a1336751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-20 17:01:43.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"45bb85f02647328dc7ea9701aff3c40976776d84","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-20 17:04:03.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"cf4491e95bf37a0564dee13e89689038f5d848b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 18:31:57.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/19a1b7625e8c40cfbe650f89bd121098 : FAILURE in 31m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/186540b0dc52414d9241e65aebbfbd2b : FAILURE in 49m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c1e85591b3444c5ca5f158782ee771df : SUCCESS in 34m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6beff2bbabe7482abba59a1a9a257502 : FAILURE in 25m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/08f8e7d350ab4525bd84f00515796876 : SUCCESS in 50m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/344e73db2e3a45328c090930a182acd1 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f20fd4061866402388d4fe63688fb04f : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34a6cf4293a445bca8cdb48f463fa701 : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b18c9dfc68104299951d3f6549469305 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5315b3c1535447d2b26290ec86342b6c : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/546def84191d477eb0282f016f0960a8 : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d40a7743bc145e1ba08aef25405ac9b : SUCCESS in 23m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dbd1c5d1fa404408b80059989fca93ef : SUCCESS in 45m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/85c8906f70d74755ad53ed8cff32a35e : SUCCESS in 55m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef7abbf082424664bd243a529e1968a2 : SUCCESS in 1h 00m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1a8d3dd3c29d4f9fa8ffe0d6e901da28 : SUCCESS in 1h 08m 52s","accounts_in_message":[],"_revision_number":12},{"id":"15216daf20f4bd2a74b840c4ad99f8464ec68cb2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-20 18:35:12.000000000","message":"Patch Set 12: -Code-Review\n\nINFO:kolla.common.utils.rabbitmq:\u001b[91mError: Failed to download metadata for repo \u0027rabbitmq_rabbitmq-erlang\u0027: repomd.xml GPG signature verification error: Bad GPG signature\n\n\nsomething wrong with the installed ones...","accounts_in_message":[],"_revision_number":12},{"id":"a351ab999dd120f270e214fde7bfe2cc7e0c42d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 19:08:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4de4be436f5d473589605f68594d8c9e : RETRY_LIMIT in 1m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f98e59ed9e84409396f4c459645006d0 : SUCCESS in 1h 28m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e02a187828ae4c07a396303fdc603d76 : SUCCESS in 1h 37m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"243d3614e5522c54a48f8b6583849daf88d8bb4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-21 08:21:29.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"252f1788eb2aa7be4959236dc80ee517f12f6b43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 09:26:51.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/68582f92bd414f5d85d9a0c5b08c5d9e : SUCCESS in 24m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4bd25ecd2b6248fca21c3fd1f67adc02 : SUCCESS in 40m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/242732c3a1724234a7ddbbb487274547 : SUCCESS in 28m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bcecbca872c7406da39ac66a4d5f66aa : FAILURE in 18m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2bc41373ba7d4efb815ff42353638bf7 : SUCCESS in 40m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1cc63bee106499b8c1fc50072f5eb38 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37045912508d44b7b92373c597ed885b : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edf7322793004f4b8393119d08055348 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dcbc8c3dc0c43a79245dfee32d44394 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1213840bc9744bfdba66f5e549fc700b : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74cf77dd7a2f4a4ca2cc62314fd26ba3 : SUCCESS in 5m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2d402fbfcf94f9e9f362bcee994623f : SUCCESS in 22m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/85620b97614641978cd498e126f160e1 : SUCCESS in 34m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9381055110444fe39096b818919d82cc : SUCCESS in 51m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6a88f1acf3a94746953a4ddadc42e618 : SUCCESS in 54m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a2429bd7c41640b2883e60a429c147ac : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":13},{"id":"a5922ce15824374574ff0ed6850c243de4ec1a17","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-21 09:47:09.000000000","message":"Patch Set 13: Code-Review+2\n\nalso fixes ubuntu/aarch64","accounts_in_message":[],"_revision_number":13},{"id":"1fc6f4903c57385638da10b43319b7f3b1023ff7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 10:05:44.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/30becf115da04c91994da1d005ffb587 : RETRY_LIMIT in 1m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebb11f74536546939d031a06f03455cd : SUCCESS in 51m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43b8a7e4c8a54c0ca6f2617bba2b6c37 : SUCCESS in 1h 36m 32s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"f514a7d9685da2c2b7b1b7d868d9762954b906c8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-21 11:41:44.000000000","message":"Patch Set 13: Code-Review+1\n\nI am generally fine with this but then what is the point of us downloading the key if it cannot be used and must be redownloaded? Perhaps we need something similar to what Marcin is doing for Debuntu - download the keys, store them in the right place, and then link to them in repo files? WDYT?","accounts_in_message":[],"_revision_number":13},{"id":"a726e65ca06eda18a932bba3be3f8710fd95d781","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-21 11:42:09.000000000","message":"Patch Set 13:\n\nI meant ~\u003e https://review.opendev.org/c/openstack/kolla/+/784923","accounts_in_message":[],"_revision_number":13},{"id":"f5e7bc7ec56acebcf3127c4d28e3432b3b6bc525","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-21 11:55:03.000000000","message":"Patch Set 13:\n\nThe common way of doing stuff for CentOS was \"fetch repo file as provided, disable it, use\" - no edits other than disabling repo. And then importing needed keys.\n\nSo fetch repo files as they are given by upstream, disable, add keys. No extra mangling needed.\n\nRPM files are signed and we need those keys.","accounts_in_message":[],"_revision_number":13},{"id":"a7e3217b7a98b89a7e112980128f05534f6d1c66","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 12:48:57.000000000","message":"Patch Set 13:\n\nSome context here in the switch from 3.7 in packagecloud to 3.8 in bintray: https://review.opendev.org/c/openstack/kolla/+/694799/37/docker/base/rabbitmq_rabbitmq-server.repo","accounts_in_message":[],"_revision_number":13},{"id":"fffc200c55c6fa7757cae97691a3691f02cdc10b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-21 13:20:05.000000000","message":"Patch Set 13:\n\n\u003e Patch Set 13:\n\u003e \n\u003e Some context here in the switch from 3.7 in packagecloud to 3.8 in bintray: https://review.opendev.org/c/openstack/kolla/+/694799/37/docker/base/rabbitmq_rabbitmq-server.repo\n\nSo I guess just disable the repo_gpgcheck and have it clean. Then we don\u0027t need more gpg keys either! Win-win.","accounts_in_message":[],"_revision_number":13},{"id":"7aae35bf244409dccb4de787048ee341e54036ef","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-21 14:30:21.000000000","message":"Patch Set 13:\n\n\u003e Patch Set 13:\n\u003e \n\u003e \u003e Patch Set 13:\n\u003e \u003e \n\u003e \u003e Some context here in the switch from 3.7 in packagecloud to 3.8 in bintray: https://review.opendev.org/c/openstack/kolla/+/694799/37/docker/base/rabbitmq_rabbitmq-server.repo\n\u003e \n\u003e So I guess just disable the repo_gpgcheck and have it clean. Then we don\u0027t need more gpg keys either! Win-win.\n\nOk, let me do that.","accounts_in_message":[],"_revision_number":13},{"id":"318f6a2853f4008493fb12fc131597f808c2a508","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-21 14:31:54.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"1d4ebfe7b24cb40869b7ff9627136ba6bbe7e022","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-21 15:01:24.000000000","message":"Patch Set 14:\n\n(5 comments)\n\nquestions, questions","accounts_in_message":[],"_revision_number":14},{"id":"a9c7d97ff1e2403be3c89a5a4a7bc80105b8f6e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-21 15:08:45.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"29d27964dcfa8a411a0e8b1243972b75317be96b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-21 15:10:09.000000000","message":"Patch Set 15:\n\n(5 comments)","accounts_in_message":[],"_revision_number":15},{"id":"bddfb6ff9dada8608a1f7c2aee25369ee1f1827d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-21 16:22:19.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"6faca2b6ff9d88a8b84318adcf08bc4aa05dc3bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 16:22:37.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9ec6e10eaeec4bfda7271a8c0d4ad0c3 : SUCCESS in 24m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3e1e12309ab34db6bacbf48fbe6dc3aa : SUCCESS in 1h 10m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/414050ce63e34e688cf1391b35fe236d : SUCCESS in 1h 05m 35s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/23de4c73b3bb470ab801b832f26449a1 : SUCCESS in 57m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b7b4986d155e410c8bcb1416d0c618ae : SUCCESS in 1h 05m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/02f6e7d97e4c4f8ca6a0fc347340da29 : SUCCESS in 34m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8cf194d241cc49fd9b9f8e365a2ca7ac : FAILURE in 23m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c0760f3cfc6f4eb6960c5b41e41563f5 : SUCCESS in 55m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c69c7057e6a472cbc45afaf3084b92a : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5ac52b5b8594cbe90dca08c7d3d7b19 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf083b54b5724aab9f84ef7252ef493c : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2abe1a6464f74def9e286212a62d1185 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87e0015f620e493daefc5b7227798c27 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cab4e75d03949c1a458a3e348fa6ab3 : SUCCESS in 4m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7abfc02c52fc4f98a4f7953d8393d6c8 : SUCCESS in 30m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/48621b1978f24167a1dadef3c33a9f6c : SUCCESS in 41m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/acec4ac674cf4466b501c2a9f0b63083 : SUCCESS in 58m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94ad44fb71e94098a4b7de677c968c6a : SUCCESS in 51m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c958c1d60d964b70b6265db259c16416 : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":15},{"id":"df770ffffd6d72c23fa8a386c1def048184c8b18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 16:32:37.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aec0b99762624fc0bddd55eeb8a78df0 : SUCCESS in 47m 27s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe89ad2e12c749fe9b20f5a7812a04b1 : SUCCESS in 1h 17m 56s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"90deec1202da8f841c563cfce54243f2847cea21","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-21 16:39:08.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"9be33126608a079e8bff8de3fef139df34b48a55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 16:39:26.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"ccd68218a489ad425a980158d618ad6ac9b64c12","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 16:44:03.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"21493a8edfddf9bd62b8a8cb95d7a7ad299397c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 18:06:24.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6cd0c3571e1248c3b2ddef0005151042 : SUCCESS in 43m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/086f8131ada94f25aaacefba349df59a : SUCCESS in 1h 11m 06s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/231d9ba2195543158b7869141dac2bbc : SUCCESS in 1h 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ccbe7ede014340a727a40ab3a99208 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ee3966e61134a2ca1fa4255b84ca3e7 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3f5dd0b7401485c94e220122ece7309 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db12192cf9174d8e8cd54c7846e5e72e : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fe4766e7c8d48b19ce5ab1143a2f8ed : SUCCESS in 4m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c5eaf1f05044e40846ae9925ece2c96 : SUCCESS in 37m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2741db473f964d3f84bb913b73a7437e : SUCCESS in 48m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4cdd55a610544a2aa5cf48319baaafd6 : SUCCESS in 54m 38s","accounts_in_message":[],"_revision_number":15},{"id":"7958f656ca0dd829b3bf16ac071e9c6d99d814dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 18:06:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"cf5cfb8a51d9a906ea04e373629132088a453b0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 18:10:47.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ecdbbadd63948ff95f2362f1d1ae2df : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eecd3eb8d7e40be94d336038e93b215 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":15},{"id":"e12d23cd4c0848119dee3b00e10178b6de9c27b3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-22 07:58:00.000000000","message":"Patch Set 15:\n\nall right, we got this in *but* it introduced a \"slight\" regression in that we are now rolling with erlang and rabbitmq versions - I doubt we want to release that","accounts_in_message":[],"_revision_number":15},{"id":"e01e8fa51105c74f91569a9964cf3b9456f2d7e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-22 08:20:54.000000000","message":"Patch Set 15:\n\n\u003e Patch Set 15:\n\u003e \n\u003e all right, we got this in *but* it introduced a \"slight\" regression in that we are now rolling with erlang and rabbitmq versions - I doubt we want to release that\n\nYou mean we may get a major/minor upgrade outside of our control?","accounts_in_message":[],"_revision_number":15},{"id":"136c3c09ddc3365796de2cd7bba92f83e7ae9e97","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-22 08:47:35.000000000","message":"Patch Set 15:\n\n\u003e Patch Set 15:\n\u003e \n\u003e \u003e Patch Set 15:\n\u003e \u003e \n\u003e \u003e all right, we got this in *but* it introduced a \"slight\" regression in that we are now rolling with erlang and rabbitmq versions - I doubt we want to release that\n\u003e \n\u003e You mean we may get a major/minor upgrade outside of our control?\n\nYes, discussion on IRC and the whiteboard.","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"641b936bc7fd33203a80bb3827fce7c44d010476","revisions":{"1d7edb6d791df84b596b25606ce1ea2522741e94":{"kind":"REWORK","_number":1,"created":"2021-04-15 11:17:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/1"}}},"commit":{"parents":[{"commit":"3330f624f0355b6d53d1e4f34b5bad72624f0a1d","subject":"Merge \"deprecate monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3330f624f0355b6d53d1e4f34b5bad72624f0a1d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:17:32.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d7edb6d791df84b596b25606ce1ea2522741e94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d7edb6d791df84b596b25606ce1ea2522741e94"}]},"branch":"refs/heads/master"},"f808a48a613185968a6052102f0266d0f5ea8afc":{"kind":"REWORK","_number":2,"created":"2021-04-15 12:25:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/2"}}},"commit":{"parents":[{"commit":"3330f624f0355b6d53d1e4f34b5bad72624f0a1d","subject":"Merge \"deprecate monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3330f624f0355b6d53d1e4f34b5bad72624f0a1d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 12:25:29.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f808a48a613185968a6052102f0266d0f5ea8afc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f808a48a613185968a6052102f0266d0f5ea8afc"}]},"branch":"refs/heads/master"},"f14306dbb64ee435bbaeb4bcf23881cec943dba8":{"kind":"REWORK","_number":3,"created":"2021-04-15 12:32:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/3"}}},"commit":{"parents":[{"commit":"3330f624f0355b6d53d1e4f34b5bad72624f0a1d","subject":"Merge \"deprecate monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3330f624f0355b6d53d1e4f34b5bad72624f0a1d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 12:32:34.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f14306dbb64ee435bbaeb4bcf23881cec943dba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f14306dbb64ee435bbaeb4bcf23881cec943dba8"}]},"branch":"refs/heads/master"},"f66f00dd64bf711e837531ada06c4bf0c4dd852f":{"kind":"REWORK","_number":4,"created":"2021-04-16 07:37:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/4"}}},"commit":{"parents":[{"commit":"3330f624f0355b6d53d1e4f34b5bad72624f0a1d","subject":"Merge \"deprecate monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3330f624f0355b6d53d1e4f34b5bad72624f0a1d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-16 07:33:52.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f66f00dd64bf711e837531ada06c4bf0c4dd852f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f66f00dd64bf711e837531ada06c4bf0c4dd852f"}]},"branch":"refs/heads/master"},"e95336b0fd478775ad8388a78c7bf65f233c899c":{"kind":"REWORK","_number":5,"created":"2021-04-16 09:14:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/5"}}},"commit":{"parents":[{"commit":"3330f624f0355b6d53d1e4f34b5bad72624f0a1d","subject":"Merge \"deprecate monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3330f624f0355b6d53d1e4f34b5bad72624f0a1d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-16 09:14:14.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e95336b0fd478775ad8388a78c7bf65f233c899c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e95336b0fd478775ad8388a78c7bf65f233c899c"}]},"branch":"refs/heads/master"},"63c02a8d0ccc5afe2b7d825761d5a3738a6e71b7":{"kind":"REWORK","_number":6,"created":"2021-04-20 07:56:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/6"}}},"commit":{"parents":[{"commit":"3330f624f0355b6d53d1e4f34b5bad72624f0a1d","subject":"Merge \"deprecate monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3330f624f0355b6d53d1e4f34b5bad72624f0a1d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-20 07:56:00.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c02a8d0ccc5afe2b7d825761d5a3738a6e71b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c02a8d0ccc5afe2b7d825761d5a3738a6e71b7"}]},"branch":"refs/heads/master"},"d2361458b3d429c80389fa517aa530aa7f2a189e":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-04-20 07:58:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/7"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-20 07:58:33.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2361458b3d429c80389fa517aa530aa7f2a189e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2361458b3d429c80389fa517aa530aa7f2a189e"}]},"branch":"refs/heads/master","description":"Rebase"},"1b3b8db46412b37a590f15ec93af1c038087c940":{"kind":"REWORK","_number":8,"created":"2021-04-20 08:35:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/8"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-20 08:35:46.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b3b8db46412b37a590f15ec93af1c038087c940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b3b8db46412b37a590f15ec93af1c038087c940"}]},"branch":"refs/heads/master"},"b7b01bc48f224b663f87d450dcf670397a036fbd":{"kind":"REWORK","_number":9,"created":"2021-04-20 16:51:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/9"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-20 16:51:12.000000000","tz":120},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7b01bc48f224b663f87d450dcf670397a036fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7b01bc48f224b663f87d450dcf670397a036fbd"}]},"branch":"refs/heads/master"},"6dfbdfd7380a8c4a7a303f5144eef2c0fca79499":{"kind":"REWORK","_number":10,"created":"2021-04-20 16:54:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/10"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-20 16:54:21.000000000","tz":120},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6dfbdfd7380a8c4a7a303f5144eef2c0fca79499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6dfbdfd7380a8c4a7a303f5144eef2c0fca79499"}]},"branch":"refs/heads/master"},"53f97555c7f5c523e198f11aee8e3f7fb6da7239":{"kind":"REWORK","_number":11,"created":"2021-04-20 16:55:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/11"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-20 16:55:32.000000000","tz":120},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53f97555c7f5c523e198f11aee8e3f7fb6da7239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53f97555c7f5c523e198f11aee8e3f7fb6da7239"}]},"branch":"refs/heads/master"},"57223b128b03160bf27858b1f386c59405c641a4":{"kind":"REWORK","_number":12,"created":"2021-04-20 17:01:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/12"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-20 17:01:36.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57223b128b03160bf27858b1f386c59405c641a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57223b128b03160bf27858b1f386c59405c641a4"}]},"branch":"refs/heads/master"},"22083f28c5e37b2e73b008ad24236ef54d8d3aa1":{"kind":"REWORK","_number":13,"created":"2021-04-21 08:21:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/13"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-21 08:21:27.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22083f28c5e37b2e73b008ad24236ef54d8d3aa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22083f28c5e37b2e73b008ad24236ef54d8d3aa1"}]},"branch":"refs/heads/master"},"5926b00c6c50e40405f9e8bb8787523d6d66f558":{"kind":"REWORK","_number":14,"created":"2021-04-21 14:31:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/14"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-21 14:31:48.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5926b00c6c50e40405f9e8bb8787523d6d66f558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5926b00c6c50e40405f9e8bb8787523d6d66f558"}]},"branch":"refs/heads/master"},"641b936bc7fd33203a80bb3827fce7c44d010476":{"kind":"REWORK","_number":15,"created":"2021-04-21 15:08:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/786390/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/786390/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/786390/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/786390/15"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-15 11:14:11.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-21 15:08:34.000000000","tz":0},"subject":"rabbitmq: Move to packagecloud","message":"rabbitmq: Move to packagecloud\n\nbintray is being shut down 1st of May 2021, rabbitmq docs direct\nusers to packagecloud - so let\u0027s move.\n\nChange-Id: I499251d13ca77aea2215f8a879b7bb8dd51f1c9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/641b936bc7fd33203a80bb3827fce7c44d010476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/641b936bc7fd33203a80bb3827fce7c44d010476"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
