)]}'
{"id":"openstack%2Fkolla~786719","triplet_id":"openstack%2Fkolla~master~I6b8935b16390a301602218ce35edfe132b140a4c","project":"openstack/kolla","branch":"master","topic":"stop-publishing-on-release","hashtags":[],"change_id":"I6b8935b16390a301602218ce35edfe132b140a4c","subject":"[CI] Drop support for publishing versioned image tags","status":"MERGED","created":"2021-04-16 19:54:53.000000000","updated":"2021-09-08 20:01:47.000000000","submitted":"2021-09-08 20:00:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786719-stop-publishing-on-release","meta_rev_id":"77ff757cd2dabcf629c3f36e94a6c7c744be5f89","_number":786719,"virtual_id_number":786719,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-08 20:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-08 18:42:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-08 20:01:47.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-08 18:42:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-16 21:09:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-19 10:47:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-08 13:59:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-09-08 18:42:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"6e20c474e3bf51e0dcf45413e0f2e2f461e4e917","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-16 19:54:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2313cb98f86a2e41d61aeec1cef28997979377c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 21:09:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7023ac648ad540148b9959ccab6ccf62 : SUCCESS in 23m 28s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6ad0ccef7d1e4f039ab506fe62230fc0 : SUCCESS in 34m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8fa033477f8a4de18368ec31a7fd40fc : SUCCESS in 1h 13m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/813514a5e65149e68cfd5237909997f6 : SUCCESS in 17m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a1a1fe3d61d740d995b05dcf5ce9f306 : SUCCESS in 56m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7e3795b62044008a55bb12f3bf7d2a3 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/645d3783e93245e3b49c27bd1560c9ab : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf79e56d875247c9a2520b28eccf3082 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbfdf93038fa481a8f4716a57fafda82 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a93cd76c1c466b8606ba2172c4c595 : SUCCESS in 7m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83c5353fea0a4095ae2803a8896b20b5 : SUCCESS in 57m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0cd036c76dc2424a9f3f2f25152bf711 : SUCCESS in 26m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a7f1b638539466faa3e82f5ddff3021 : SUCCESS in 45m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e2fa61460994cba8d46f1568f4f4490 : SUCCESS in 50m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/880b94a0f10a4d6ea82ea24740f7b1fa : FAILURE in 1h 01m 25s","accounts_in_message":[],"_revision_number":1},{"id":"dc0bea0ce39b4db8ba5702fa3c7035bde65091e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 21:26:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/366370dd03ed49e4861235f6da7402a2 : SUCCESS in 46m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/82ded1e0a67244f9bcbbbb61b753683b : SUCCESS in 54m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2cc420091fb6464f9b0e24f6912c7d21 : SUCCESS in 1h 24m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5cc8ddb1afedd80347be5c21119cb289d35a933","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-19 10:47:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4b6c7eb46d9386a3cec23f7bbb5c5205bba9aca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-08 13:59:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cbe795c461af70a50e3b165e0b5d5401cc35c111","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-08 14:54:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"46c0e78900a060ec2bfe527083e00b7b3a6a5930","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-08 14:54:28.000000000","message":"Patch Set 2:\n\nrebasing just in case ;-)","accounts_in_message":[],"_revision_number":2},{"id":"caefbffb4a7aadefbd69cf46297152c4c185660f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 16:19:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/263aad09a562421a93d13f5e79345210 : SUCCESS in 45m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/961f1aa52ccb418ea43df1f6bdc3c955 : SUCCESS in 56m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b7163183d3d2423197fbfb2f36230616 : SUCCESS in 1h 10m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0ef54b97eb6a4266b0b353aea0acf4d1 : SUCCESS in 1h 11m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/13c72fa713544735925d789d8f5c6fca : SUCCESS in 1h 18m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df7c15073723470c99ac098d7896d65f : SUCCESS in 30m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b5039a8c15f44333826f8192e85fe224 : SUCCESS in 19m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4bc5cded53e04849925bef460e5ef9ba : SUCCESS in 57m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98752685d9a2487d9558d421902a43b1 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c378ac694854452ba5868c5325b8cef : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee4748ad25ff4ac38ae08bd93bd34615 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/176c9a6851d1413ca2b75cb497bcae09 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8816b0bc4dc45048ee6ed10cc6f53ee : SUCCESS in 8m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/897c03301f724b43aae4e443b673d697 : SUCCESS in 22m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d87d26d57e7b4e429feb31ecaab90c83 : SUCCESS in 32m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/11fe78977c4749f683f817980ed450d4 : SUCCESS in 50m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/486b3e87072a4760a4833023c3036a0f : SUCCESS in 52m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d0be2ff2aeb3438b8b57fee9e3ff1c12 : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6cd54053db498a1b576761761fdf9d5ac58d7428","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 17:04:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/34d36eb3e5e54b33859df2bf45437937 : SUCCESS in 1h 42m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e8d1b8c85004161907725929b3ed173 : SUCCESS in 1h 51m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6bea05558d840ebb285561d605fe916 : SUCCESS in 2h 06m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d108ce509474558bbb60736dc6a8000 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2649995512e481093fb10078c707cac : SUCCESS in 8m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f084c2b181d424ea6f0b46aacedb567 : SUCCESS in 36m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a20ee74f8f78f3d39c62de5a93a26d101b95f56","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-08 18:42:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"afeee7173845e865f2873e35079b13e6d22c4091","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 18:42:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"355f02aa31c73d3102b42e3a4d80859071b85c4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 20:00:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/25041013a39f4d7790936e5eb59dae7d : SUCCESS in 44m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/db4b5f3fc19f4fbba312e71f8b9fc6c0 : SUCCESS in 1h 04m 34s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2f8d128d20a943bfb29d607cbde98f0f : SUCCESS in 1h 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/301ece43bd504dc681301dbe365b4f55 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76a60b6153a94728b5924d643eb93063 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35f0c5386c2042eeb90edb9a3d163097 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75acdff4a2a74d63b7dc08131035c6d5 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5cd179f70bfa4544b96cd1d6b2f8ef85 : SUCCESS in 32m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/95f7cd6633ce428a92eb9825204531c8 : SUCCESS in 55m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/99b9399501d040f3bdebbc8d5c69924b : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ec8861b6ea40725de43dccbcde90426608e191bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 20:00:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3e9f7be837a00fdee5a11f24e18e5e65bfc9c848","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 20:01:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73dca309681f46ffa58fd59d7ec4eb6c : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b4b206de42abcb53b8536b208e3370967eb9da0d","revisions":{"b2e65cef1a347d87772a53e7c98b1fbe897f39df":{"kind":"REWORK","_number":1,"created":"2021-04-16 19:54:53.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/19/786719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/786719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/786719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/786719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/786719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/786719/1"}}},"commit":{"parents":[{"commit":"9314b534bf0175b6e1b9f26197a3638109565547","subject":"Stop publishing on release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9314b534bf0175b6e1b9f26197a3638109565547"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-16 19:53:06.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-16 19:53:06.000000000","tz":0},"subject":"[CI] Drop support for publishing versioned image tags","message":"[CI] Drop support for publishing versioned image tags\n\nFollow-up on I3259c0736a296751129aa349e8d378c0123957e6\n\nChange-Id: I6b8935b16390a301602218ce35edfe132b140a4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2e65cef1a347d87772a53e7c98b1fbe897f39df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2e65cef1a347d87772a53e7c98b1fbe897f39df"}]},"branch":"refs/heads/master"},"b4b206de42abcb53b8536b208e3370967eb9da0d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-08 14:54:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/19/786719/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/786719/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/786719/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/786719/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/786719/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/786719/2"}}},"commit":{"parents":[{"commit":"7319b1ba98d95c0eb2f5d2e2b1c62ab48404adcb","subject":"Merge \"Move clustercheck script to mariadb-base\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7319b1ba98d95c0eb2f5d2e2b1c62ab48404adcb"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-16 19:53:06.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-08 14:54:13.000000000","tz":0},"subject":"[CI] Drop support for publishing versioned image tags","message":"[CI] Drop support for publishing versioned image tags\n\nFollow-up on I3259c0736a296751129aa349e8d378c0123957e6\n\nChange-Id: I6b8935b16390a301602218ce35edfe132b140a4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4b206de42abcb53b8536b208e3370967eb9da0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4b206de42abcb53b8536b208e3370967eb9da0d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
