)]}'
{"id":"openstack%2Fkolla~786938","triplet_id":"openstack%2Fkolla~master~I28fa085fd19dbf177985f2ce385a066e65dacc79","project":"openstack/kolla","branch":"master","topic":"prepare-for-kolla-wallaby-release","hashtags":[],"change_id":"I28fa085fd19dbf177985f2ce385a066e65dacc79","subject":"base: do not overwrite APT sources.list","status":"ABANDONED","created":"2021-04-19 17:19:27.000000000","updated":"2021-10-29 15:55:24.000000000","total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"2669308f1774bd84c66923ef30df26985e297b20","_number":786938,"virtual_id_number":786938,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-04 16:14:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-08-04 14:52:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-19 18:35:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 18:06:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"d662c2ad3b1c2527d0016028f1360e38ab4f5a11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-19 17:19:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"487ee3266625032ca1ea7d98558d9f545591cfaa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-19 17:21:41.000000000","message":"Topic wallaby/no-apt-overwrite removed","accounts_in_message":[],"_revision_number":1},{"id":"31525091cceff367003a8e2ff4596e1a6896d225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 18:35:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4337fe45b9fa401e9f6413398f0f28eb : SUCCESS in 34m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d88d2275b0b94681890238f63f747077 : SUCCESS in 43m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4e6ee41fe0eb472d9e09cc2324260620 : SUCCESS in 43m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/24a6605a64f542a8807a83cecd5a2839 : SUCCESS in 30m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7df1540fbe89407f911c505c4fb331d1 : SUCCESS in 1h 06m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcad5061fce94c809c3e21fb5ddbf11a : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5479ae2dd3954ce58ea8680c2bbdf345 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70aa2657dc9845a7b29d8d6e47c60c05 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a1c696a510e46028f31ca649f012759 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0ea573c9884a32856b89d7a18a7a70 : SUCCESS in 8m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ad408f2f9d7c497fb6519dc0c4eccb21 : FAILURE in 25m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d69d2d43f650412185097e319fa8a987 : SUCCESS in 42m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/65662b9f653e434ca577e6780201bb3d : SUCCESS in 50m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f766d11cd6748c3bdb54c6d826849de : FAILURE in 17m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/20ee04b0262e4d279fe9fddc7e7560b4 : FAILURE in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c1da3985a55ccd4d386f68983aeef50a53b22325","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 19:56:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1422017460b400dbcc37216e2ce494b : RETRY_LIMIT in 1m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/879f5888454e47a895edcf02cf086d37 : FAILURE in 1h 27m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f892703526e14da7afb496b82b01a0b7 : FAILURE in 38m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47c19697a6a3fcce9e958435febea7d0c41eceee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 18:03:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d295dc3c346633e050aa1172cafc1c70ecd93888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:34:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b53f0c7d4706462eb7c32da2892ba9d7 : SUCCESS in 18m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b35e0bc39c114a1ca7141af8f870473d : SUCCESS in 39m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a14fc58826f547639eaf0329916e0eef : SUCCESS in 1h 04m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/dc8d0d7bb24d4d089d50ac33346ca198 : SUCCESS in 1h 07m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/458e2be636984fa7b87862431747d168 : SUCCESS in 1h 09m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b7c2c333a1a74edebbbb60fd06899bcb : SUCCESS in 36m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1b521d0219724fbe980ca1881517cc83 : SUCCESS in 20m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7c9bef4fc52d4966b61bd61ae26b3b73 : SUCCESS in 56m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15af401041b24fa885744aef4ae3c610 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2594537a9c6444c8f991df65e3ac0aa : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43cd5a6cb1074c8182d0e757da63a416 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/827007404cf046a1a046ff9069931104 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab2b63027fe442f8efcd1030d20aa4d : SUCCESS in 5m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e7b221ed768d48118d933e65b21dbe58 : SUCCESS in 25m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/92f436c26ea6438ca0ea18baa0025cf8 : SUCCESS in 42m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca14c913554341bda6e0fddd8e3b1c17 : SUCCESS in 54m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/df0096129789470cb5fa21d38ab0e65e : SUCCESS in 48m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0ae69b9812724d058d27d5204a665452 : FAILURE in 1h 05m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4740994aa017158f0bacbde8e05b98b88f149569","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:39:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea2ff7227b954ba48a11916ccdeb7b73 : SUCCESS in 59m 10s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d532c2b834646eb8042757bc4462824 : SUCCESS in 1h 27m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea836680b702eb5367a7b3c8a9a4ba0af1e35c3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-03 18:14:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"26238dcaef9a7be288c3c92bd60ca7b969df0fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 19:37:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7679c60184a14016b8d59ac7b57bf281 : FAILURE in 27m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1d8502a81a084538bf33bed31bda63be : FAILURE in 38m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b1666ec453aa4697a7bb7daa2fc7c8f9 : SUCCESS in 1h 09m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4a2725ed500a4f0492bf7b751df24c72 : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6481f147f4b946caad2442edd898954d : SUCCESS in 1h 16m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0f16c0b8a7d847078a4fee37ef4b9c9b : FAILURE in 44m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/24532830c4454ab58a80a06a78da055a : FAILURE in 19m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3b2cdd55fc7c49a2bd56d4cdfdc1ef97 : SUCCESS in 56m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1774bf2c8e4c29a6a882075c7f950a : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86baee8a81fa453a8f69f227e8d5e783 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64cf102d8e1a4e58a43558003e8cccb8 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c712b58e2e540859c9ff10ccf3d3e5c : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcaa7c19956544f88cff62f84428f56f : SUCCESS in 6m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c94d053db70f41fc89d30c7896bf8029 : FAILURE in 21m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/58e3fd6b14ac48fba14ee110c81ac718 : FAILURE in 47m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2d17d6ce04e4fa89538cc0554d36692 : SUCCESS in 54m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/208ed5ff07bc40779ad1cdcda4712961 : SUCCESS in 54m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/693d28d7e62643c2ae971145dd7b56bb : FAILURE in 58m 46s","accounts_in_message":[],"_revision_number":3},{"id":"cb23e916c0a0767ddb56fd95effd9d8924d1d6c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 19:42:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a23cb9e1f9a3419896d17215a065c787 : SUCCESS in 51m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/da0e8b27896d413189ec5ab7b4407f62 : SUCCESS in 1h 23m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1ad168763d4345abd2a3faa9e1428c30aa45cdf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-04 06:15:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a9218367ab1d16f86217201225aaa2371687e2e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 07:37:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7ec9d82b1af14b9dbe16bdf3bc24825d : FAILURE in 51m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1e0d3e35809049539a0ec50ac8c245bd : FAILURE in 43m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3cda6ae2c73548f0a6070d18c12881e1 : SUCCESS in 1h 18m 00s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a3eb6d0ea83c41fba62d1066f6ed515c : SUCCESS in 1h 06m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/44e3f8b47480459da9e9f9f630380cea : SUCCESS in 1h 11m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b50b22cc6d7b453985453603dd250569 : SUCCESS in 36m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c0b09ad2a8b44c0b918a87b3bfb09bc0 : SUCCESS in 21m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/13fd3ce080974635921c16618064ddba : SUCCESS in 50m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3635774cc3a64f0c824eef2cbe6d14e8 : SUCCESS in 9m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c4f4a800cb94b89aae8e9a349d4207f : SUCCESS in 11m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62b97e8618204f3ca900f41dd2233775 : SUCCESS in 10m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f1e66819ca34c5780bf4cb5e331fd8f : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50621ffed19847d3bf992b2320bf4b33 : SUCCESS in 8m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/407c7c8b91904295bb7b265bbd4b016a : SUCCESS in 34m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8052374b7edc4364ae7d621caeb6aca0 : SUCCESS in 37m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/38c1321aa3ac4c4386b7e52bde4cb0f0 : SUCCESS in 58m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae73abe166684d41abd420415e00de73 : SUCCESS in 1h 04m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e57f6d0c612d4501b2b0674fbc09dd8b : FAILURE in 1h 09m 12s","accounts_in_message":[],"_revision_number":4},{"id":"910030de92dafe037df96c10454f278ed001aa2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 07:58:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d04c174b57b4401b87a9626fa284464 : SUCCESS in 57m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d1e055f6b5d1464f994f4e26b20152c6 : SUCCESS in 1h 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"408396b864a985211d1156de8bc35c7778e8fe5d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-05 15:48:03.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4942a496284f1a4fa3751e544502c9ddfed83bb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-05 16:53:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae0e6ecec32abea3258ec49f1d2cadda29615bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 18:06:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/52691688eaf84826892e2c869817c2a8 : SUCCESS in 25m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/86d80ea3492947258c5facf13f42e1f0 : SUCCESS in 27m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f5ffb69d5e5f4ece98061620881a0d05 : SUCCESS in 1h 08m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/17ccfbc79cc1436f92862835435b3f70 : SUCCESS in 47m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ac1ba3816fb04a28af289a00c698c25f : SUCCESS in 1h 11m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/76a28043e44e4e7cabe22c3f875a1891 : FAILURE in 5m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6452017153b040c7905723c884a00826 : FAILURE in 7m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/33965bd169f64eb19e2f672af79165fc : FAILURE in 9m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22820108ba3448d7af4844cdefa49041 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19f4cdd6b3ab4078bda8231707619ba9 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acc1caa8379b4314bc7b99a59833fd29 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b75e4fc771f84ad1b619fb3971a76169 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898d764d59f34525bb4433f6df923d8e : SUCCESS in 6m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a96913ed4ce041e38883939ec3841694 : FAILURE in 5m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f37b6bc063474508b250fe83846f1fd8 : FAILURE in 5m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf990c8d1227454884549281241e374d : FAILURE in 11m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b95c659d42c46eba55a13f126aa8d70 : FAILURE in 10m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/907ca858a4f246c4a9e15dee2b840a97 : FAILURE in 10m 37s","accounts_in_message":[],"_revision_number":5},{"id":"469487e3c3dc8363180ef18d1b4a1a43b68857e3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-05 18:21:09.000000000","message":"Patch Set 5: Workflow-1\n\nlost \u0027trusted\u003dyes\u0027 on a way","accounts_in_message":[],"_revision_number":5},{"id":"e4950e19b3bc343ffdf45d3f6cc9ab8240478ee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-05 18:23:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"250d1a8d14f8187b9c0e1c83f022eef5cb37aa69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 19:43:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/738da81f58ba43598747721f201f4fdc : SUCCESS in 24m 14s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/07a93417e519455b8e42cd992e43a4f4 : SUCCESS in 37m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2d3595f6017b462686ea628f6a2bb3e6 : SUCCESS in 53m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e0c5bb7fb8254cd7835c27de9c7807c4 : SUCCESS in 59m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9b69a64658414a6d9640f59035241ef6 : SUCCESS in 1h 15m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f5661851d4f7495194d5ef2a92b854af : SUCCESS in 35m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bbdf046dd52346bf8cc44f16529a6821 : SUCCESS in 21m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7afc4b12a50f461c859d9925488acfd9 : SUCCESS in 1h 00m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65186ad6fad243a6bb9291d1d237c302 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f803a36ea2f24639ba794b53e5aff1fb : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/614c8fe47e2f4703bd2e88f84560f25f : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdcf0196d03346b9b4f627510414a55e : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8219df055edf4757a3cae49fb8f836c4 : SUCCESS in 6m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aab2fc2adee14f0eb9878e805c8640d4 : FAILURE in 5m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c5677825f8d44d959333a146c16afe65 : FAILURE in 5m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/61a45cb50f074c9db98019cef7143273 : FAILURE in 14m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b995e59229c4a71ba746a7810462fe2 : FAILURE in 14m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/791872f81e5d4a258b3d2257f02dfff3 : FAILURE in 9m 19s","accounts_in_message":[],"_revision_number":6},{"id":"4cf7b92d94c2fc153a1302c2ade12f2d60c36d66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 20:52:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a180310050c4de5a11cc43406bc219e : SUCCESS in 2h 01m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4dae5b4f3e074a92a7c110ad208d57d0 : SUCCESS in 47m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ba0507ebf8f4e0aa13b8a00bbb81e80 : SUCCESS in 2h 16m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e360a4d3b71b47328fa69c776f9ed823 : FAILURE in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"87958b3a8fbd790ad2cd76bc20c8c328f46867c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-06 06:09:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8e7307ae27c4f603d5af3563fb8ebace24532ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 07:58:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8d12e7f85530471cbccc05c3b8fa8250 : SUCCESS in 26m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/23a8e6c75dff4ceeb251a456dd4d0a85 : SUCCESS in 36m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e00cfeee72a84dc3b96d14692fe5036f : SUCCESS in 1h 22m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f7aba193c0ea471c9a395a0cfd9879c9 : SUCCESS in 1h 06m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6a5ad781c27c40229b068899485b2aad : SUCCESS in 1h 16m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/503ad9111cbc4426bfa79f0f0a598a6b : SUCCESS in 39m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5d5d723890614762a4c0417a438f03c1 : SUCCESS in 16m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/97edc76ec04948a79f774aa1bc7ee04a : SUCCESS in 51m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6699a09af75941a0beded5bc7291536a : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27212ebe9b33491fa9e428c6ae90d55f : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7d49bfadbe140628d3310427bba3c47 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d53f911cfac24fe8a08de99d4b79cc87 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70da478415d543c497ed428a847231a9 : SUCCESS in 10m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/855bac1ab78e4885afeb93195d6105b1 : SUCCESS in 27m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6bb5adbd3e8341a098b38a146a4f10ba : SUCCESS in 38m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc0767cc4ff141ac8ef0b03b2df82118 : SUCCESS in 1h 20m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94e4654876a645c6a2cd56a8aa4472f8 : SUCCESS in 57m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/80705299106c491e80695fb657cfcb65 : FAILURE in 1h 38m 39s","accounts_in_message":[],"_revision_number":7},{"id":"3113dc9285c04ca73b52438ac9c50fd6268d9a79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-06 08:05:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c705e7fccba99c40511bd42132abd1c99f57ad4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-06 08:22:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dc3fce30211fc93a361c0feb1372d4f99a6d3c92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-06 08:29:10.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3b2f9c69550b34789c9498b5fe1a129d6e3af07f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 09:59:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8a855e3f861c429f86199d0e0a420528 : SUCCESS in 23m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/193e5ed1f24047aa97e96078b7472a65 : SUCCESS in 47m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ca75debd3fe345799d5a71b2acd7ba78 : SUCCESS in 1h 16m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/19dd405f75aa4d078338418b257314c2 : SUCCESS in 1h 08m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a5d83dd241524165870f87f90424c6f6 : SUCCESS in 1h 26m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/101cfdda81864c90b82739537052d58c : SUCCESS in 44m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3fe6c1f7fa09444f8da5f01b54eca7df : SUCCESS in 32m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/89cda5dda52e4ea99ba277a303c752cb : SUCCESS in 55m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/514a10d40bba4c17b620d86414328afc : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58fb98cfae9b4206a6103d24fc79703c : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/751050a9e4804d4e871b45d354956d8c : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b421a5423c94e5191b69d614e0901a1 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d10b41b89f7041ae8750208bdaec4b9e : SUCCESS in 6m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/654e12df675643e286356731486725df : SUCCESS in 32m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2c892d598ed4498a207b9a0e760f6cf : SUCCESS in 38m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ffa770294f22440ba6290a2403fb9ac0 : SUCCESS in 54m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/472d21911def4e359fa93fdd1faf1e2c : SUCCESS in 55m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dff61770217e45e4b95f5bbc50fbbde7 : FAILURE in 1h 11m 20s","accounts_in_message":[],"_revision_number":10},{"id":"d8dba03a9a62a2b8cf93edd70853a3a4d6253fd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 10:52:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f1fdbd1aaf924234b073b06312fcb5b3 : SUCCESS in 1h 51m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f483d2081fe04fb58cf6e1a31a9ee247 : FAILURE in 23m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc7f4cdcc6c7492785dbb00f9d4d757f : SUCCESS in 2h 21m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/142760c65ab944deae841a893ba324f6 : SUCCESS in 1h 42m 06s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4fb768eed7d566c17f1e7aa4272942df46ec0506","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-06 12:05:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1019268fe4dcfa62f5bba988f1cadc71d2636e1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:37:07.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/dddf48e191224dabb432d13406062f55 : SUCCESS in 31m 11s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/897f07e31ca14acdaf66067ba5d3ad35 : SUCCESS in 54m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e726f3b337ba418589646bc09a1041a0 : SUCCESS in 1h 30m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/17fbfe65b2ce4ddb9f6e4eb25c0fc480 : SUCCESS in 59m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/491f82f3282a485a9c2f1376d30a4379 : SUCCESS in 1h 13m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cc474fb0080e41b6bd8eb93e11d188ad : SUCCESS in 38m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7f15bd962e154c9e97e005cecf370ebe : SUCCESS in 20m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/93410ae87a0a4473a56a86bb4fb464d7 : SUCCESS in 1h 05m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35d4b5c131cd4d7ab2a96969afa46988 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78aa9a02b2864962a85852e718684722 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e0d911661f24789bae1d9b88a8eec79 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f62ac0c4db6409592efb6378d3a5b89 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c793413066e545ac987936442b9bd4d7 : SUCCESS in 9m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/35650de46cd545c79fde66c7ea9efa1f : SUCCESS in 39m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/05887f48f8fe4f4cb25a38c7ae32483e : SUCCESS in 46m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/01dce649ca264d2aa754000bb3a57c5a : SUCCESS in 56m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6425c43eca824cd4b41d0f132c14ee92 : SUCCESS in 1h 06m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b1f9b23fcc204b9893812e18fdf50c67 : FAILURE in 1h 07m 12s","accounts_in_message":[],"_revision_number":11},{"id":"89e2772585050bfbd901b3180de2f18fb11ecd7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 16:13:31.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a58549f7703a44a5bb2805f7b1d98cf4 : SUCCESS in 1h 58m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a50b67f192f4f138d27d14d89ea4169 : POST_FAILURE in 4h 02m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/abd92c2c25da4a36a0d7f57336beadb4 : SUCCESS in 2h 10m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ae68ac306bc4b47b0f83fed55965b78 : SUCCESS in 1h 43m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9cfafa333a5f7fbd94777eeb5df97df3ec16cf4e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-10 18:06:52.000000000","message":"Patch Set 11: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"9229bd5e6d41e2acc7bd1370a9ba7d26de8d7c60","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-12 11:55:55.000000000","message":"Patch Set 11:\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c2cb52c8e0da740c8f0a94450168ca85816bd337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 07:00:15.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"fe0fa2eb368408b2667cb7b5faeda27edeea3c45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 08:25:51.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8e5ba04f004043489f866a92abb6ced0 : SUCCESS in 25m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c5ebc0772a474b9abe4597281efee641 : SUCCESS in 46m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e9a585e318914a738d9d19643234d7a3 : SUCCESS in 1h 13m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7e9dd50d8a044178b7192414e656c675 : SUCCESS in 1h 01m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c7511c7656bf4039b35217cdb6715e26 : SUCCESS in 1h 20m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b8bd965f384f4c35861eb76c8a12839c : SUCCESS in 29m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1b86644e723d4ea0b4ca0ebbbfa4d5b4 : FAILURE in 18m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/59b495dd2cba43d1895d464f720970c4 : SUCCESS in 59m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a376a81ec003492a84ee89dccc857930 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcc93b6487be4ce89741512a3e817a1b : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82d09fc0526142f683d542ade67a6fbf : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5592db7ae08b4dc6b2536028997cf464 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f566b510f64befa10f8e069315c3f3 : SUCCESS in 6m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/35536022d7f448108ab408b881d311c9 : SUCCESS in 21m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/12b79ec52bcf4d64a75280c6782ae927 : SUCCESS in 40m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0edb5ad96a454a698bd053126be9a0c2 : SUCCESS in 49m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b4196093e1524e97a3003635a73dbd92 : SUCCESS in 57m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/504a4bc9d17547ca8f9d0308b548e950 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":12},{"id":"5ca6447230d01c6efb318a97af4eae6831ee594b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 09:10:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4521d490813e4812bf3443873ce06ae0 : SUCCESS in 2h 05m 16s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6efd99d282a4a52a93dc9e339bf6789 : SUCCESS in 1h 53m 59s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/06fabf8275704f38afd27cc024619c81 : SUCCESS in 2h 07m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc0e540a52754fc39a81caaa6c94cb4f : SUCCESS in 1h 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2a8caaa0b18cc404e4388c09cf16794f7340cc02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 12:39:55.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"1c0b4a4d3f80b8c018c3002ae4ba1f29ceb98572","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-17 13:37:56.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"382712e281ec81e9b1e55d74b3037a0de8919c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:51:25.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f0f7d2eb9e8e45e3a88f0ae83ec76946 : SUCCESS in 1h 13m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6789164e1d1848348fcf311c6edc8e66 : SUCCESS in 2h 05m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/65fd0841c6934fa593c3aaa13efdbf8c : SUCCESS in 52m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fd55f9947e894dc1bb9d0ce3c7a2337f : SUCCESS in 53m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/aa45e8a4627f41338b4d12223fd0b1a4 : SUCCESS in 2h 08m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d0b08f2bee514dbfbed7bda0fea9ba0b : SUCCESS in 1h 30m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2b9a02fb5dcc4f26a39c969801d6810b : FAILURE in 1h 37m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/19a017a6db104d24869030b7a8c41fc6 : SUCCESS in 49m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2dfb3684ad4412889cc905734283d1 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81439e56ab214d3abcec9c3ae79e84c0 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/414ad3b0e43148caab641ba33d5cba6c : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17d9873071ea4b13be1a122a109d8c6c : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d09b054049641a9b84256be139ae6eb : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57f19113d9e84ffcbc1186ee89f15361 : SUCCESS in 24m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c5e409be0a6f4f789d7f70007877fa7a : SUCCESS in 35m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c052300a43894924a56af2f41f044118 : SUCCESS in 44m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1565549dceb4c16a5f283049cfbdb6f : SUCCESS in 49m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/18ba8eff297f48168e192299d945b618 : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":13},{"id":"45ab122e48a047523d58590f0834cae007e3a459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-17 16:31:15.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"9bb57578920fe27e1be936c1dffb6b09b1c8f147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 17:40:40.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b6d0ffda6fb1488480172ffa482e29d1 : SUCCESS in 26m 41s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f593cac9605f42a996f6264009abafd9 : SUCCESS in 38m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/af26ca4954c24a868a56017daf38f716 : SUCCESS in 56m 01s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c3fff786d15d427790e5a267f352342d : SUCCESS in 57m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e05241e8800f45c3b4c7741c9acada3b : SUCCESS in 1h 07m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/50d50e5a45714ab1a26380cae4d23ddb : SUCCESS in 31m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0200f5192aa741ec9b06094c4de08bea : FAILURE in 22m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bab78cc6a3504fa4af3851e0da9bc3cb : SUCCESS in 55m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feae426bf8ad442da1eceda16430795a : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60f7d27d8a8c479989c78c39c818d00b : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/125244c9eaa74db5b563560e6d2e037b : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b191ad81a8b47df8717e405e3e1b58f : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/735dec9016504fc7ab1c3e624fa0710f : SUCCESS in 7m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c981f02041e468a8ab4dca013ddf699 : SUCCESS in 25m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3b005f5b5cfa4bc5999ab72daf630417 : SUCCESS in 38m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/604b6ce1b6d6463ab3e2a5001766a31a : SUCCESS in 45m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/84cf22045198440e8e663b5be5576689 : SUCCESS in 44m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/58a65baafa13445ea8afb4b4297918ee : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":14},{"id":"501bc7488b34aa036f7771d6b89c25e3748f97f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 18:39:12.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e33d299492de4fa983f5416cd0ba1f9b : SUCCESS in 1h 51m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b57e2a39560d48d19cccaaad94ede4d6 : SUCCESS in 1h 09m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c02a12bb22d94c15b40c0d1a7bbe42d8 : SUCCESS in 2h 05m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/182f56e7ec254acf8bbda4f7a2b4497e : SUCCESS in 1h 34m 08s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b07be4aaefc097a5945d1bfcce7abb076b6d5431","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-03 12:11:30.000000000","message":"Patch Set 14:\n\nI am wondering is it the proper way to go.\n\nIf someone wants to have own mirror used then can replace sources.list.* files and do build without using template overrides.\n\nAfter this patch this would require making own base image.","accounts_in_message":[],"_revision_number":14},{"id":"d09cc970eaf1106cc74292c43bee31e261d9663f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-04 14:45:47.000000000","message":"Patch Set 14:\n\n\u003e Patch Set 14:\n\u003e \n\u003e I am wondering is it the proper way to go.\n\u003e \n\u003e If someone wants to have own mirror used then can replace sources.list.* files and do build without using template overrides.\n\u003e \n\u003e After this patch this would require making own base image.\n\nBest have an early block to customise the repos.","accounts_in_message":[],"_revision_number":14},{"id":"87298af668b39d3c474532e82f743d17961ce881","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-04 14:49:06.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d7095e82de63207b1715729bdb98479728eabf99","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-04 14:52:45.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"2a333950cfbcde7ef296876b1e7f4bf8e2837724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 16:14:40.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/861a5b7f9be04e908cde8acc00a472b8 : SUCCESS in 23m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5f382dafaab344a4b15c851b669b8c9b : SUCCESS in 39m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e7741679e1534dacbb609e7457ee0ec9 : SUCCESS in 1h 10m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/57e294198d644080a617859349ebe1cd : SUCCESS in 1h 04m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/95cbd1b20cad429490cd13d66e2dff6f : SUCCESS in 1h 24m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/808ce08525e046b2aa0fe550a897aa15 : SUCCESS in 31m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dfc4277dbd1143038ad378f9ebeb5680 : SUCCESS in 16m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6357bc848ca5402dbbb1e8cca3334eb3 : SUCCESS in 1h 03m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6e6968b27f4ddab22304add9392ca9 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b84583f1e8514b229c081e07b4ba1505 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf215156abbd4fc2933b2758ff945867 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae5f1620debf48e4a72b34b60b7494bb : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35e12ec9e746496eb36579ff11a5f8f4 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/88d2a33659a746a18c9b8ce92d05956a : FAILURE in 20m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/24d05bf3e7af44d598825950a766c530 : SUCCESS in 32m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a7e3acdfa924d56a967facc4cf1372c : SUCCESS in 45m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1a014ed65d0b4f33bdc87f4333642c6c : SUCCESS in 53m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0e415cbd32be453dbfb42d20159bd01a : SUCCESS in 57m 28s","accounts_in_message":[],"_revision_number":15},{"id":"b24c6eae2ce30b10d95b278afa40c9d822641544","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 16:46:29.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c201a367c5f3413f92d126ca445feeea : SUCCESS in 1h 24m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/15532f0e62b2468d88da782339cff78a : SUCCESS in 1h 13m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/91e4f55e83b847dbae0054d601498d78 : SUCCESS in 1h 48m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5864ea0626b248e4bea2bb86aa0017b1 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba1c65309e2747e19fcf10c7cb0721a6 : SUCCESS in 5m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4c579afa9dc47df87bb708da35fc3f7 : SUCCESS in 1h 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d8bc9694cdfd91867cbff25ba5f9795e24cfff52","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-04 17:06:11.000000000","message":"Hashtag added: test","accounts_in_message":[],"_revision_number":15},{"id":"1d550af3b1f490c13166061ed49fafbfc6b0883e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-04 17:06:18.000000000","message":"Hashtag removed: test","accounts_in_message":[],"_revision_number":15},{"id":"f9ffcf6a69edee6330b380a95b93a2d15c24dcef","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-10-28 10:59:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":15},{"id":"2669308f1774bd84c66923ef30df26985e297b20","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-29 15:55:24.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"6d407ba9a74f51294faee549b3118d43d180d326","revisions":{"4095ae3cba142a5de388b16057145993cb40ca6b":{"kind":"REWORK","_number":1,"created":"2021-04-19 17:19:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/1"}}},"commit":{"parents":[{"commit":"d043385205a914d2d39f050ad9391e8dd8df7a51","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d043385205a914d2d39f050ad9391e8dd8df7a51"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-19 17:18:54.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4095ae3cba142a5de388b16057145993cb40ca6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4095ae3cba142a5de388b16057145993cb40ca6b"}]},"branch":"refs/heads/master"},"7df7a4d9c3aedfe22d4116a21186f92e5bd2bbd6":{"kind":"REWORK","_number":2,"created":"2021-04-26 18:03:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/2"}}},"commit":{"parents":[{"commit":"3c33d83bed7a761895f970ef1d2ce764eafffadf","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c33d83bed7a761895f970ef1d2ce764eafffadf"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-26 18:03:32.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7df7a4d9c3aedfe22d4116a21186f92e5bd2bbd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7df7a4d9c3aedfe22d4116a21186f92e5bd2bbd6"}]},"branch":"refs/heads/master"},"dcd445b378762f0530947ac93ef3e7f97238d3a9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2021-05-03 18:14:00.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/3"}}},"commit":{"parents":[{"commit":"4955003e52b05b753a1341c739e4319643b60ed3","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4955003e52b05b753a1341c739e4319643b60ed3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-03 18:12:41.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcd445b378762f0530947ac93ef3e7f97238d3a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcd445b378762f0530947ac93ef3e7f97238d3a9"}]},"branch":"refs/heads/master"},"9efc77ed10a833e4e818b87839ea58d7625334d1":{"kind":"REWORK","_number":4,"created":"2021-05-04 06:15:09.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/4"}}},"commit":{"parents":[{"commit":"4955003e52b05b753a1341c739e4319643b60ed3","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4955003e52b05b753a1341c739e4319643b60ed3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-04 06:14:41.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efc77ed10a833e4e818b87839ea58d7625334d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efc77ed10a833e4e818b87839ea58d7625334d1"}]},"branch":"refs/heads/master"},"4d4cfa74d87dce128240d5658b4abcc7b1f0b5c5":{"kind":"REWORK","_number":5,"created":"2021-05-05 16:53:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/5"}}},"commit":{"parents":[{"commit":"9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-05 16:53:04.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d4cfa74d87dce128240d5658b4abcc7b1f0b5c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d4cfa74d87dce128240d5658b4abcc7b1f0b5c5"}]},"branch":"refs/heads/master"},"90f2f532bd78caaf42fd8ea9de97f539fa6ba661":{"kind":"REWORK","_number":6,"created":"2021-05-05 18:23:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/6"}}},"commit":{"parents":[{"commit":"9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-05 18:22:27.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90f2f532bd78caaf42fd8ea9de97f539fa6ba661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90f2f532bd78caaf42fd8ea9de97f539fa6ba661"}]},"branch":"refs/heads/master"},"1e116904bc4b81266299f33a8d0bb0acc96f027a":{"kind":"REWORK","_number":7,"created":"2021-05-06 06:09:12.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/7"}}},"commit":{"parents":[{"commit":"9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-06 06:09:01.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e116904bc4b81266299f33a8d0bb0acc96f027a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e116904bc4b81266299f33a8d0bb0acc96f027a"}]},"branch":"refs/heads/master"},"0277fb6963c4a843bcb9b00bda65f8c80125a82f":{"kind":"REWORK","_number":8,"created":"2021-05-06 08:05:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/8"}}},"commit":{"parents":[{"commit":"9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-06 08:05:30.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0277fb6963c4a843bcb9b00bda65f8c80125a82f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0277fb6963c4a843bcb9b00bda65f8c80125a82f"}]},"branch":"refs/heads/master"},"bfc731ad29c114e0bd4cd5fd4b247f82cd812472":{"kind":"REWORK","_number":9,"created":"2021-05-06 08:22:01.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/9"}}},"commit":{"parents":[{"commit":"9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-06 08:21:14.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfc731ad29c114e0bd4cd5fd4b247f82cd812472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfc731ad29c114e0bd4cd5fd4b247f82cd812472"}]},"branch":"refs/heads/master"},"f9ecb424fc53cbab5b278edc95323257c2a9a4b1":{"kind":"REWORK","_number":10,"created":"2021-05-06 08:29:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/10"}}},"commit":{"parents":[{"commit":"9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b5da3c07e03e75babc7ffbb8461bcee4e90d9a2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-06 08:29:02.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9ecb424fc53cbab5b278edc95323257c2a9a4b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9ecb424fc53cbab5b278edc95323257c2a9a4b1"}]},"branch":"refs/heads/master"},"7554edb37b649bca0d7e4f2e074a924ff80fcb76":{"kind":"REWORK","_number":11,"created":"2021-05-06 12:05:46.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/11"}}},"commit":{"parents":[{"commit":"36e0c7e198654180e759cf07f041e838f8f035fd","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e0c7e198654180e759cf07f041e838f8f035fd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-13 10:57:59.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-06 12:04:34.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7554edb37b649bca0d7e4f2e074a924ff80fcb76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7554edb37b649bca0d7e4f2e074a924ff80fcb76"}]},"branch":"refs/heads/master"},"c5d1a435a030f966fc755d69e1e9f015dc406e8e":{"kind":"REWORK","_number":12,"created":"2021-05-17 07:00:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/12"}}},"commit":{"parents":[{"commit":"882307862cfb208b497751b6a1ab0d95100ff95a","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/882307862cfb208b497751b6a1ab0d95100ff95a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 06:54:40.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 06:59:31.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5d1a435a030f966fc755d69e1e9f015dc406e8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5d1a435a030f966fc755d69e1e9f015dc406e8e"}]},"branch":"refs/heads/master"},"d97bffba4de8cba123a9826ac7f51d9a46bbd715":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-05-17 12:39:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/13"}}},"commit":{"parents":[{"commit":"11a60e438af93ecce75aa19eb2e722fa5ebe1809","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11a60e438af93ecce75aa19eb2e722fa5ebe1809"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 06:54:40.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 12:39:22.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d97bffba4de8cba123a9826ac7f51d9a46bbd715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d97bffba4de8cba123a9826ac7f51d9a46bbd715"}]},"branch":"refs/heads/master"},"dc5f99bca7d6574e3198bec420cc3270cbe4317b":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-05-17 16:31:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/14"}}},"commit":{"parents":[{"commit":"eb58d81cc7b146d6c7c7b707141c8f008edf94a5","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb58d81cc7b146d6c7c7b707141c8f008edf94a5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 06:54:40.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 16:30:55.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc5f99bca7d6574e3198bec420cc3270cbe4317b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc5f99bca7d6574e3198bec420cc3270cbe4317b"}]},"branch":"refs/heads/master"},"6d407ba9a74f51294faee549b3118d43d180d326":{"kind":"REWORK","_number":15,"created":"2021-08-04 14:49:06.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/786938/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/786938/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/786938/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/786938/15"}}},"commit":{"parents":[{"commit":"12c3754687aff3d9a70ba2290c91bbcfa0321762","subject":"Debian/Ubuntu: handle APT keys in proper way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12c3754687aff3d9a70ba2290c91bbcfa0321762"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-17 06:54:40.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-08-04 14:48:58.000000000","tz":120},"subject":"base: do not overwrite APT sources.list","message":"base: do not overwrite APT sources.list\n\nWe handle external repositories via repos.yaml file so our sources.list\nfiles are basically same as ones in base container images.\n\nOnce we start requiring packages from bullseye-backports repo we can\nmove to \u0027debian:bullseye-backports\u0027 container image as a base.\n\nBoth distros have their OpenStack repositories enabled in official way:\n- Debian by \u0027extrepo\u0027\n- Ubuntu by \u0027add-apt-repository\u0027\n\nChange-Id: I28fa085fd19dbf177985f2ce385a066e65dacc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d407ba9a74f51294faee549b3118d43d180d326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d407ba9a74f51294faee549b3118d43d180d326"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
