)]}'
{"id":"openstack%2Fkolla~787265","triplet_id":"openstack%2Fkolla~master~Ibd955869a6f9485dfa4d08a8ad2f4b28b7d59c15","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ibd955869a6f9485dfa4d08a8ad2f4b28b7d59c15","subject":"Drop ppc64le support","status":"MERGED","created":"2021-04-20 16:36:47.000000000","updated":"2021-05-17 11:12:37.000000000","submitted":"2021-05-17 11:11:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787265","meta_rev_id":"204a6f58153a2ef84a6d6c67781224b142944cff","_number":787265,"virtual_id_number":787265,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-17 11:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-05-11 13:18:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-05-17 10:02:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-05-17 10:02:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2021-05-17 11:12:37.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 17:59:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-21 08:47:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-04-26 09:20:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-05-11 09:00:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"21b7cebdc4743df1cde622c8bbcf6560528e9eed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-20 16:36:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcbe0345876a430062483bca513846cb87b29088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 17:59:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/488afcd5df6d44eb948d795f7a9bd09f : SUCCESS in 25m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9b514cde8e1c4ee3af5e184665f9641a : SUCCESS in 47m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e14a9a486c224803b564a9f9d5e5073a : SUCCESS in 36m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4e7a030aabfd4793830dfd11b34223c5 : FAILURE in 29m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a1eb199851a14e4a9929c987676003fe : SUCCESS in 58m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d4873205c3940f7b4cab2e79527c229 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b97532beeec24517a39a5b86a7914b57 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a8257800b5b4da596359c294b5e184a : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50b451b4094d4446924919427c38844e : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd59235f15b04ba5911d126a4c00122a : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a185276d7d2a4b24a6cf0d8b8ff1764c : SUCCESS in 5m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf167e19777045eebcd1a80f9bdbab09 : SUCCESS in 28m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f3de0555bbc45189d8af00bfe7bce75 : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb4667b17e2f4095abdda032fbd0e032 : SUCCESS in 1h 06m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43346162ddd84d1d869668f4d95ea1c7 : SUCCESS in 53m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/106998ee622a422eb0ad1e195e160baa : SUCCESS in 1h 11m 51s","accounts_in_message":[],"_revision_number":1},{"id":"40ffe93ed55a86d53ef49419ba91b6863f2c6479","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 19:21:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6a70a21ebf94fb6bf51492d761ece99 : RETRY_LIMIT in 1m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/86d759cb88fe4656892713fe64321a02 : SUCCESS in 2h 07m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f64bc8615dd540bda08fd7884d949218 : SUCCESS in 2h 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8f8cade2e3efdfecdf0abf6541a74f4381b9f4b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 08:47:38.000000000","message":"Patch Set 1:\n\nDeprecate in Wallaby?","accounts_in_message":[],"_revision_number":1},{"id":"6084b0b53413f4d45ad887c96ca093ffb6170f6b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-21 09:34:33.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Deprecate in Wallaby?\n\nDrop. We do not publish images, we do not test, we do not support.","accounts_in_message":[],"_revision_number":1},{"id":"db60d8c81b557fd725af3bcfeb0da583e8583d16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 09:17:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7b2475db6e29596bbed2ec147add90c72a40a756","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 09:18:41.000000000","message":"Patch Set 2: Backport-Candidate-2\n\nXena only","accounts_in_message":[],"_revision_number":2},{"id":"26757755a3d8c0d0f6d565f07bd090557ee462b0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-26 09:20:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"20da1f8aaeed6cedd148d19ec0a4c9f905d53cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 10:45:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b8f4ba44493c41ff85fe6212da2131af : SUCCESS in 25m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f43241a068364e82898cdf3e8ccc8987 : SUCCESS in 39m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0865bf4fb5204d579e1dbe060aab4d83 : SUCCESS in 1h 09m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c598968459764904bd671c96559b6b42 : SUCCESS in 1h 07m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/de412c873d2d4ffea23bb39249f095f7 : SUCCESS in 1h 20m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d80efcf47c574d588d761743b8d7f889 : SUCCESS in 36m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ba1eb1952c9d43eeaf89a2c34e4f566e : FAILURE in 20m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4f98c7a457fb469392dc152333d23c2e : POST_FAILURE in 15m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc171f7d69b0410b89dbe92abb6eed49 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92edd89b654f4d4bb7fb8895818a7984 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3a62212e3454c28b88c831f55c4d6b3 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf3cf13610e64ae6b428c88776ab8c48 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae295455f1047f583d45f98e0c45aa8 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e242e72bb80f42f0902238b7250b86f7 : SUCCESS in 4m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afeca2a88a5c4b9995ed046d27b5ddb0 : SUCCESS in 22m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c97f801a9c224e7a9919abe4fc06c9ff : SUCCESS in 55m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f5c53dd740e6476db9eaafc8f2dda858 : SUCCESS in 1h 00m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7cc9ca926574487688c30854de9ac3ca : SUCCESS in 56m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0e3ba4acbc0c43f5a54c47663c858b64 : SUCCESS in 1h 11m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d1981254894810684cf9ccc05f1b271d582e5cca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 11:03:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cabe3f9f8c91444dad43ffe574d716c2 : SUCCESS in 1h 43m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a90273fa574a4fcbafd4c6053e3c988d : SUCCESS in 1h 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d79bbef642e9e615056403ab5bf2c6bf67ac43bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-27 14:15:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d56503b14a4fd697475e4bf46eedb0f66b2b1f4a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-27 15:52:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ab3b4981b9d936416fa725045f6a7e29273be561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 15:53:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e4da66e9b2e444a3a38d976f0a269192 : SUCCESS in 21m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4812cd66b51e4225b222f5942a6cb08f : SUCCESS in 42m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4a0eef9827e649b589afdf79927e40c8 : SUCCESS in 1h 08m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/780cdb3aa46c451fb81296f1486bc04d : SUCCESS in 1h 27m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/76f2186a116e4ab591757039145168c4 : SUCCESS in 1h 08m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5bf4c58cefcf426b80f2aecec15712cf : SUCCESS in 48m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d8296a9817014fbe9a5ffd3792fe4464 : SUCCESS in 40m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/009ee2998292483ba9ceca1ff0641c9b : SUCCESS in 50m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f92fd4237264feea9ae12a10ac8399b : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9005e8803928407a9250de3c17aca495 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78e60130be204c06ad9dcb3ad848d327 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/377e2da81341448fa59b542d5893df66 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd7ddd487884e869228b76dab7a35eb : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca0bb0bfb4b744d1b522baf6b6bee4b5 : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/163902a6a0284990aedd553e7bb8a0a5 : SUCCESS in 23m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/27044c9887f74ee49e2fa83703392438 : SUCCESS in 53m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6660e2d7932d4ee2ade6e04f55c097ed : SUCCESS in 46m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3cdd50c8921e47d482215e07ef874c91 : SUCCESS in 51m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d10955dae54495cac73b085dee26817 : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":3},{"id":"672f9e131884a8423badf0a486a23ce219bd9896","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 16:45:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80745a3dd90140368bde954f01688c67 : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/536f287cbdb24f09b569a29044848a94 : SUCCESS in 2h 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"97e6df5c27a46907fc3f1cfadaa71a44d4f14183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-08 09:01:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"73ecc77bae1f98cf89f53984ca8a5c54c2a272e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 10:44:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6e3a983850cd4003a8f5e089100947ed : SUCCESS in 41m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bc4fecabe20c407f978a8b171ccbc348 : SUCCESS in 50m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e8786ec38dd146999a01cd628f8bcabe : SUCCESS in 1h 25m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d35c5ed3cccf429e92e86f8b2bdb646e : SUCCESS in 1h 18m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/27abc5fef9b846b9adaf7fdeaef1b4dc : SUCCESS in 1h 37m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1b034f7957d948f28da5365468862cd2 : SUCCESS in 49m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bea04c3950444563b746627318e899fe : SUCCESS in 37m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/99b0740e0f80436cac20434de6be27c6 : SUCCESS in 1h 07m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d90ace604c4513b5694a3d1d8c23a3 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c9a028887b547609b1c9c4c0614e3b6 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f5f72f378cd4e3981d6ae18c6c01354 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3b3dc69aaf4449da3e1f1c9a90063a5 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee862133806447886f5764c70f84c40 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3a5f6d8054b4a77a72faa0728e1b823 : SUCCESS in 4m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b261c01c85ab461bb7aa8991960efc8b : SUCCESS in 36m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/87e2395b5c72493284af302eb6c30d43 : SUCCESS in 45m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/570dbe87de4d4f1f923e6a04de8e6410 : SUCCESS in 1h 04m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/424f5df27c2b4576bf8f9950552bc5af : SUCCESS in 59m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f10b7842c14e4760ad34c27b8b0a11f4 : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":4},{"id":"cc1ce84cbc5ea777c5e8129ef20f053a653cb878","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 11:24:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec4957d9231a4b81a839c58ffd50a2a1 : SUCCESS in 2h 18m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e4fd783eff1b49c49dccda4bb11d6e49 : SUCCESS in 1h 14m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1f839d69da44de88fdb65f1283e61aa : SUCCESS in 2h 20m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/93f0658346e84ab09b67e9601d0abd2a : SUCCESS in 2h 06m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b5e164a8886cd35b4f39871d023eae68bf3c61a7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-10 17:54:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8acb830992acb25dbafd3deccb46ace259bc97bf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 09:00:46.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nMerge conflict","accounts_in_message":[],"_revision_number":4},{"id":"6753133cdf798d90b376259c2ad3f2ebb5ff4ede","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-11 10:43:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4d5d8c9addeb6bc354a22130985e0cfe743c4fb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-11 10:45:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"37740ad4bcd7813df3bd3947e19d5e68f645eb2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 12:04:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3d5224e7b6f046dfb94030dd0ebb6346 : SUCCESS in 20m 03s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5e024acafabe4b2dbecaa0e266bced0b : SUCCESS in 34m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0b2234846b204f999b1773c32b65a9b5 : SUCCESS in 58m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0f15a3d75b15498981cdfb3fa4d5717c : SUCCESS in 1h 05m 44s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dbaee8dee632412ca7a053ccb25b9ce6 : SUCCESS in 1h 14m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/29bd5b8ef8ec4fe4b139bda14ab4f468 : SUCCESS in 32m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d63508e16882418bb653269cead837a3 : SUCCESS in 17m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/62fa35f33bf6483892de73987ab9437c : SUCCESS in 48m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b638e4a83a4106a71fa3a98bfeaf85 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4294449dcfb4ce8a0c1544a872f52d6 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dbf1a0588a34937bd13444e9a2234a3 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecd976a6adaf4be1b786ea1349e1cb07 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e691598fdd8d417e8af7ce18b4571d85 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b65e604063364d5c836ab820a5fe9a16 : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4cbc2b9103484e739f85c6369277007f : SUCCESS in 21m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/730a29ea45ce4cc08ffdf5bcbaa5fd47 : SUCCESS in 33m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4fb59b6ce5f1471ebda926460bd12760 : SUCCESS in 44m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/454092809fb7412396f36222247ca083 : SUCCESS in 49m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1ee41be0229d40cb87f7f59c2cfbb7ca : SUCCESS in 1h 08m 52s","accounts_in_message":[],"_revision_number":5},{"id":"6b2f5ff8b7279ed57954c8d32d4fc8a357816256","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 13:06:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a30fb0e4e04460e87bc9d5addfc400c : SUCCESS in 2h 12m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a5a2f183fdd46898ba2bfa360739801 : SUCCESS in 49m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/78ae6b5e9f6f447c82d6500a214ddce0 : SUCCESS in 2h 19m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea9e4dc993e24e16b22c7b121624cac8 : SUCCESS in 1h 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2de46276b795e03f9bdb4816916c2432d2a4e0b6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 13:18:20.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"eddc5d92af2f5a0273aa3c293f8f4bc1651bf630","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-17 10:02:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4660a0b67b11c2afb4bb014b8319ef8e2545b32f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:02:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ba9250348fb7e6513a74b99e8848f435b93181e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 11:11:10.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a070ea692d87489c83a49b6fef5dd9bb : SUCCESS in 30m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6367151c925f4ffabd895ed490db7d09 : SUCCESS in 1h 02m 58s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4946b4036f314b60839667d91d27707d : SUCCESS in 1h 07m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/092bcee03b624405855b23b3de47ca47 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f72afce12784d3883d243b11990036f : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1209a4c6eb84f0792fbc1fc93671cf7 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70f9075118a48ecb57f4caf74c59534 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/795f5c00597f4204bb5be70fe996ec59 : SUCCESS in 4m 47s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/411702555a7448649f977af7e1cb4d78 : SUCCESS in 39m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2effbe2184384e3fb99880e991b272ef : SUCCESS in 45m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9066d458617f4f95a26e530870fe5b69 : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":5},{"id":"097f3969c9488de4b70d4135fd53f75c4c618f4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 11:11:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"2ffa8a47bca7ca2f39f4e4b692a1f9371ec1d5ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 11:12:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cdcf48432f41a2a6b6c60be16a0be4 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a778ef27922d429586476bc7f61394f9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0506ede84f97159a8156ad51aea395d0868e6141","revisions":{"96cf8b74e357e8533f1e0b0bcba405cca0ece9f2":{"kind":"REWORK","_number":1,"created":"2021-04-20 16:36:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/65/787265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/787265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/787265/1"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-20 16:25:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-20 16:36:28.000000000","tz":120},"subject":"Drop ppc64le support","message":"Drop ppc64le support\n\nWe do not test support for ppc64le on CI or other systems.\n\nIn previous cycles it was used by TripleO and now they have own way.\n\nChange-Id: Ibd955869a6f9485dfa4d08a8ad2f4b28b7d59c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96cf8b74e357e8533f1e0b0bcba405cca0ece9f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96cf8b74e357e8533f1e0b0bcba405cca0ece9f2"}]},"branch":"refs/heads/master"},"36c50aea738d94cf71c93808bc2ac11d443db21c":{"kind":"REWORK","_number":2,"created":"2021-04-26 09:17:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/65/787265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/787265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/787265/2"}}},"commit":{"parents":[{"commit":"f00961da2251ad2c374a03f081db4bf0f089cf73","subject":"docs: deprecate ppc64le in Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f00961da2251ad2c374a03f081db4bf0f089cf73"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-20 16:25:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-26 09:17:22.000000000","tz":120},"subject":"Drop ppc64le support","message":"Drop ppc64le support\n\nWe do not test support for ppc64le on CI or other systems.\n\nIn previous cycles it was used by TripleO and now they have own way.\n\nChange-Id: Ibd955869a6f9485dfa4d08a8ad2f4b28b7d59c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36c50aea738d94cf71c93808bc2ac11d443db21c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36c50aea738d94cf71c93808bc2ac11d443db21c"}]},"branch":"refs/heads/master"},"d61ca5889db5be6b442b760e05caddbf839ec32a":{"kind":"REWORK","_number":3,"created":"2021-04-27 14:15:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/65/787265/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/787265/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/787265/3"}}},"commit":{"parents":[{"commit":"edc14536d0544c81e2e5f3d011a9828177800547","subject":"Merge \"releasenotes: we move to Debian bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/edc14536d0544c81e2e5f3d011a9828177800547"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-20 16:25:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-27 14:14:39.000000000","tz":120},"subject":"Drop ppc64le support","message":"Drop ppc64le support\n\nWe do not test support for ppc64le on CI or other systems.\n\nIn previous cycles it was used by TripleO and now they have own way.\n\nChange-Id: Ibd955869a6f9485dfa4d08a8ad2f4b28b7d59c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d61ca5889db5be6b442b760e05caddbf839ec32a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d61ca5889db5be6b442b760e05caddbf839ec32a"}]},"branch":"refs/heads/master"},"183462618c65d9e2bc91198c56b3a127def062b7":{"kind":"REWORK","_number":4,"created":"2021-05-08 09:01:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/65/787265/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/787265/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/787265/4"}}},"commit":{"parents":[{"commit":"6dfae74813289575e77aaa7dfe43f7e6a81a5f26","subject":"Merge \"[release] Use UCA Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6dfae74813289575e77aaa7dfe43f7e6a81a5f26"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-20 16:25:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-08 09:01:07.000000000","tz":120},"subject":"Drop ppc64le support","message":"Drop ppc64le support\n\nWe do not test support for ppc64le on CI or other systems.\n\nIn previous cycles it was used by TripleO and now they have own way.\n\nChange-Id: Ibd955869a6f9485dfa4d08a8ad2f4b28b7d59c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/183462618c65d9e2bc91198c56b3a127def062b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/183462618c65d9e2bc91198c56b3a127def062b7"}]},"branch":"refs/heads/master"},"0506ede84f97159a8156ad51aea395d0868e6141":{"kind":"REWORK","_number":5,"created":"2021-05-11 10:45:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/65/787265/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/787265/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/787265/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/787265/5"}}},"commit":{"parents":[{"commit":"73c456f85b4e861e4be12282a67ca59f5f319357","subject":"Merge \"docs: update supported OS distros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73c456f85b4e861e4be12282a67ca59f5f319357"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-20 16:25:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-11 10:45:16.000000000","tz":120},"subject":"Drop ppc64le support","message":"Drop ppc64le support\n\nWe do not test support for ppc64le on CI or other systems.\n\nIn previous cycles it was used by TripleO and now they have own way.\n\nChange-Id: Ibd955869a6f9485dfa4d08a8ad2f4b28b7d59c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0506ede84f97159a8156ad51aea395d0868e6141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0506ede84f97159a8156ad51aea395d0868e6141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
