)]}'
{"id":"openstack%2Fkolla~787274","triplet_id":"openstack%2Fkolla~master~I7b689422a5f5f35c04122898a70ce8c373563135","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I7b689422a5f5f35c04122898a70ce8c373563135","subject":"Add LetsEncrypt images","status":"ABANDONED","created":"2021-04-20 17:42:42.000000000","updated":"2021-04-30 00:13:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6b64f0c4683a79ef467b90c51773ce2c2f201021","_number":787274,"virtual_id_number":787274,"owner":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-20 21:55:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 19:14:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0008a5735cfccc8aff0ec4b337933357bcb59e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2021-04-20 17:42:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3d2ce3238bf2f8c972d7d8bb5d21f6bd20acaea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 19:14:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/194782b4e5d447a1b88b375155f94873 : SUCCESS in 1h 19m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c06f08e6b2324cf3b05f55aabf82c81f : SUCCESS in 1h 05m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9c26471803e448308ba500120ca07f7f : SUCCESS in 48m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/26bd83bca20041269685ddac5295aea2 : SUCCESS in 22m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0a55f63724a0459aa70ca7c0c3b85e71 : SUCCESS in 1h 25m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189ed5579aea49afaf93d7bd80f53f53 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b294a27855148a9a1496684748e9566 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6348d96a42ae432c88b39c997f5ef47c : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2da4e161cbbc47e98af42fb82cec3f0e : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9bd04f9a9bc493481bb3ed9240c5b36 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d66f1a8a7944d97af65fcdf60267cda : SUCCESS in 5m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fa22db65ae8843dfa1a0b386e444520d : SUCCESS in 27m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/229c3f2e2e2a46f48d0fa063590eb635 : FAILURE in 1h 01m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5dfd711f4cda42369cf7c4819ab847e4 : SUCCESS in 1h 01m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c570a2f63574acc94444be6fa92e243 : SUCCESS in 51m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/301b711b2e95456ca340d91a19bb0cc4 : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b28cb189f156efd9b2e7d507683b045c5d3e0e62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 19:36:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f688625cdc354494a4e94fdf1c3871b0 : RETRY_LIMIT in 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a1b4df4adb040abbeb5aae9a3b86883 : SUCCESS in 1h 51m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a17e0214b7d94ff3a6b0fd11289a22cc : SUCCESS in 1h 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fbdd05870e279f1d591c7f28a09fbfe0cb3d724","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2021-04-20 20:44:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25849db4e2453b40d33855a3d9811dcb0080e348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 21:55:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/52c96dfdfeb0437f90d4bb9a997dec52 : SUCCESS in 23m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1c71d498a3724165bec7bcb0febaf0f5 : SUCCESS in 40m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/89c3a1114bf64196a8283ec6e22f6a8c : SUCCESS in 37m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d1d5ecf4d1d40cbace2ce3c09d9ce4a : FAILURE in 19m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4e00f85626aa42669acda7c70df03dad : SUCCESS in 46m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c6e7e008b449d89501abb51e64ff19 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/544ac2388b6b40d094ad8cc63d9228c5 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c781db473954f90b5996c239185fec7 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f08783ccc59d4347afe22b0afbfc50be : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df821d27416645efbd96ec93d531c3ed : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9d2f5f41b994f27966cda3d896d47b1 : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/443258c07d7548c68cfc0533dd2de223 : SUCCESS in 24m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/15fa255741df4151b34cdae98438b905 : SUCCESS in 33m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/87f4eb2cb6f745b49cd2e6b60b27bff6 : SUCCESS in 58m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e041badb6ba54006a41dd1d6dce3f801 : SUCCESS in 1h 00m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ada6ce93841748b48366e0dcd493d28c : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ea42cb5494a0acea0e4264abda2eccdc953a778b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 23:05:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4495b21a6fda4236934c1c7049321147 : RETRY_LIMIT in 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6e210b1490e4d26995c216477232a34 : SUCCESS in 1h 25m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d26fa73c8dc4a239b42fcd4eff41179 : SUCCESS in 2h 09m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b64f0c4683a79ef467b90c51773ce2c2f201021","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2021-04-30 00:13:17.000000000","message":"Abandoned\n\nusing original change, 741339","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e35fd4ee22a76827ea9549452168f73965e732c","revisions":{"8e35fd4ee22a76827ea9549452168f73965e732c":{"kind":"REWORK","_number":1,"created":"2021-04-20 17:42:42.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/74/787274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/787274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/787274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/787274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/787274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/787274/1"}}},"commit":{"parents":[{"commit":"c96a4316b324d2105326544b5026e5d825ceaf1b","subject":"[release] Use Wallaby sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c96a4316b324d2105326544b5026e5d825ceaf1b"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2021-04-20 17:42:35.000000000","tz":-420},"committer":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2021-04-20 17:42:35.000000000","tz":-420},"subject":"Add LetsEncrypt images","message":"Add LetsEncrypt images\n\nAdds container images to support LetsEncrypt service integration.\nAdds container to support ssh transfer of TLS certificates across\nOpenstack nodes.\n\nChange-Id: I7b689422a5f5f35c04122898a70ce8c373563135\nImplements: blueprint letsencrypt-https\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e35fd4ee22a76827ea9549452168f73965e732c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e35fd4ee22a76827ea9549452168f73965e732c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
