)]}'
{"id":"openstack%2Fkolla~787320","triplet_id":"openstack%2Fkolla~master~Ic976765ef4aec30d8153a20aba5ab389265554bb","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ic976765ef4aec30d8153a20aba5ab389265554bb","subject":"CentOS Stream: update kolla-ansible Zuul jobs","status":"MERGED","created":"2021-04-21 08:06:32.000000000","updated":"2021-04-21 11:11:22.000000000","submitted":"2021-04-21 10:52:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787320","meta_rev_id":"bb74bf6f54352b80a438b0e51119a5b081ca47bd","_number":787320,"virtual_id_number":787320,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-21 10:52:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-21 08:26:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-21 08:27:41.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-21 11:11:22.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-04-21 08:27:41.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-21 11:11:22.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-21 08:06:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-21 08:16:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-21 08:24:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-21 08:27:41.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"c30e4401a3b9e564291da8e6e6d8edbdb24348b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 08:06:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e32beea5be83ff71b365e8dad7474026421de6ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 08:06:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Job kolla-ansible-centos8s-source-ceph-ansible not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-centos8s-binary\n          - kolla-build-centos8s-source\n          - kolla-ansible-centos8s-source\n          - kolla-ansible-centos8s-binary\n          - kolla-ansible-centos8s-source-upgrade\n      check-arm64:\n        jobs:\n          - kolla-build-centos8s-source-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8s-source\n          - kolla-ansible-centos8s-source\n          - kolla-ansible-centos8s-source-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-centos8s-source-quay\n          - kolla-publish-centos8s-binary-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8s-source-dockerhub\n          - kolla-publish-centos8s-binary-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos8s-source\n          - kolla-ansible-centos8s-source-bifrost:\n              files: ^docker\\/(base|bifrost|openstack-base)\\/.*\n          # Test rabbitmq \u0026 mariadb changes in multinode ceph jobs.\n          - kolla-ansible-centos8s-source-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-source-upgrade-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-source-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-centos8s-source-scenario-nfv:\n              files: ^docker\\/(base|barbican|heat|mistral|openstack-base|redis|tacker)\\/.*\n          - kolla-ansible-centos8s-source-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-centos8s-source-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-centos8s-source-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-centos8s-source-masakari:\n              files: ^docker/(base|masakari)/\n          - kolla-ansible-centos8s-source-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8s-source-prometheus-efk:\n              files: ^docker/(base|elasticsearch|grafana|kibana|prometheus)/\n          - kolla-ansible-centos8s-source-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"14388618243d346a7c32c95e692ba159fb26d36a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 08:06:55.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"417b6cc588c2cbc55ae6b6d7b7e39e50f38bdf62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 08:07:04.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Job kolla-ansible-centos8s-source-ceph-ansible not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-centos8s-binary\n          - kolla-build-centos8s-source\n          - kolla-ansible-centos8s-source\n          - kolla-ansible-centos8s-binary\n          - kolla-ansible-centos8s-source-upgrade\n      check-arm64:\n        jobs:\n          - kolla-build-centos8s-source-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8s-source\n          - kolla-ansible-centos8s-source\n          - kolla-ansible-centos8s-source-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-centos8s-source-quay\n          - kolla-publish-centos8s-binary-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8s-source-dockerhub\n          - kolla-publish-centos8s-binary-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos8s-source\n          - kolla-ansible-centos8s-source-bifrost:\n              files: ^docker\\/(base|bifrost|openstack-base)\\/.*\n          # Test rabbitmq \u0026 mariadb changes in multinode ceph jobs.\n          - kolla-ansible-centos8s-source-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-source-upgrade-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-source-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-centos8s-source-scenario-nfv:\n              files: ^docker\\/(base|barbican|heat|mistral|openstack-base|redis|tacker)\\/.*\n          - kolla-ansible-centos8s-source-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-centos8s-source-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-centos8s-source-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-centos8s-source-masakari:\n              files: ^docker/(base|masakari)/\n          - kolla-ansible-centos8s-source-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8s-source-prometheus-efk:\n              files: ^docker/(base|elasticsearch|grafana|kibana|prometheus)/\n          - kolla-ansible-centos8s-source-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"7e95064d525f61639ad74a62c6799426771a34f2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-21 08:16:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67cb4924a0f051f8bfbbb608bf2f8e396bc81511","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-21 08:24:28.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027m really sorry, but we don\u0027t do ceph-ansible anymore ;-)","accounts_in_message":[],"_revision_number":1},{"id":"4511fd32eb9e39f9540a711430bd47c99cdc10eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-21 08:25:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"069f79ac0027e58d6f5b43eb7edcab172fd05db0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-21 08:26:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1e570c019a255585762a3f4cd1ac2810081c2d67","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-21 08:27:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a443cad9b4016834f8f494f0504410e61990a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 09:38:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b1681017d8994c3a93789be156f5ba8b : SUCCESS in 52m 52s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d5b2197cdeed43a2abdafdea51621095 : SUCCESS in 58m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/992abe9f4c3a41938dc2100e5b78a5ed : SUCCESS in 1h 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89b2bc52448848fb95eecf9155ebdac4 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44f283c0724a49f9a4c62f2d468f79de : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21eafec08e504f2d804111edf4c75c26 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9238597bade44f2798a8510cb23f25ae : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc4c9fd18734a83803005f77b59a551 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"c675b54759501c998afe6b2ef281d2c61d1b36a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 09:38:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"034361614512b4d779433f42f1c8bc141f731df4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 10:52:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/471ee9d7e0e343f5a02ad8b0b42317db : SUCCESS in 55m 07s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5afcf1d62fc64b17a098e4b929bea103 : SUCCESS in 1h 02m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851bbb1156e14b02a2228a5af4840df5 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df7afcb857a246c18f509706b4de47de : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/395021d3992d4da8bc8470ff16acd10e : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf10d89ee1d541fdb923b5f180d2ec37 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"34e957f9d5d23b14721a5ac7ba81c892867e3343","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 10:52:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e67af2bcd6bf1fbd4115ac10cad9028eaa2dda2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 10:56:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01893880490a4910b52aa77be3a4c9fd : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2},{"id":"51ddf246155d73fbb04583eb62e353f5fa1acd23","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-21 11:11:22.000000000","message":"Patch Set 2: Code-Review+2\n\nnice, and cephadm as well, very well","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b64fd1d02f576d0bcd581e692701d2a1cf024c63","revisions":{"97aa399ab47b2e11dcb597f8fba15315d7fba41b":{"kind":"REWORK","_number":1,"created":"2021-04-21 08:06:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/787320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/787320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/787320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/787320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/787320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/787320/1"}}},"commit":{"parents":[{"commit":"d3734a376d735700e33668f933ef2cada26258ac","subject":"Merge \"deprecate RHEL as base of images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3734a376d735700e33668f933ef2cada26258ac"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 08:06:16.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 08:06:19.000000000","tz":60},"subject":"CentOS Stream: update kolla-ansible Zuul jobs","message":"CentOS Stream: update kolla-ansible Zuul jobs\n\nThe kolla-ansible jobs were renamed from centos8 to centos8s when they\nswitched to CentOS stream. The Kolla pipeline must reference the new\nnames, otherwise the jobs will not run.\n\nChange-Id: Ic976765ef4aec30d8153a20aba5ab389265554bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97aa399ab47b2e11dcb597f8fba15315d7fba41b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97aa399ab47b2e11dcb597f8fba15315d7fba41b"}]},"branch":"refs/heads/master"},"b64fd1d02f576d0bcd581e692701d2a1cf024c63":{"kind":"REWORK","_number":2,"created":"2021-04-21 08:25:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/20/787320/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/787320/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/787320/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/787320/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/787320/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/787320/2"}}},"commit":{"parents":[{"commit":"d3734a376d735700e33668f933ef2cada26258ac","subject":"Merge \"deprecate RHEL as base of images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3734a376d735700e33668f933ef2cada26258ac"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 08:06:16.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-21 08:25:26.000000000","tz":0},"subject":"CentOS Stream: update kolla-ansible Zuul jobs","message":"CentOS Stream: update kolla-ansible Zuul jobs\n\nThe kolla-ansible jobs were renamed from centos8 to centos8s when they\nswitched to CentOS stream. The Kolla pipeline must reference the new\nnames, otherwise the jobs will not run.\n\nChange-Id: Ic976765ef4aec30d8153a20aba5ab389265554bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b64fd1d02f576d0bcd581e692701d2a1cf024c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b64fd1d02f576d0bcd581e692701d2a1cf024c63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
