)]}'
{"id":"openstack%2Fkolla~787473","triplet_id":"openstack%2Fkolla~master~I9ee58429731a8df86d8a730833a529638cfc6f29","project":"openstack/kolla","branch":"master","topic":"prepare-for-kolla-wallaby-release","hashtags":[],"change_id":"I9ee58429731a8df86d8a730833a529638cfc6f29","subject":"[CI] Remove an invalid entry","status":"MERGED","created":"2021-04-22 07:01:07.000000000","updated":"2021-04-22 10:03:14.000000000","submitted":"2021-04-22 10:01:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787473-prepare-for-kolla-wallaby-release","meta_rev_id":"00a592e4e92044df174b949174f885c936dc6598","_number":787473,"virtual_id_number":787473,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-22 10:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-22 07:16:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-22 07:33:04.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-04-22 07:33:04.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 07:16:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-22 07:33:04.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-22 08:34:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98471ebe384f49a23b50dd8505d5ed749e80b592","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-22 07:01:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e17f383ec378ad78667b6805d51b95d4347f3b16","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-22 07:16:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0568635af675e9e223304bb66eeadb160c321705","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 07:33:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"645b05523b9cd33e2f8829fa62490ebe91eeaabd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 08:34:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/502e68fdbe6a4c0ca255bf763e0a5957 : SUCCESS in 40m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8488bedc16284dbabd23f50d1ca7a3b1 : SUCCESS in 59m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8835efd60aa14f90aa281ef6093c758f : SUCCESS in 1h 27m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/37f2e06d00d24a28af5660529e79b1f0 : SUCCESS in 1h 16m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a2b790a8b058453fb61f2135690ecf6c : SUCCESS in 1h 28m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5e38e5e3a6294699a8c4412a8bbecf25 : SUCCESS in 39m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ee0095c5905741d884f6209a8ef52b8a : FAILURE in 19m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/18a32878603d455aab2abc5c86ef4c5e : SUCCESS in 1h 05m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e168afa77c66465f9c0a85c9453ba1d1 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1716da8ee9954d31b790e439609475e1 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d55195da034843d2862258cd6f47fcb0 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcf1ce26ad9f47d8951e91416762c4ad : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c5d8057b7847598ade279998c99cd1 : SUCCESS in 6m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0e7cfd12fb264248ba81e3ec23e35c4c : SUCCESS in 22m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8fcabf6b5134405597ee6ae90a69a3df : SUCCESS in 34m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e25224752034e13a7930c27d1340b79 : SUCCESS in 1h 08m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8cb270608609423182596b40da27b0fc : SUCCESS in 1h 08m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/efbca7a3e7b9436f9138d2efddfd8f71 : SUCCESS in 1h 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dbfae975924394d46365abb7b04c37821aa0ad70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 08:34:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"683e31dd1b506658d537bd5335b8c1ccffcc14e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 08:51:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/02856d2bd66e4f1ea56458b6a4ed3e15 : SUCCESS in 1h 47m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/70d008c827bf4ea2a96a51fee8142235 : SUCCESS in 1h 33m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a1dbfcea0e67f283de5b8adc3f5ac8ac1727dae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:01:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e56572fdea014444ab07039c9b717f7e : SUCCESS in 42m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ec56bea4babf4073806d7cceab0021d5 : SUCCESS in 1h 07m 41s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bacc46ba5d5848ddb33ccbc9aeeae76b : SUCCESS in 1h 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0573c24fc0a44f70a1a66473755c1d5a : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/067f60beb4684e8d9fbc305890f91416 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58ccbc9146b841d593e137303b8f3f6f : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3841675f250b4283aa6c218b51346f37 : SUCCESS in 7m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f43aec6e42d4226bc382f0ecd1560d3 : SUCCESS in 31m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/54de2d1e4cd34e7a8590c26eb0a2f899 : SUCCESS in 57m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b02f6ee9e14a49deac4dd2ad7fcf54ff : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9453f2de881fb5a914d4e3d20c0eb2bfe0cc279a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:01:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"00a592e4e92044df174b949174f885c936dc6598","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:03:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba5ea49489c44b789bf3f9e622adc923 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ade7ae9c685a48e242e4641f034237305f688677","revisions":{"ade7ae9c685a48e242e4641f034237305f688677":{"kind":"REWORK","_number":1,"created":"2021-04-22 07:01:07.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/73/787473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/787473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/787473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/787473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/787473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/787473/1"}}},"commit":{"parents":[{"commit":"44b193a2dc5a988ea0fd7ec5dc89a6147600ae85","subject":"Merge \"rabbitmq: Move to packagecloud\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44b193a2dc5a988ea0fd7ec5dc89a6147600ae85"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-22 06:58:52.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-22 06:58:52.000000000","tz":0},"subject":"[CI] Remove an invalid entry","message":"[CI] Remove an invalid entry\n\nopenstack/networking-ansible does not exist, it\u0027s under x/\nand not included by default.\n\nChange-Id: I9ee58429731a8df86d8a730833a529638cfc6f29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ade7ae9c685a48e242e4641f034237305f688677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ade7ae9c685a48e242e4641f034237305f688677"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
