)]}'
{"id":"openstack%2Fkolla~787503","triplet_id":"openstack%2Fkolla~master~Ic9ae2202492de7c99b812f10669f07000609c551","project":"openstack/kolla","branch":"master","topic":"prepare-for-kolla-wallaby-release","hashtags":[],"change_id":"Ic9ae2202492de7c99b812f10669f07000609c551","subject":"[TrivialFix] Remove unused aarch64 repo mentions","status":"MERGED","created":"2021-04-22 07:57:27.000000000","updated":"2021-04-22 10:57:48.000000000","submitted":"2021-04-22 10:56:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787503-prepare-for-kolla-wallaby-release","meta_rev_id":"ef19c5a60ca3e56677c556f87e21bc81ce302aa5","_number":787503,"virtual_id_number":787503,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-22 10:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-04-22 07:58:10.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-04-22 07:58:10.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 07:58:10.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-22 08:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-22 09:23:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35f59cb9009f52b5e8f0ec0f54744af0855647f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-22 07:57:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"871234283a1f426c277c40807e0be05c74160a77","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-22 07:58:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"994648c31987e68646b0f3a830bd19485d1a4c1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 08:57:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b13dfedd8f5a44eeb73663279f4d282f : FAILURE in 53m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58b7043feb424a1b90751e58af410ffa : FAILURE in 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37955dced97fb72bfa24068ec20277ecb7145023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:23:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/22b1d8575b594065a783ebf80fe31222 : SUCCESS in 26m 08s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5b64f945fab44c4c9ff21546cd360d6e : SUCCESS in 46m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e8749033edb54e67839ec8dec9745076 : SUCCESS in 1h 14m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e026b20c8460491d84e3cac8fb4fef4a : SUCCESS in 1h 08m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/760ca71b65a646d2a50a6c54e7aea38f : SUCCESS in 1h 24m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f400bb76af374849bd852b967a31c449 : SUCCESS in 41m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6403ffc6144f4a8dbffee33f0e54b1f8 : FAILURE in 20m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aa4ef538d6f04efcb5061504bdb36f3e : SUCCESS in 1h 00m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd56b3f6bdd24451980514ff54b84a51 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f34a2013bd445168078fd07829f8a7e : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e64b0e5cd184382b87692e738a6d977 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afe73bc249bb4951b56969109c3ca734 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa5a78fe9ca4a6cbd9175694a900387 : SUCCESS in 6m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/709d0fc0cb8f46149d00dac4787e350f : SUCCESS in 22m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9aeda876a3f8498f8e9fe363231d213f : SUCCESS in 41m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4fc8e082b40f427aa5fd78a595ee6c13 : SUCCESS in 39m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/95e473d5e4de45cbb806fc798027ead3 : SUCCESS in 54m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/54a4656301324bb2be3c68cc956df553 : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9c603f77bd108634433857046bee36615085cbeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:23:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7cd6eaf782b179d225b72e94958e3f85c42165bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:56:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c6e57a7da383466791ebcc70aeac5bda : SUCCESS in 42m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/14ea43d2b9fa4d29b5e1a1d08e6aee87 : SUCCESS in 1h 06m 29s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d7077e3375a54266a9748bbe5de922c0 : SUCCESS in 1h 05m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a457c513a64079a569784926a3e451 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1a4a7c733ab4dfba8d811342e06c87c : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c100adefd55e417ea09de443dd1053b4 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d73912adc8c45268e205571ce7d72b3 : SUCCESS in 9m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2942fc04d6742879e6b8609f85bd039 : SUCCESS in 1h 02m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4ce918c02554e868a8a5cb9b49df95c : SUCCESS in 52m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/150f046115764ad3b5d24570142be446 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":1},{"id":"222bbf800e27fbebb85885c056944dcdf442b9e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:56:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ef19c5a60ca3e56677c556f87e21bc81ce302aa5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:57:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a481f564276f44b99e98293e3c90fdc0 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c842d925dd6177bf9e878085e2dd36feb3fe1c7","revisions":{"0c842d925dd6177bf9e878085e2dd36feb3fe1c7":{"kind":"REWORK","_number":1,"created":"2021-04-22 07:57:27.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/03/787503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/787503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/787503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/787503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/787503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/787503/1"}}},"commit":{"parents":[{"commit":"44b193a2dc5a988ea0fd7ec5dc89a6147600ae85","subject":"Merge \"rabbitmq: Move to packagecloud\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44b193a2dc5a988ea0fd7ec5dc89a6147600ae85"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-22 07:56:50.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-22 07:56:50.000000000","tz":0},"subject":"[TrivialFix] Remove unused aarch64 repo mentions","message":"[TrivialFix] Remove unused aarch64 repo mentions\n\nChange-Id: Ic9ae2202492de7c99b812f10669f07000609c551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c842d925dd6177bf9e878085e2dd36feb3fe1c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c842d925dd6177bf9e878085e2dd36feb3fe1c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
