)]}'
{"id":"openstack%2Fkolla~787986","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Ia4d326601d14c5d55e8f67d8996458fc796583ea","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"Ia4d326601d14c5d55e8f67d8996458fc796583ea","subject":"Disable trove for Debian/binary","status":"MERGED","created":"2021-04-26 17:41:33.000000000","updated":"2021-04-28 18:20:26.000000000","submitted":"2021-04-28 18:19:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787986","cherry_pick_of_change":787958,"cherry_pick_of_patch_set":1,"meta_rev_id":"687089816c0c517d77c412c5962823c22244f6db","_number":787986,"virtual_id_number":787986,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 18:19:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-04-28 08:19:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-26 17:57:06.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-04-28 08:19:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 17:57:06.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-26 19:14:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 19:42:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 08:19:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"286bd0cdda7d1ec2cfba86cf2573ccabd17a040f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-26 17:41:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2ff8dabe6b5d28f8ae3d5f8af6467a4c7ecc0b00","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-26 17:57:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8002126f9baea5da67e77bfeb1078301b9659af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:14:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/05038d2c92064851a0ea4b6351be9ff9 : FAILURE in 1h 28m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab431f524aea4cfeb700f6cafcf7e13f : FAILURE in 1h 30m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38c388c556d64d42a7697d9ff4c37bea : FAILURE in 24m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b525a93155324e8aa54457eeeed9f0e6 : FAILURE in 1h 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1436beddc47610cb3776e16e21149468ead54d61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:42:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/12bc74ba152a42da90cfd74a1660f707 : FAILURE in 1h 59m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f147700b7ca24374945c26ab28c26f80 : FAILURE in 1h 59m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b095ce6e9a7d43df976daaa1d3e2a1e2 : FAILURE in 18m 35s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8c76a9ee64da460e9287f616e8398afc : FAILURE in 23m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fb0c64bc96854577b0b2c2fb551fa5d3 : FAILURE in 1h 10m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/affc405e1a1d47a4924d63c332d668e7 : FAILURE in 40m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/57be2e05264c497984ef734ea6578b31 : FAILURE in 21m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f1d90eaaa6634b3b99dc3a8b834d76e1 : FAILURE in 12m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348ea93b01194c83917407a545fcbbca : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b03f780ec0424d1e942f5b999606589b : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e915872dc6184625a7ae4a6b0e8029a9 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ffafe9e2ebe41afa9822ee6af4c54aa : SUCCESS in 6m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb58a509d560415f8bf67df6fd33ccdc : FAILURE in 22m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6821cd8af37b4062ac8bbadfb7e5d1b2 : FAILURE in 24m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c0002a0f7d74b23a0ae79854c7a4fba : FAILURE in 16m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/45a77c088ee54234aa51f1b7910039c0 : FAILURE in 20m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d2e9dfe30ca642c19277a69ac2269d74 : FAILURE in 17m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9e648b88be226449f8b92c42c348e2fd0cd3638e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-27 16:12:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a8714f74b85e5fe3930d5051a1737c9297f07071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 18:16:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c550035e41e7497887cf99bff56f70b4 : SUCCESS in 1h 35m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e8551bcc0b6e4cde913c1e706b01b44d : SUCCESS in 59m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d3a264b89c514093b8b18402e68a11ab : TIMED_OUT in 2h 01m 47s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6d28ff5cdf0140c48af0468964cd7a5b : TIMED_OUT in 2h 01m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9edd4dc8aaef436e9c7ea9ec673fd98d : SUCCESS in 1h 09m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/94d7482903bd40febe7225caf24c9b46 : SUCCESS in 21m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dd43c3a60546430cb4a9273efda9dca4 : SUCCESS in 18m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f98744deff884f9cb7629b5117655fbb : SUCCESS in 54m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eb2c25cd73d4bbab66bf9c6bdf0ab78 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a1cd174d1b841a591d4ec90927c5c45 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf0217ddb76a425b8790ab29cb8f734a : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b2e156279a443cb8b1a70e94ba5a55f : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/27cdf451e0994b35bb74b0bc559556d5 : SUCCESS in 21m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/95de630706ae471a9b623a9c458e2fbc : SUCCESS in 24m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/100e3e10b68b4ccea760fac7e48d1203 : SUCCESS in 47m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/51b0b47ff6764b599a561d4788d5eb8e : SUCCESS in 46m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/035414ec86b74331b67856da564a72d5 : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f5e800114fe63422aecb40fd2bf06dc04029cb59","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-27 18:25:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0ed087a383554c317f8a4585c9dc93f373a9c89a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 18:46:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/167bc2002373416c9721fbeffc56a84f : SUCCESS in 2h 02m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c367b33222d4ba189c2812a4cc40106 : SUCCESS in 1h 40m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9cc85fe82d4a4137a358ce114c9f0d92 : SUCCESS in 2h 30m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8a0d36753544d81bf6a6f472cad8bbe : FAILURE in 6m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02cad172e6cf7abf660b7d3d136779627bdb4d99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 19:37:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ba4c53cd54bd46419f17d0b48cdc51ea : SUCCESS in 22m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cc9e08c5006f449695e90243d2776eaf : SUCCESS in 23m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a7666fe64eb74df8ae5a907829262111 : SUCCESS in 1h 01m 15s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/22492bbd06c148b98aa216bb5ae4cec4 : SUCCESS in 1h 03m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8eba796b99b24c7f89c4628b0420d2f5 : SUCCESS in 1h 08m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2913a3ed255e4f419afe4d20ebf23c46 : SUCCESS in 19m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1cf38525d9fd48a9bb6a8d60323cd65e : SUCCESS in 21m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7651e95fd9234e65803660a3d5e63026 : SUCCESS in 55m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d65acc3f42a04b23bb968fbc81ff64b2 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f6086dbbeb84465bf1b178c035041ad : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/418c679b42714e33abf25939e8c54b9e : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e51f2039497c462a8e1f49ec3581fe4f : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/98c916f6100d46d28fc3ed2b731c0083 : SUCCESS in 39m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d8c0d18905b40269799e1697ede71b1 : SUCCESS in 24m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3676654e7c2d446b9ecab96816d6c5d6 : SUCCESS in 50m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a6b93ee5557f4dad815147ac23f86248 : SUCCESS in 53m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4cff0f9677f841b4a30df7563843519d : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":1},{"id":"937ca202fcf2952f54e6530a1cb4dd4775c4d828","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 08:19:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e6e037ed2aefc70ec6434ffb10993fca3a7f1f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 08:19:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d87195739804f9e40e1022e74d96d3f16e029046","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 09:25:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/342368aba546460ba89614afbfb70de8 : FAILURE in 6m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/404bcde4f47d4a518233e81ecd043113 : FAILURE in 10m 20s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a9bf2b97c30e45a9ae32b8010e7139a4 : FAILURE in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f4758c9b0243a2af3492bfb7bf66f6 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/853a172a437f415ea45863835689b5a1 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12c7db8b7fd7408da1981ac97ff37d5e : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acec8eea0619442584a41302bf4fd0f7 : SUCCESS in 6m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/928b9b6f8bdf4c4cb079826ad65e79cc : SUCCESS in 21m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/65a2030f6c5b4df182547d6d38d4aa5b : SUCCESS in 52m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c7a158c35fa54d8d9c8ddcf50ee45e0e : SUCCESS in 51m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3a854b8445bda7492c606a55aa84fa7b55a47b3c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 12:11:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3bfab57144ce14a44122b7b4104b6d2220242316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 13:25:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ff7a40f7c51f485ebdfb1e4e296d6b64 : SUCCESS in 24m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e554673d211345ea9bad6fefbf5e6335 : SUCCESS in 28m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/374342dbf68046f8ade9d57f1f6ef859 : SUCCESS in 1h 08m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/354168d022bc411d87778bfdc45a2855 : SUCCESS in 1h 02m 22s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a154d963dc25412f88567601d58b0989 : SUCCESS in 1h 07m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/db2dfd4c180448ef9eae20d61da435fd : FAILURE in 17m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2953f8be59c44d96a9c84c348fc75ab6 : FAILURE in 20m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2074cee0b7124a27b41fb919dfa9b02b : SUCCESS in 54m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c16a99e90c54118a56469926179eaa8 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/822ce88f001245f3a5357b2ae569148b : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34edcdd383c34a988daeb2c43c21783b : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f470928016fd447b84bae232e1563346 : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0dfb27edd87f41eeb63d9d4a270fa01b : FAILURE in 21m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/27b07f600df6451493f61840369a9f2a : FAILURE in 25m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1b474990cef40ce8ab44fac3c6e0eb7 : SUCCESS in 56m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ec8fdff853f40c9a919084d5e01f71c : SUCCESS in 52m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4d1de0ebc1454f938242b97bb8a49b01 : FAILURE in 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b9f9ce3f89e36ed2525b8993bfb946d4362d3be6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 14:18:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/95138bc4afcb4e81ad65d086bf07ff49 : SUCCESS in 39m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04cbcb03a4664039a10f88c2fa1d8cad : SUCCESS in 1h 37m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/65a5034e55774f7e9d8086e46a165505 : FAILURE in 2h 04m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/16d2c240b34b4bccae69c69690bf9bff : SUCCESS in 1h 40m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fac98ce9be2b3264478c030507d9c94ec23f702f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 14:52:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"86910772287ec24c28407b77c1935d3aa4eac70e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 16:15:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/676b214f66a0452696589ad24c8ee6c1 : SUCCESS in 33m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a2d72c7c79ca445c9a70009d385f3541 : SUCCESS in 35m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2c1e11964faf4593b6e9a20df63f2b16 : SUCCESS in 1h 17m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/054686d473404cdfb67b3d8bed99c93c : SUCCESS in 1h 10m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a41f0b64b6ec46368f2637d240e0891c : SUCCESS in 1h 06m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/819ae3c4838744ac847c40ab73b6ca3d : SUCCESS in 21m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4e822111bd524c38af6fc0028cd4f75b : SUCCESS in 22m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/583ded253224443ca34aab3e0111c127 : SUCCESS in 1h 10m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89b0be9b4982405286e9fc01d019446c : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30539382cbdb4330baf8c33a6806a1ff : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8e60cee88f149ebabac9871407d64d8 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e8cd32ca2b4aa3985494e91a33184c : SUCCESS in 6m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0045486e29134519b21f93a09df6b135 : SUCCESS in 33m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c807a479492f4e9baf35d681ddd22847 : SUCCESS in 34m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/51c3063baa594ca4a0623a2266e3de9f : SUCCESS in 1h 04m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/41c2633fde684f8b8e2a85f7f3e92c79 : SUCCESS in 1h 05m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d7a6a912c9e44ec962bbe05c42fcea3 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0b40e13f58068ac6b2f3cc8b1f4772f42f22a6f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 16:16:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89bfd262b7d18ff4cd1e5593eb4de6a71b3daf54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 17:34:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/499a23391abb4d169ea39e32705edd71 : SUCCESS in 52m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/55f84a75f13a44d78aebde7ae0e9e915 : SUCCESS in 2h 15m 59s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c47a4edf61eb400a9fcec6ce034b26bc : SUCCESS in 2h 39m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2bce553fc40042f897cdbea358b72f13 : SUCCESS in 2h 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18c9c00a857fe851b2c26a261dd6502b0cc6d7e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:19:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/84f72f48b190433fbd3dbf3c58283a77 : SUCCESS in 27m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0b540d4fa980415daf55543ec959ec56 : SUCCESS in 1h 52m 31s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f96325d28c634b45969c25a2cb31a1e5 : SUCCESS in 1h 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b1044416c9443d6bdbdd01f1050490a : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd26008f370c4370a21bd8fc9b8c3e11 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96806856f9734d2fbb1a5f06f1881c4e : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc96ffbc24c41e69daaba3c1d31e459 : SUCCESS in 9m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/98e080369e3842dc9dd842713e907b5a : SUCCESS in 22m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc75f172406b4f3abeb2c735d516c0c8 : SUCCESS in 47m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d0d02db3dc2e48dcbbd7b8f779cdba94 : SUCCESS in 1h 16m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2e77ac7430c3cbb92905d52ad01bffb502bdb36d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:19:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"687089816c0c517d77c412c5962823c22244f6db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:20:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdcd279443e2465ea1fe00831429ff27 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c366e9fbc3294fd5f9822c31538766deeeafbdc","revisions":{"4c366e9fbc3294fd5f9822c31538766deeeafbdc":{"kind":"REWORK","_number":1,"created":"2021-04-26 17:41:33.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/86/787986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/787986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/787986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/787986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/787986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/787986/1"}}},"commit":{"parents":[{"commit":"cc310a525c9c2f1add957abf7a10caa5715d39d2","subject":"[CI] Publish images to quay.io on a daily basis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc310a525c9c2f1add957abf7a10caa5715d39d2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-26 11:15:33.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-26 17:41:33.000000000","tz":0},"subject":"Disable trove for Debian/binary","message":"Disable trove for Debian/binary\n\nPackage installation fails:\n\nSetting up trove-common (1:14.0.0-2~bpo10+1) ...\ninstall: cannot stat \u0027/usr/share/trove-common/api-paste.ini\u0027: No such file or directory\ndpkg: error processing package trove-common (--configure):\n installed trove-common package post-installation script subprocess returned error exit status 1\n\nChange-Id: Ia4d326601d14c5d55e8f67d8996458fc796583ea\n(cherry picked from commit 51c3c90af9a0c4ed27b4af5e0650b4e3a9eaeb53)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c366e9fbc3294fd5f9822c31538766deeeafbdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c366e9fbc3294fd5f9822c31538766deeeafbdc"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
