)]}'
{"id":"openstack%2Fkolla~788073","triplet_id":"openstack%2Fkolla~master~Id3f824b741a5aa98efc7162fb7d49657e86a2bee","project":"openstack/kolla","branch":"master","topic":"add-swift-lock-path-and-set-permissions","attention_set":{},"removed_from_attention_set":{"30609":{"account":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"last_update":"2021-11-23 12:52:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id3f824b741a5aa98efc7162fb7d49657e86a2bee","subject":"Add Swift lock path in Swift containers","status":"MERGED","created":"2021-04-26 20:16:13.000000000","updated":"2021-11-23 12:54:57.000000000","submitted":"2021-11-23 12:52:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788073-add-swift-lock-path-and-set-permissions","meta_rev_id":"a153513b51ebdfc589e73d07dfa990d92c305851","_number":788073,"virtual_id_number":788073,"owner":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-23 12:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-11-19 08:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-11-23 11:24:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-11-23 11:24:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 21:25:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-28 10:09:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-11-02 15:16:54.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-11-23 11:24:01.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"fe11694d1e859204ad7f12c072911ecce4e086e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-04-26 20:16:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dd1be37ad09154e0764767a4b8279774522a5b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 21:25:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e61947ac06124deba280299a557e60b3 : SUCCESS in 24m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f12795da9f04484ea4dfbba5637a0bdb : SUCCESS in 34m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d925344e7e6c48fdad8fe979e732e3d9 : SUCCESS in 1h 05m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/af21aad1f8d84f869e95593bbf013590 : SUCCESS in 1h 05m 18s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b71a8147d04e4e49b0d1b7a8867da9e9 : SUCCESS in 1h 08m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/39af8687758f41eca48d135d37a290fe : SUCCESS in 34m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/25a06d05187d4f138e481bfc22f80f41 : SUCCESS in 19m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/285cdcb21ff5417ea164eadc5be2f8ee : SUCCESS in 57m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b323ec5ec04b55a771bfedca8e8b9e : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b104adb1a73a4123ac4a274c3cfca5d8 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5078fa569c3b4e949b4a8fe9d4876736 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/affa17b0c4f742bf97c3c6bcb49cb737 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aec087672964aec8c6b8ae35dd6b528 : SUCCESS in 6m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2e9f187294214a8f94d0416f697b9d9d : SUCCESS in 16m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a44dfb346e74d969922f8e3eeeffe3a : SUCCESS in 37m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ead6fba8fb94983a31b9386e480dfe1 : SUCCESS in 1h 05m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a0476e5e9073451ebea51ff8b9d5c87b : SUCCESS in 47m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2d0055bf4f3d4f8a969e21512925a74c : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":1},{"id":"516993b4a4806894a6053b9e66f86173938cad09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 21:58:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1bebae521acb4c238dec1c07e5812c64 : FAILURE in 48m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/657386630a6744bfb11f026fd7c9b0b6 : SUCCESS in 1h 35m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51a076e858271dc870a25af0716c1f7e8a607dc6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-28 10:09:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f6c95184f466b69e4c34f6cf3ee7ffb85eef5d1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-11-02 15:16:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca65fa02dc3887d002076656812280de49311455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"date":"2021-11-19 01:21:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ccd55e43dd2e38bf2219aed6df565eb252729f1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 02:58:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/06eb8880777f454d99c0f0afe0dab42c : SUCCESS in 26m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e7117b0f5d734f9cb5e6547a9077bf5d : SUCCESS in 34m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/15fc663d1ad0451b8b0591abd047f80d : SUCCESS in 1h 04m 48s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8aaf8230adcd4507b866700ed7002955 : SUCCESS in 1h 09m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cd6b508d1fc640008b3d8b9bdf568af6 : SUCCESS in 1h 33m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/66eef346042c499ebec885a550e3f374 : SUCCESS in 21m 42s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a0a50a0669054e6096ab1c688313edc0 : SUCCESS in 20m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4a32517738424123a64038860da0fdfe : SUCCESS in 1h 11m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/aa94b25010ab49bb91136875fea1f424 : SUCCESS in 55m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e791a4962bb4dd89fdf4d3d810d0f30 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e71653e6db2443b998d25c8bd563037 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d8a669afed74eaba16e9c132af5926c : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c495e422535c48f19ee678797a42be98 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dba5b70b2bc475e9e614a98c4dea58c : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f33397229ec4d75aa7c429f7ca2d783 : SUCCESS in 3m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f30b56a99fa34d55b3c79e675418d940 : SUCCESS in 17m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba7a3b887e204224a117f46890d5e010 : SUCCESS in 30m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac96cfc98c884e0a90d6b2baa6b9c252 : SUCCESS in 56m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd11014817fa4f8d8251d7f745d4eb9c : SUCCESS in 36m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ae2521cfd36494e8b579eb487b111b1 : SUCCESS in 1h 08m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a22979bd8b9b4a9a0d90b7af63f14f250d8de67d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 03:15:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dfb0c43b37f54eb4928d56ad142efec2 : SUCCESS in 49m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/34ee7fda650b47918f8c64c06889b5a8 : SUCCESS in 37m 37s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d4335fd5d404d3ab7fa29324053e095 : SUCCESS in 1h 52m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b6f123f30ce47ceb323860f475142a0 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9b704103b8345c4b901b4d4f4d9817e : SUCCESS in 5m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4048c26f6d84b61bb1bf7560a9485a4 : SUCCESS in 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0450232d0140a6b419b839c7d0347be4de0afba1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-19 08:43:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"826ea859bfe0932e24630ebd29ddaa46c8ec3a00","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-23 11:24:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2fe987a4e3e6b72dcdb704434d4dc474e1902fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 11:24:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"09e5e056f79735bf44de74e72837f90a62c181c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 12:52:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/86ec98850d56483cb63c02484f7035dd : SUCCESS in 54m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ded35628ffcf4d5391c280cae9506d5a : SUCCESS in 1h 02m 38s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/33bf680556f04b0ca4fa7537e6c5d764 : SUCCESS in 1h 26m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/70444ac7e22248b890691c9fc48863bc : SUCCESS in 30m 01s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f1d34c1917f1465ba807940e8256741f : SUCCESS in 1h 02m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/968dd269ea4b46a79fb9dde37ed589bb : SUCCESS in 1h 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/805bab539f204888888f72c88ab9ed2a : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bfbe7c74d22475eaf1bfb13b9c5a742 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9bfbd64aad342c18e6924012a88c900 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70b267b771e43e09e33bee30ceff960 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/264090248394420b8da4901d180b064a : SUCCESS in 4m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cba5a9e82445411abd9135275a5b79e3 : SUCCESS in 34m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e732d13ba1e54bd590cf06871f57ab4c : SUCCESS in 50m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e74f5780b1d475f8c295dfe882ef0a3 : SUCCESS in 55m 29s","accounts_in_message":[],"_revision_number":2},{"id":"3e353a2666811c2e85c8425afc524286141e5bde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 12:52:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a153513b51ebdfc589e73d07dfa990d92c305851","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 12:54:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac4694df09d4842994b25c6abd56f89 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfee6cf9d1024b1b8536d1631bca2926 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff25b500d23155485dc57d10aed9070a4cba88b4","revisions":{"f3ac9f2b9e253e661f94913670b485738d7b4501":{"kind":"REWORK","_number":1,"created":"2021-04-26 20:16:13.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/73/788073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/788073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/788073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/788073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/788073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/788073/1"}}},"commit":{"parents":[{"commit":"af17cc180e40cc19fb08fb3186de43d5a6c7d83d","subject":"Merge \"Disable trove for Debian/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af17cc180e40cc19fb08fb3186de43d5a6c7d83d"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-04-26 20:06:58.000000000","tz":-240},"committer":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-04-26 20:06:58.000000000","tz":-240},"subject":"Add Swift lock path in Swift containers","message":"Add Swift lock path in Swift containers\n\nSwift-recon-cron requires rw access to the lock path\nspecified in in object-server.conf file. Currently it\ndoesn\u0027t exist in Swift containers\n\nCloses-Bug: #1926203\nChange-Id: Id3f824b741a5aa98efc7162fb7d49657e86a2bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3ac9f2b9e253e661f94913670b485738d7b4501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3ac9f2b9e253e661f94913670b485738d7b4501"}]},"branch":"refs/heads/master"},"ff25b500d23155485dc57d10aed9070a4cba88b4":{"kind":"REWORK","_number":2,"created":"2021-11-19 01:21:07.000000000","uploader":{"_account_id":30609,"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","username":"sdmitriev"},"ref":"refs/changes/73/788073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/788073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/788073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/788073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/788073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/788073/2"}}},"commit":{"parents":[{"commit":"af17cc180e40cc19fb08fb3186de43d5a6c7d83d","subject":"Merge \"Disable trove for Debian/binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af17cc180e40cc19fb08fb3186de43d5a6c7d83d"}]}],"author":{"name":"Stanislav Dmitriev","email":"sdmitriev1@gmail.com","date":"2021-04-26 20:06:58.000000000","tz":-240},"committer":{"name":"sdmitriev1","email":"sdmitriev1@gmail.com","date":"2021-11-19 01:20:26.000000000","tz":-300},"subject":"Add Swift lock path in Swift containers","message":"Add Swift lock path in Swift containers\n\nSwift-recon-cron requires rw access to the lock path\nspecified in in object-server.conf file. Currently it\ndoesn\u0027t exist in Swift containers\n\nCloses-Bug: #1926203\nChange-Id: Id3f824b741a5aa98efc7162fb7d49657e86a2bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff25b500d23155485dc57d10aed9070a4cba88b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff25b500d23155485dc57d10aed9070a4cba88b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
