)]}'
{"id":"openstack%2Fkolla~788304","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I543965a2741cebfa759576a4c75669a7bacd4208","project":"openstack/kolla","branch":"stable/wallaby","topic":"prepare-for-kolla-wallaby-release-stable/wallaby","hashtags":[],"change_id":"I543965a2741cebfa759576a4c75669a7bacd4208","subject":"move to Debian \u0027bullseye\u0027","status":"MERGED","created":"2021-04-27 14:06:41.000000000","updated":"2021-04-28 13:57:18.000000000","submitted":"2021-04-28 13:55:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788304-prepare-for-kolla-wallaby-release-stable/wallaby","meta_rev_id":"6764735e7bb14472650a7937c33acf07d34c2744","_number":788304,"virtual_id_number":788304,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 13:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-28 12:32:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-28 12:04:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-04-28 12:32:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2021-04-28 13:57:18.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-04-28 13:57:18.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 15:49:18.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-27 15:54:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-27 16:10:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 08:34:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-04-28 09:04:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-04-28 11:00:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"}],"messages":[{"id":"b3e2fee2f7b3b29bccdc8745202f12d91aedafdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-27 14:06:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"614a5d59e09902af52225c11eafafdd237ed3247","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-27 14:11:08.000000000","message":"Patch Set 1: Backport-Candidate-2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"bfbfe3cc7f7dabe03de12902d84cc7deca537f1b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-27 14:16:12.000000000","message":"Topic set to prepare-for-kolla-wallaby-release","accounts_in_message":[],"_revision_number":1},{"id":"ed18b746f695f0ee44d62c91d533876ffde1173a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-27 14:16:56.000000000","message":"Topic prepare-for-kolla-wallaby-release removed","accounts_in_message":[],"_revision_number":1},{"id":"4661a7aa461a58beec12ac001652e3513ea7fc1f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-27 14:17:00.000000000","message":"Topic set to prepare-for-kolla-wallaby-release-stable/wallaby","accounts_in_message":[],"_revision_number":1},{"id":"11fe8af58836ebd82b4f87733277669a6e92153d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-27 15:49:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa6635e930dbaa0b0f5fbd5af35993f11f8aca85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 15:54:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c7cfb3136bd414caa3be0059c346327 : SUCCESS in 1h 44m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6fec495fa25483cb260ccd705e882ec : SUCCESS in 1h 24m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5e8e64c78840b016225b710d64dbbfb6d68aaf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 16:10:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a1224c844c1a47f3861bba593cea6dad : SUCCESS in 44m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8a762b153a5740cc9fff5d59d3c12ec4 : SUCCESS in 46m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9d83628dd4e843519a3486f1b250c7a5 : SUCCESS in 57m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d53d1d0218964bddaae3c5e79761f75b : TIMED_OUT in 2h 01m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/904df00791814b42b939e3e1f63e2b59 : SUCCESS in 1h 08m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f943be2e2cd24d94b2648200cf94c460 : SUCCESS in 41m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/461ec88279ac40d38e773a46691e2165 : SUCCESS in 22m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/edbc48e91cdc4bfa91e99c628e81de1a : SUCCESS in 51m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4dc04099c74ab9841a28a9582282db : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7677f7cc4c66459b8d9aa0a19b849a80 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c669e0565635435eb55c8d4a29d8c584 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e6c37ec878b4d42b0a1e04acbd043e4 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31f36cf876d408fafc57b9636372eb1 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c9b9d327e90e415e8da55bd69ddaebe1 : SUCCESS in 22m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7999e006cba9401b9bca8c5b52583764 : SUCCESS in 36m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/564735b28780484c89047ffd42d11dcb : SUCCESS in 50m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/696934d9dd5f4add9a16b2977904a604 : SUCCESS in 44m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7703afcb41334ab498da41844da73fe9 : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":1},{"id":"cdb001649a660aff68114944f6909206277d04d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-28 08:19:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ae96bff645ecfce1f1a9166538d09441cc3f2e82","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 08:34:53.000000000","message":"Patch Set 1:\n\nConflicts","accounts_in_message":[],"_revision_number":1},{"id":"377722bd841c7c78405c30319b192a73656c90b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 09:04:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"558bb17bd660eccbfacf94447fa11e4d99c6d21d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 09:26:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/682ef225af3f4b88b295547aa30381b5 : FAILURE in 7m 41s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/481d8c76d7c14db78b6e4cd811eae70b : FAILURE in 9m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/01fadbf8dc9c47c7ad9f5cc18700d3e2 : FAILURE in 16m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/307f52075aca4e03a9f5212db7831c70 : FAILURE in 12m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f12c5f888dd546fd935be74233f948d1 : FAILURE in 11m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/075cfe6cae3f42798b2f082e05822ea7 : SUCCESS in 35m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/43247240cada41e382c1478a76822028 : SUCCESS in 17m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/31131f824c4d4fca8503edd769676ebb : SUCCESS in 55m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fef6db02f4c42a8bea84ec806889917 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aa85cedf420450db3dbb9a978700f6a : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1385af0d64a64a089df979aa0c696411 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ae1b34d422a4a76be6b60728dcb183c : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c4aedfc9f54bf38ff18629890197d9 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d256f3b47c54947b6628659784cd846 : SUCCESS in 26m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3960734683748548f71dbe7f82a9e5d : SUCCESS in 38m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5b538f69c7e44ae69d48baa98e740abc : SUCCESS in 54m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e1609b15a1b40f79c78e0ce6342372a : SUCCESS in 52m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e24fc13be8384276b633c652efb887fd : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e3672d8c4cf74eaae55b741a5bc9fa8e174521b8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-28 11:00:26.000000000","message":"Patch Set 2:\n\nrecheck (CentOS EPEL mirror in OVH failed)","accounts_in_message":[],"_revision_number":2},{"id":"b36337c8ff41dd66f49c229cc1c78a4f294bfa1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 11:36:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/01f2a0d38d7447feb746dd3cf8507ba4 : SUCCESS in 2h 10m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43a62bfcbd1b48ee9d25595b9d0163fd : SUCCESS in 3h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"458cf4de4fc1cc8de94b75155f5028810761dafc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 12:04:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a1a3841b384b0cb244e527d19999d357c87417d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:09:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ddc2b7e4fd964337a796da23fe1896fd : SUCCESS in 24m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6a88a6c3ad2d47048dba7c2d91494eb7 : SUCCESS in 42m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f785d4ff173941ec931867d480cf243f : SUCCESS in 54m 02s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/51a7ace999de42c5890208805e7eec13 : SUCCESS in 59m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e5d9f943671d4f72afdf2c41fa9a2b1b : SUCCESS in 1h 04m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/af31c1e0c3eb43b5ba910c04668ff360 : SUCCESS in 38m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c78f9561aa2843cea7fe311c546c4d82 : SUCCESS in 17m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/017fd084801147bc95b3401cfb4107c6 : SUCCESS in 58m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49df3996f89b4cba802dbfb8e6f88038 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f44efe5e74e74561a7ab7eef7d4ea152 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39344c1b18bd4fac8c9712610494909d : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8300315ab84b4fee9c2939c3f9a2f9f8 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56348cd3044046ad87c65c6ba42dbadf : SUCCESS in 7m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1b2af5e3acb74fd3ad434cafa9ceff1c : SUCCESS in 27m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/29ed42d6d66f426fbe437c4bfe94b124 : SUCCESS in 34m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8457de3721ff406291e3685a133f2004 : SUCCESS in 42m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a75ed3654e234c7b9bcc6977b73abb1d : SUCCESS in 45m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e0b1473a36ce4be5a2d46059878535ed : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d6c4283f0c42f6db86711ba6f70238907056036f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 12:32:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"05ed6b2a101079fab3d5d7e15a4323ee434e1f44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 12:32:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1da3194374fcb509b800ee62f2aaf5b360750cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 13:55:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6e452441a8e84fb5979c0943c145f7fb : SUCCESS in 42m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c45476ac41e945e28fb58c60578e2f9a : SUCCESS in 1h 09m 39s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/707e1f2695c6488d83f2b9b79871c875 : SUCCESS in 1h 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1f1815f8b844e58c59b83484d5bb82 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05abc6e805ff4bf6851d363207175ccd : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bfe64f373374758b2b16e7ff492f5c6 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a43585d9064262aefe3bbb28dbfb37 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/89e34edc97da4dee952c1d419417dcef : SUCCESS in 48m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a835cc2728e424e91eea8f34d538cb1 : SUCCESS in 56m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7dbb1169feb54d04ab974eaec4973210 : SUCCESS in 1h 08m 32s","accounts_in_message":[],"_revision_number":2},{"id":"6b19d36012da2294313c4789defda0dbc95ef637","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 13:55:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"aa859e0881d43e91f77d9f146fe1667fe6ca9a9d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 13:57:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/604308fc51c64791a2c6ed824e1c2521 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1f5014e01104c2ccf275c2ed082b6861c331ac65","revisions":{"43ba709b0d1ca707e10a37f5be6ce8159bd33985":{"kind":"REWORK","_number":1,"created":"2021-04-27 14:06:41.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/04/788304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/788304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/788304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/788304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/788304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/788304/1"}}},"commit":{"parents":[{"commit":"b85b63ec9a603550e87f293b7278208667e9cca9","subject":"docs: deprecate ppc64le in Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b85b63ec9a603550e87f293b7278208667e9cca9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-21 17:41:45.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-27 14:05:17.000000000","tz":120},"subject":"move to Debian \u0027bullseye\u0027","message":"move to Debian \u0027bullseye\u0027\n\nBullseye is now in hard freeze cycle and goes for release.\n\nhttps://release.debian.org/bullseye/freeze_policy.html\n\nCo-Authored-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n\nChange-Id: I543965a2741cebfa759576a4c75669a7bacd4208\n(cherry-pick from master e0f8cbc50cf7184930a501b0f5452a64f523cd6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43ba709b0d1ca707e10a37f5be6ce8159bd33985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43ba709b0d1ca707e10a37f5be6ce8159bd33985"}]},"branch":"refs/heads/stable/wallaby"},"1f5014e01104c2ccf275c2ed082b6861c331ac65":{"kind":"REWORK","_number":2,"created":"2021-04-28 08:19:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/04/788304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/788304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/788304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/788304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/788304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/788304/2"}}},"commit":{"parents":[{"commit":"13abc31c2f34ef8cf8772c4209ad0102c839d2ec","subject":"Disable trove for Debian/binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13abc31c2f34ef8cf8772c4209ad0102c839d2ec"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-21 17:41:45.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-28 08:19:08.000000000","tz":120},"subject":"move to Debian \u0027bullseye\u0027","message":"move to Debian \u0027bullseye\u0027\n\nBullseye is now in hard freeze cycle and goes for release.\n\nhttps://release.debian.org/bullseye/freeze_policy.html\n\nCo-Authored-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n\nChange-Id: I543965a2741cebfa759576a4c75669a7bacd4208\n(cherry-pick from master e0f8cbc50cf7184930a501b0f5452a64f523cd6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f5014e01104c2ccf275c2ed082b6861c331ac65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f5014e01104c2ccf275c2ed082b6861c331ac65"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
