)]}'
{"id":"openstack%2Fkolla~788455","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~If21a21296489c1f243bb46f2595ff94d999bd686","project":"openstack/kolla","branch":"stable/wallaby","hashtags":[],"change_id":"If21a21296489c1f243bb46f2595ff94d999bd686","subject":"neutron: Remove vmware-nsx and networking-mlx from binary","status":"MERGED","created":"2021-04-28 15:07:33.000000000","updated":"2021-04-28 18:28:05.000000000","submitted":"2021-04-28 18:26:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788455","cherry_pick_of_change":788317,"cherry_pick_of_patch_set":1,"meta_rev_id":"3f971363c41a0ff30bea2d785473b7b8e846f30c","_number":788455,"virtual_id_number":788455,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 18:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-04-28 15:09:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-28 15:08:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-04-28 15:37:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 15:07:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-28 15:09:01.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-28 16:38:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f24a7668d7d502336fc9919906b3e5705f0a148","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 15:07:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4a4202217922c2c4dee69d5ab7511db0cb394e14","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 15:08:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e9e1b6cb19a9556d5fe946f33cfd335bbd38f8d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-28 15:09:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5310707fd29d34b6d09081908d28c1b893a036d1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 15:37:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"250804ab2f7558c37e791a619ca104db8f7b1bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 16:38:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2472bf4655cb43a8a32c1771cc4e941d : SUCCESS in 25m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a5bacd0013e24a22ad87fbde111841f9 : SUCCESS in 48m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/914b24fd7f394a06b78ca814d6a42f3f : SUCCESS in 1h 12m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2f0a817b9ccc4fb98413bd84e2972790 : SUCCESS in 1h 07m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1d276acd670c4b89a0a82c2e52a8554a : SUCCESS in 1h 08m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ae128b2652a64e80a90c9c8b3a82e39c : SUCCESS in 37m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9e01c1fff27d4ec5a468d8449182060a : SUCCESS in 22m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a09a4036622b456fbc40caa9f5c1a103 : SUCCESS in 58m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b5a31ea938146019db20cd0f65e3834 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/428b4817d9c94aa6b7f0411660376705 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a469463ffe94cb7a476ffdb60da7537 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a75ab8d022844679859156a6091aedff : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd90e5f20725441da111473fffdff586 : SUCCESS in 16m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6081c54e1f84326b8bad7717b01ecfe : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/45f3a76ae530492fafe432d76e2c3d0e : SUCCESS in 26m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0d9c3ce07cd4f7baac1eabc89549a17 : SUCCESS in 38m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d71f7928af2945179bcbf692426f0c7d : SUCCESS in 46m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5445ede06dea46deb8c32553449f04c8 : SUCCESS in 59m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/67bd0b1af48b4551b30978aa01c617db : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4acd1a4e3fdff66c91144aa9563e2d2c4de4d2da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 16:39:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"894d4ad81bda13fb73b591ac9b0f25d67f4b95da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 17:30:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6f4fdf6de1d74f06b8c2844b27ec57cd : SUCCESS in 2h 18m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/16cddf1fea954d5a9f546df94867b38b : SUCCESS in 1h 42m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cc0c99036da53fdd31d94d71cef3d3bcfdbd22b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:26:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/edac9c52a5b84690b0e4e940afaf4e22 : SUCCESS in 39m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1832a38f5ef84c28a177663463bcaa79 : SUCCESS in 1h 06m 49s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/95815cc21e7846959b1cf3da36aabaf0 : SUCCESS in 1h 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d90ba2e6cb9d4f9d84d8532d173d4b93 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f06899674d2426eb290ba7e34b8f144 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70e52469bc204c54aa26cd40294f5c6a : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3790f3b35ebd4db897bcd7094cb7b94a : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6ed4c1d4f1c40e6a71f0d5e0f817557 : SUCCESS in 6m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/21bfb5c0c9ee42a09623b07de21a1c02 : SUCCESS in 29m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e36d2bba51c42afb35615f9d1dceec8 : SUCCESS in 50m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac5a9a3f026b40ef8503f5e04cf3efc4 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c73defa564300243c2ed288994ba90e3c1c3cfeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:26:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3f971363c41a0ff30bea2d785473b7b8e846f30c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:28:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3acce4887594bcab468e570e39910b0 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52a27683b6534ea882f8bd22f2d8a562 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1ad81338870ad4b5a07b6a0590b87472f2277b7","revisions":{"a1ad81338870ad4b5a07b6a0590b87472f2277b7":{"kind":"REWORK","_number":1,"created":"2021-04-28 15:07:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/55/788455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/788455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/788455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/788455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/788455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/788455/1"}}},"commit":{"parents":[{"commit":"989fe5301ba0ffdd92208037fd2f098f31d549ca","subject":"Merge \"releasenotes: we move to Debian bullseye\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/989fe5301ba0ffdd92208037fd2f098f31d549ca"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-27 15:23:26.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-28 15:07:33.000000000","tz":0},"subject":"neutron: Remove vmware-nsx and networking-mlx from binary","message":"neutron: Remove vmware-nsx and networking-mlx from binary\n\nThose plugins have been already removed from source images, this\nchange removes them from binary.\n\nChange-Id: If21a21296489c1f243bb46f2595ff94d999bd686\n(cherry picked from commit a4f6a8f05b45838338ee1c6a78f278d49486ac03)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1ad81338870ad4b5a07b6a0590b87472f2277b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1ad81338870ad4b5a07b6a0590b87472f2277b7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
