)]}'
{"id":"openstack%2Fkolla~788462","triplet_id":"openstack%2Fkolla~master~I320d0795ce11ca00398cec6183ed140d31d3fd80","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I320d0795ce11ca00398cec6183ed140d31d3fd80","subject":"centos: do not install CentOS Rabbitmq repo twice on AArch64","status":"MERGED","created":"2021-04-28 09:35:07.000000000","updated":"2021-05-10 12:14:11.000000000","submitted":"2021-05-10 12:12:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"788462","meta_rev_id":"d99077e61a4eff15e806825d475ba411dfc0364f","_number":788462,"virtual_id_number":788462,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-10 12:12:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2021-05-10 12:14:11.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-10 06:11:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2021-05-10 10:58:46.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2021-05-10 10:58:46.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 10:46:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 16:22:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-05-10 06:11:34.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-05-10 10:58:46.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"64c7315648da85d9b8cc17dda84297741c2f8d7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-28 09:35:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18d9aaa335d46a575bf08755b6fc43f993136a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 10:46:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/571124850b9b4629a90af5e165167c73 : FAILURE in 8m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6e0370c5f1ce4f1abc0e95497f793adf : FAILURE in 9m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6ed0b12596f34ad28cb441c67366688b : FAILURE in 13m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5b70050509294147bc3473fa3d57e504 : FAILURE in 16m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/97c5fe0befe0456cb16b63e63a827c84 : FAILURE in 12m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2faa51ac2ec34ff7b075071836fa0a11 : SUCCESS in 33m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/91a906a3717344eab8bcf80bc393e080 : SUCCESS in 16m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/342da8effcc94277be7eb13695044e69 : SUCCESS in 49m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900e5a61a6c64c108790813b4414703f : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9169c8c47a30414d914f559ef5e1f766 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1c79e0d42864f4d84ef1703ad811119 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb13a52824e6493f8041c69470d96d97 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6711a2664cf04a15a10a7999e8d53cb4 : SUCCESS in 5m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/23d07e743e1f4eb09cd20b4e636195bf : SUCCESS in 32m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/704a32a7cca640f1aba5e03af325c140 : SUCCESS in 44m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4559b99925f4cd0a981e174011f1f15 : SUCCESS in 57m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/133159facee546ef9703657f33eced67 : SUCCESS in 51m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ed2d53ab80b54417824695c94dd98c94 : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ff8adc784dffef657330fbdef4f08ada946ac708","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 11:11:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d410826409349d38b219d2bf6246852 : SUCCESS in 53m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ad562589bd54cc181ca595c51626fe4 : SUCCESS in 1h 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b2007fba27d5079b061e8300a348140f11258e9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 16:22:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f4ee1112e26b1d3b5821cb5a30d9bc9b915be0e0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 16:22:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"99367d39e509c43e4a4272202cc35ab4dc0f3689","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 17:47:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e0333defebf436cbf8d558d411de7da : SUCCESS in 1h 00m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6665dd51a630462483dba616175e4e62 : FAILURE in 1h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e43c4264b0fa7e0150ee068ad877836e078944de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:37:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1f68ffa5268846c9a43901b3e896fd02 : SUCCESS in 18m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a3ff6262636443aa95a23ee1c6c23b3c : SUCCESS in 30m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3b5e1a88aa374cf38f0acffb7b75d589 : SUCCESS in 55m 25s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bef56543b36041a8996ad71fec725459 : SUCCESS in 59m 25s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/90baba84bbfb4bcd8d702b774c494b76 : SUCCESS in 2h 01m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/97c8dbfba87241189b4bf7ac4ec3aaa1 : SUCCESS in 42m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fac94c2708a44bf0a666262f6e6aceea : SUCCESS in 46m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4567a829d7614d8d96b6c2f80cd5098d : SUCCESS in 54m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ccf52e997c484496f504c640a72751 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/541601ef1ad445319f1846d83c05f891 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b25f2b6320b436e91f1afff454bba2f : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ae58c8fb7704437819f20d77b578e84 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ff19fc8c8a43d5820189a61bd266e2 : SUCCESS in 9m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4401bf6657c14b98b99dbc07fa269c1f : SUCCESS in 25m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f9047ad11e2b4a109f16a341de2e2b88 : SUCCESS in 2h 08m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/11d2f6086181449aa562616630a38eb1 : SUCCESS in 53m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e96e092263bb425c9490573d2d52bcc7 : SUCCESS in 40m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1fe618fd5b79499cb690efde72cbdcba : SUCCESS in 1h 16m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0517b1fe0cd90e18f4d9c3348ac51005f573fe7c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-30 18:08:13.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)\n\ngot confused while reviewing the queue","accounts_in_message":[],"_revision_number":1},{"id":"aa63bd9f4908ae8394ce62adef9a6a4a184cd1e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-01 09:33:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4d9bded5f986ad210b1ef1647df7c0c4fb8b4a5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 10:52:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ac909d29c4b4f13b8338aae44542a5f : SUCCESS in 39m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/230f65fb96644e0ab99e5f3de5894403 : SUCCESS in 1h 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a91a419ad867164ec84d3cd2628b91b8f086d0e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 11:00:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9590122c25c44b2dad52a7c122d93b2e : SUCCESS in 53m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0e9af7c855bf4712b7d5638fd10a9ba2 : SUCCESS in 33m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/10fa187552184232bd62f0b0e2fea2d2 : SUCCESS in 51m 35s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9622771f429e449ba27c7c01c93cc6a7 : SUCCESS in 53m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0c992b2ba3d64e51a2a5f086b18a26d6 : SUCCESS in 1h 15m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a23cd514c0bb4dd299920bdd92d8ec45 : SUCCESS in 33m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6855efbc139b4509a87c85db9378aae6 : SUCCESS in 18m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/01cf8146f0314324a6536e6c35a42bbf : SUCCESS in 55m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2ceb7ae939f4b1b8bc305c7523cfbf1 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91912fd453d04a3dbc90a7a7a9b180ed : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d79fa81b66d24963ac4b864916c43d0a : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bda10ce610434ffbbb079b28cb7d248e : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2080504b1d1241648d7aaec3e22a6730 : SUCCESS in 8m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2e587b8adc545fb81a53f745e8e88fc : SUCCESS in 37m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/695d6de3a66f453b9d6b90013ba428d2 : SUCCESS in 37m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/000f56201dc846f88fbf5ebbbbb182aa : SUCCESS in 55m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e5ab703c16aa417fa843b718aba5de99 : SUCCESS in 49m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/887c352cd46a4a1f97997905076582ba : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":2},{"id":"38e3ccbcdda0e55cabed00e85bc4ede21cc09351","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-03 18:00:49.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7b88eee443f7ca18c8082b9a6d6ef2ae2459e8a4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-03 18:00:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0bf280b53d464696e91766ad93095d972ae48754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 19:22:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4e1046978ecf4ee3bdc5288e736a6d3b : FAILURE in 23m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a492a3412ab44f5bb59fa8e9d9791d00 : FAILURE in 41m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3469b00dc7924efb8b93e84dec2da388 : SUCCESS in 1h 02m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8fa9ab620c9a4524abc1f69d691d7b8c : SUCCESS in 1h 13m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c1778c3e4e064b06a5459e17dda03046 : SUCCESS in 1h 16m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1c74969b8fa14aa2857f249e01bf021b : SUCCESS in 36m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c7e9f04fca5e4028a8b26b6d84397cd7 : FAILURE in 20m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e5788017b6ae4415b5b8b12b36309a0b : SUCCESS in 1h 02m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c003f53452410d8a483836dc4380ec : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21d7d36cc945433c971c62db38874956 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e65469823b0140d1916fc3f8e558cad3 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07a0e174ca804256a71b69f959d8b3be : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6380c68dea974baca92f671cc2268f88 : SUCCESS in 6m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/07129b09f2f4448d897353cd37142d0a : FAILURE in 22m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/753a11d54c4f4e6fb7c14998af51d354 : FAILURE in 39m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c1357137abd454bb9ddf398feaef6e3 : SUCCESS in 1h 00m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6f164ca443943c08feb97cc38001b8a : SUCCESS in 54m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d5ab952eda3413082bce05eea81c555 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":3},{"id":"c56fcf41874bf9fdcd22d595049a451a34a635f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 20:12:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b74d5a53fe324f6eb25c965e0da0cf85 : FAILURE in 1h 46m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/56675b4865c34b4cb00c6d367843c557 : FAILURE in 48m 15s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7999746d31c4360bf925b868488043e : SUCCESS in 2h 09m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9b67d5e51af61c49511ab3b47f5dad518e222ecc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-04 08:53:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"763c4db2d90faa6229a77618bac1a04f1efae509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 10:11:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e278a9b6f07542c2a39020aaed8e6094 : FAILURE in 25m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8df3fab0bfac4d0ca46cb32b8b1da897 : FAILURE in 49m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c9bf1ce8be0c4d95b5686ad620e7e07c : SUCCESS in 1h 12m 04s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f7bdadf6d04945939d2c22654bbbd2ee : SUCCESS in 1h 06m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/06f2f87ce9824c81857d638154d19953 : SUCCESS in 1h 16m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ddf5925ee4dd49db8e31daa9184d3966 : SUCCESS in 33m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/608f8c2307154309a7c2410838c85c3b : FAILURE in 17m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/202afc9ee73e463380146e07218ae662 : SUCCESS in 51m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc31afd6c404cac88e26f641e065826 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b71eb6990044cf099641d371cb6acfb : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a649ad8f5fc047d187e5622404d81fb4 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c948fb1012574b919c83295a52adf4b8 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01cba04cd9b43bfb647c4dc88e76e40 : SUCCESS in 7m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/50b933d3f6be4776bfd5f0b261fff5d1 : SUCCESS in 21m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/69833e5e36ed408b9d86a4b81763a31a : SUCCESS in 33m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6afd35e353dd4ebd89780d3cc09f6fb8 : SUCCESS in 58m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dc84ea0d016f499db52cb78f5b1b62ae : SUCCESS in 55m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ca3a1555af44125ab45bae1f9072056 : SUCCESS in 1h 14m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c80383ec4e2ffe50a8428a1051444bdef9a4fcb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 10:29:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ac9b3ad879e34cb7af295439acce5a73 : SUCCESS in 1h 24m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1a5b7e3a3964873b9c31bf7a282d09d : SUCCESS in 52m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a3079e883c24998bf119c269d04e1d7 : SUCCESS in 1h 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"641a9b2899dbdbac6d2a9ec71a3ffbb7afc904df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-08 08:57:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1ef463c4a76b797f04578bf446e896dccb608060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 10:02:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b846070e5e574756998b0600d9f5f5a4 : SUCCESS in 26m 58s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e4e1e8cf25024fcbbb827458c6b8f44b : SUCCESS in 33m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/930fbc5045da4482a72d276bc5837eba : SUCCESS in 55m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/53be88ef56ad42b49f5c9a6a39e11102 : SUCCESS in 54m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9f11654e1b1f4b56b738036f6e5916b0 : SUCCESS in 1h 04m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2b286153c30249268f9858e54b85aebe : SUCCESS in 41m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8f63d072702a4c308f00b6636a5076eb : SUCCESS in 20m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/78182fde56a74ea38429a353fdea88a7 : SUCCESS in 48m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af06338d0264446cae0871d110828955 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6919f200f7b344ba9936b14aca5fa045 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65202e95d73f471a91f863f7814532f4 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfdf40f25adc4a6184e66bb0fc04fcff : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb78f0482414d248efe388e989b8947 : SUCCESS in 6m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49d5d3530ca241b1a03aa14593d867f9 : SUCCESS in 23m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a07edc94eef34610bcb333f6b42494cf : SUCCESS in 37m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f05ec42848f04b37bc577a1892785522 : SUCCESS in 54m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/832b3942d2c146e4a016e9c35edc0747 : SUCCESS in 39m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/775a84b72a8e49c287f8de82cdf1ef23 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":4},{"id":"ca7653ee4ab85d5f9dd6604a93a7e3047fefb396","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 11:25:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bdcd609a70564d5a96e082d5b01c3f0a : SUCCESS in 2h 25m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ace325d169d49ceb6194f6ccc7e7514 : SUCCESS in 1h 14m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cce7e5d8557a4b038790bddd92fb1167 : SUCCESS in 2h 21m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb6f36e352bc42b7a5a3babad95f445e : SUCCESS in 2h 05m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"50b5c23b1d87141675304dcfc449e61992d50dc0","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-10 06:11:34.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"43b59fae344a2d856386109995ba7886203bfc3b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-05-10 10:58:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"278302a0e3b45a1f4b1dc32bfad8bebd886ecb67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 10:59:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b5a229875ac312148092338693850b1cad015da5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 12:12:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5052506028d54b0a86e02a1d899c4ce3 : SUCCESS in 42m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/258aff101a5e40cba3a24225fc037c00 : SUCCESS in 1h 08m 34s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f53c26cafe1347c4a0ea1198f24cd535 : SUCCESS in 1h 08m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3dc027ddb0a4751938ed291cf776943 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3d2314672e5480a8eca493ba2b8f47a : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d51ea597e25c404f876420b4d397fe2e : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa2e2df7be04e6a83ab700ecc46660e : SUCCESS in 5m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f9145406d6d74cd3a82a461caa92bdc9 : SUCCESS in 45m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a61f9fdc37694bef9ecc2cea34cc750b : SUCCESS in 47m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/49b33918546c40c596fe880b2b9fdb59 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":4},{"id":"c8e091581b7fb82e969ccfed929090278ce288fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 12:12:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d788f94f95092e55b2db57df600667ad1cd3ad86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 12:14:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d332cc76c4bf43fe961a15a83cc84c22 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"038074097727340e7a507a4693af0794797a849d","revisions":{"135a7079a412c40dc4846f908533cbcd01aefed0":{"kind":"REWORK","_number":1,"created":"2021-04-28 09:35:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/788462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/788462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/788462/1"}}},"commit":{"parents":[{"commit":"a4f6a8f05b45838338ee1c6a78f278d49486ac03","subject":"neutron: Remove vmware-nsx and networking-mlx from binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4f6a8f05b45838338ee1c6a78f278d49486ac03"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-28 09:34:57.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-28 09:34:57.000000000","tz":120},"subject":"centos: do not install CentOS Rabbitmq twice on AArch64","message":"centos: do not install CentOS Rabbitmq twice on AArch64\n\nChange-Id: I320d0795ce11ca00398cec6183ed140d31d3fd80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/135a7079a412c40dc4846f908533cbcd01aefed0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/135a7079a412c40dc4846f908533cbcd01aefed0"}]},"branch":"refs/heads/master"},"01727c605a0a95469f5fbe4c5394542ddaea3167":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-01 09:33:45.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/788462/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/788462/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/788462/2"}}},"commit":{"parents":[{"commit":"a4f6a8f05b45838338ee1c6a78f278d49486ac03","subject":"neutron: Remove vmware-nsx and networking-mlx from binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4f6a8f05b45838338ee1c6a78f278d49486ac03"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-28 09:34:57.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-01 09:33:40.000000000","tz":0},"subject":"centos: do not install CentOS Rabbitmq repo twice on AArch64","message":"centos: do not install CentOS Rabbitmq repo twice on AArch64\n\nChange-Id: I320d0795ce11ca00398cec6183ed140d31d3fd80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01727c605a0a95469f5fbe4c5394542ddaea3167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01727c605a0a95469f5fbe4c5394542ddaea3167"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b4e657bfa86d69a735585f379bf53eea21d40e74":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-05-03 18:00:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/62/788462/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/788462/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/788462/3"}}},"commit":{"parents":[{"commit":"bfea49246db981ea2f0a6fbc5343dd8f2a55d746","subject":"Revert \"CI: temporarily disable CentOS/AArch64 testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfea49246db981ea2f0a6fbc5343dd8f2a55d746"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-28 09:34:57.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-03 18:00:49.000000000","tz":0},"subject":"centos: do not install CentOS Rabbitmq repo twice on AArch64","message":"centos: do not install CentOS Rabbitmq repo twice on AArch64\n\nChange-Id: I320d0795ce11ca00398cec6183ed140d31d3fd80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4e657bfa86d69a735585f379bf53eea21d40e74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4e657bfa86d69a735585f379bf53eea21d40e74"}]},"branch":"refs/heads/master","description":"Rebase"},"038074097727340e7a507a4693af0794797a849d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-05-08 08:57:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/788462/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/788462/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/788462/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/788462/4"}}},"commit":{"parents":[{"commit":"6dfae74813289575e77aaa7dfe43f7e6a81a5f26","subject":"Merge \"[release] Use UCA Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6dfae74813289575e77aaa7dfe43f7e6a81a5f26"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-28 09:34:57.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-08 08:57:31.000000000","tz":0},"subject":"centos: do not install CentOS Rabbitmq repo twice on AArch64","message":"centos: do not install CentOS Rabbitmq repo twice on AArch64\n\nChange-Id: I320d0795ce11ca00398cec6183ed140d31d3fd80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/038074097727340e7a507a4693af0794797a849d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/038074097727340e7a507a4693af0794797a849d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
