)]}'
{"id":"openstack%2Fkolla~788490","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I635964a84f0cd5cadccd7ce8040fa36e1300c765","project":"openstack/kolla","branch":"stable/wallaby","hashtags":[],"change_id":"I635964a84f0cd5cadccd7ce8040fa36e1300c765","subject":"centos: switch to RDO final release of Wallaby","status":"MERGED","created":"2021-04-28 13:34:40.000000000","updated":"2021-04-29 07:49:42.000000000","submitted":"2021-04-29 07:48:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"788490","meta_rev_id":"efe5bc47cecfda69fdac5d0f8e7927f0c864be8f","_number":788490,"virtual_id_number":788490,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-29 07:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-29 06:31:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-28 16:04:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-04-29 06:31:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 13:34:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-28 15:05:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-28 15:11:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 15:12:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df844b1768df4f6725741d49123a44713f1e46de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-28 13:34:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f9956fe2cf51bd9bb17472bc0977d260de24059","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-28 14:31:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c72e7279e0db65407814861e8a6690e1b30bf2b7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 15:05:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a95731effbfe65d5601592e3cf339ba77edfbea2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 15:11:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5344b49769a14f429466b86b660d9489 : SUCCESS in 50m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/73d5b4d111f7442a9cfc830c7f8c8455 : SUCCESS in 1h 30m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f66200c942a1055febf213197ee229c0212e89d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 15:12:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/05ae57349e8d41a59590c92167a7c453 : SUCCESS in 27m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/024620e92cab44a2821f17647b1db885 : SUCCESS in 33m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5cf6d6d79f6a4384bba3f3d376f174c0 : SUCCESS in 1h 07m 58s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3d61fddb512842cbaa09a0ba9862bab8 : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d623d8de1da6465b8bb988f00365091e : SUCCESS in 1h 32m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ac93e0e8b9734989a36e74e4dc27d4bf : SUCCESS in 26m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b01f66504ce14c3eaf3a86f3034dd08b : SUCCESS in 22m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e82ee39f78314da2a62d4d6b2b362d37 : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ea3b125b964d5fa5231841963e7cf3 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fe115a2964c415b8bf11efe618e95f1 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d7bc0c196174ba29874918c258ce581 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7eeedbeba10546c7ae2291fd5e931cfc : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b9f3ca31ec24876b78dc4c1deba4fdf : SUCCESS in 6m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b078a92ca29e4de688e990607f422ce9 : SUCCESS in 24m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b15d52aa92b0497680f60b264cd09676 : SUCCESS in 38m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/85b1ba4f0ab3481aaa27ec310b9961ae : SUCCESS in 1h 00m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/411e227e0583495b9c0c992b269b3f9a : SUCCESS in 57m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/380915f6450844ce978b68661f8f7b47 : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a58cf705336d99d05eb7540bd62b68aaa56f5eda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-28 16:02:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e17fb6a2a1cf14af29ee2fb8a377d91845bea559","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 16:04:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2252bf2277736485a7fb541b07cce00d67cebf14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 17:20:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ca05333d42f74f1bac355d2b6fa4ac93 : SUCCESS in 26m 16s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/563f6894e5334665aeea1321e13a7402 : SUCCESS in 45m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e2a6af3eff9f476e87fb157ffdcb7891 : SUCCESS in 1h 01m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4cc586f7de6f436088251a26cbbe4f82 : SUCCESS in 1h 07m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/92443a2868dc4c87ac572f091a982201 : SUCCESS in 1h 13m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df8c473ebc244b219c7818a6b8841945 : SUCCESS in 35m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2b197175338346a890fea5518262667d : SUCCESS in 19m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/92ead230248e480abdd7a4f8b0a9e3c6 : SUCCESS in 50m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab8226061f7c4ab0a30eac008f8149e1 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8352e97daba4c8a8646c5e63dcf5f78 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67a95674488f4915aa7b0d1fc34321e8 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa0b296368f4471d962f1ea5f29736be : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2159399b7ce44223908d5b84f663b703 : SUCCESS in 7m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2cb8eb4f22e84be18a059389ea93e329 : SUCCESS in 22m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec00ddc45da941ad80ad22d50e9243c5 : SUCCESS in 41m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ea34af06ed4c4e929ab3776f808dc1a6 : SUCCESS in 50m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/39ae65279a104d2e86e537fd74b023f8 : SUCCESS in 1h 05m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/51b7828614d746baae11fc7aaab484da : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":2},{"id":"9e4b5bfd70e4514e78802b23bd78c753a1237446","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 17:45:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8f094fb40c8417ca251137482807ab2 : SUCCESS in 1h 02m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b24cab636504a6ea623328cb27037df : SUCCESS in 1h 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22d05fc0c5bda3c0a331669d473e427702557d6e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-29 06:31:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c7778067092127d0ee3c8e5f9be810275dd82f61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 06:31:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6fa8252f83e283b6a953d95dd0e435b7aa8db7ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 07:48:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b09f2ddf84d2486abb44aff73ee7e510 : SUCCESS in 42m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/339c9f4b91314dcabe14b686b257e607 : SUCCESS in 1h 06m 31s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9d574bb8fcf7425195823fd6b2667d76 : SUCCESS in 1h 04m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cfc5089f0b04229a376260a1ffb6142 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c51a8a5fedbc44fdb9148f927bd2fa69 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b098cb49f804029868d3e080de4c0e0 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2349591d774cce9cd5e6f365ca85ba : SUCCESS in 7m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b510283a42e4a79b7212acee0c69d3d : SUCCESS in 35m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/75afa216ac614fcd8d7359ccb64bfa45 : SUCCESS in 50m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0125c779dd10428f97b771d7e0f26c36 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":2},{"id":"4be7eda2fb2a594778b5a22380bcf04b964e6298","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 07:48:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"efe5bc47cecfda69fdac5d0f8e7927f0c864be8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 07:49:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe082fbecac4cd891073ea7b70d4af1 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53225b2bfd7507dca4c574cae404342eb07131dc","revisions":{"24e997acc7e9a7b96fead80e06e9095a0bc43a31":{"kind":"REWORK","_number":1,"created":"2021-04-28 13:34:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/788490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/788490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/788490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/788490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/788490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/788490/1"}}},"commit":{"parents":[{"commit":"940baeff65b8e75fed8a77e56cb385920d72ed01","subject":"Merge \"[TrivialFix] Remove unused aarch64 repo mentions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/940baeff65b8e75fed8a77e56cb385920d72ed01"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-04-12 12:32:47.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-28 13:34:22.000000000","tz":120},"subject":"centos: switch to RDO final release of Wallaby","message":"centos: switch to RDO final release of Wallaby\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24e997acc7e9a7b96fead80e06e9095a0bc43a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24e997acc7e9a7b96fead80e06e9095a0bc43a31"}]},"branch":"refs/heads/stable/wallaby"},"53225b2bfd7507dca4c574cae404342eb07131dc":{"kind":"REWORK","_number":2,"created":"2021-04-28 16:02:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/788490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/788490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/788490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/788490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/788490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/788490/2"}}},"commit":{"parents":[{"commit":"940baeff65b8e75fed8a77e56cb385920d72ed01","subject":"Merge \"[TrivialFix] Remove unused aarch64 repo mentions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/940baeff65b8e75fed8a77e56cb385920d72ed01"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-04-12 12:32:47.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-28 16:02:31.000000000","tz":0},"subject":"centos: switch to RDO final release of Wallaby","message":"centos: switch to RDO final release of Wallaby\n\nChange-Id: I635964a84f0cd5cadccd7ce8040fa36e1300c765\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53225b2bfd7507dca4c574cae404342eb07131dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53225b2bfd7507dca4c574cae404342eb07131dc"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
