)]}'
{"id":"openstack%2Fkolla~788525","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Ifa491b102ad0db7f6dc645ca552d83b53bcf3b46","project":"openstack/kolla","branch":"stable/wallaby","topic":"pin-erlang-stable/wallaby","hashtags":[],"change_id":"Ifa491b102ad0db7f6dc645ca552d83b53bcf3b46","subject":"Pin erlang","status":"MERGED","created":"2021-04-28 18:40:22.000000000","updated":"2021-04-29 09:59:11.000000000","submitted":"2021-04-29 09:56:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788525-pin-erlang-stable/wallaby","cherry_pick_of_change":788045,"cherry_pick_of_patch_set":1,"meta_rev_id":"47b2fc417a3f811ca835a68aeee04bf3f572a5c7","_number":788525,"virtual_id_number":788525,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-29 09:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"date":"2021-04-29 08:34:01.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-04-29 08:52:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-28 18:49:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"date":"2021-04-29 08:33:57.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-04-29 08:52:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 20:29:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 20:31:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 08:23:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-04-29 08:52:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8888c4c2c06e4ad9400a964decda2ab741807108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 18:40:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e1300490920edb4c590e4d5f1d0188ea2cc9db51","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 18:49:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46134ca64b22a29f7e8bb1b51505ee9bbf1b3b2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 20:29:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/489a12fc4d244f3491fc994120bfa087 : SUCCESS in 1h 44m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a9973b3c0c64db99f93ba2b43e64d93 : SUCCESS in 1h 20m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"509b2ffbf3a06fabe9a3aca9d282861424ce74ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 20:31:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/753ce39729e64e51a3ab33216e6429ec : SUCCESS in 50m 11s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b050d3269b3542d69c3d6c9da7f2847e : SUCCESS in 57m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0581729333214089a5849cf2c0c4d622 : SUCCESS in 1h 32m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f60a2320bf9d4566855d01fce204453b : SUCCESS in 1h 29m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4cf5ffa240ee4804a57f0480bc34a0e6 : SUCCESS in 1h 43m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4e19816e71c04ad6968af6d258d90839 : SUCCESS in 59m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/df05e56c8d7648d3a925cc82d8e1a0e7 : SUCCESS in 50m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/412ab8fe52ee4b48a0a64b82a95684e7 : SUCCESS in 1h 22m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f49540f0d84329863287086628e439 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41c6c1b3b5ad47a28a8f3c543a00a1cf : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/129f3064ae134466811bcfff89a3f9e5 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47fa2f691ad841019a8cd982925163ff : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309fa051b38542f3aef98fb690393003 : SUCCESS in 7m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1752d50d69f451082c507b6efe28348 : SUCCESS in 48m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/127829a481e54bb289b040b0a16cbb4b : SUCCESS in 54m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1041297a8304d0288558eefb3776b50 : SUCCESS in 1h 13m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9495f34326ec4633ae587d9728e83856 : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4265ffc5165e4b89827c947f1a340452 : SUCCESS in 1h 30m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e4e89574458ba6b7ccbdfd69c930ce173ff51644","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-29 08:23:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e68d5a0a3f7626751d49b86e02738b621647ca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 08:23:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58a0a28b1f1368e9f08bc89fb81943d65e9f1a48","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-29 08:33:57.000000000","message":"Removed Workflow+1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"002d47cb6f3b440a83e17a94124ea87b655b5089","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-29 08:34:01.000000000","message":"Removed Code-Review+2 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"3c70452a3b6603ab0c0a8b3e74d61557a14aeca1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-29 08:52:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b27c94d943791a5abc2288fb0c539f70220ffadb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 09:56:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f386afaf449d4c84ba4fb2342ca9e24c : SUCCESS in 40m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/de6aa13f18b34f5daecdc86628f720a9 : SUCCESS in 1h 08m 57s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/352219e3519d4deba3b5be59d32face6 : SUCCESS in 1h 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc63d3484c84beb90650961ebe3cb8a : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/385754f4a3714a3980ebf3162ea3b76e : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5eddbde027f34591ade887c0fa211249 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2140a391fc6745239370d38ba589d20c : SUCCESS in 6m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/89effe55e61e428b846c11f0d5b024bb : SUCCESS in 1h 31m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/74ea55854a8545cdb576a9c4689db77b : SUCCESS in 53m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ce683d934f741aba949fd72708824d9 : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c81b5ec6b8cbc0dabce7a69eac70d522e483302b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 09:56:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"47b2fc417a3f811ca835a68aeee04bf3f572a5c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 09:59:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b9c8b8f3d64b24afca9b11ff1a1f7f : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"084dd5e4ecb4f98a7758a63bfcb3a6f29a2825b7","revisions":{"084dd5e4ecb4f98a7758a63bfcb3a6f29a2825b7":{"kind":"REWORK","_number":1,"created":"2021-04-28 18:40:22.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/788525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/788525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/788525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/788525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/788525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/788525/1"}}},"commit":{"parents":[{"commit":"a1ad81338870ad4b5a07b6a0590b87472f2277b7","subject":"neutron: Remove vmware-nsx and networking-mlx from binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1ad81338870ad4b5a07b6a0590b87472f2277b7"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-26 16:54:11.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-28 18:40:22.000000000","tz":0},"subject":"Pin erlang","message":"Pin erlang\n\nNow that we run without erlang pinned by the repository itself,\nlet\u0027s pin it on our side.\n\nChange-Id: Ifa491b102ad0db7f6dc645ca552d83b53bcf3b46\n(cherry picked from commit b54789d093d8f9e35265d313e95f9f89aceb4f4b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/084dd5e4ecb4f98a7758a63bfcb3a6f29a2825b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/084dd5e4ecb4f98a7758a63bfcb3a6f29a2825b7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
