)]}'
{"id":"openstack%2Fkolla~788528","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ifa491b102ad0db7f6dc645ca552d83b53bcf3b46","project":"openstack/kolla","branch":"stable/train","topic":"pin-erlang-stable/wallaby-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"Ifa491b102ad0db7f6dc645ca552d83b53bcf3b46","subject":"Pin erlang","status":"ABANDONED","created":"2021-04-28 18:49:55.000000000","updated":"2021-05-06 09:34:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"cherry_pick_of_change":788527,"cherry_pick_of_patch_set":1,"meta_rev_id":"9e346ee8c5688a46f1038a5a68206484325b2332","_number":788528,"virtual_id_number":788528,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-28 19:16:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":-1,"date":"2021-04-28 18:50:12.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"reviewer_updates":[{"updated":"2021-04-28 19:16:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92822f8ae9665153677f8d10050ea694b802f57e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 18:49:55.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.\n\nThe following files contain Git conflicts:\n* docker/base/apt_preferences.ubuntu\n* docker/rabbitmq/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"3b7bbef3949932b96ded5b72a5d3f3fe377cdda6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 18:50:12.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"10fc6ae352dfa365f299de6dcc8e34b2ed7c7cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:16:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d80206de57444c2dacebaa73c48fc6ce : FAILURE in 5m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7da7118c6a0342bebbe4952c2dbc0f2e : FAILURE in 5m 22s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4223d735baac433cbfd80574cee7f09d : FAILURE in 5m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d7a82b47206244a4b5786e48bdea017d : FAILURE in 13m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c581cfe17f2a4fa18bc678aba5c3078e : FAILURE in 9m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2104b659899f426e9192830769687e94 : FAILURE in 10m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0a4a40e3f8e74f33a97d60c5bca3279d : FAILURE in 9m 03s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1fd89652507748e2b1eb2973d229a688 : FAILURE in 9m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a6ce5bcaadb9486ab696685aa534f5af : FAILURE in 15m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/16c540fb98854f559b64b58690aae9c3 : SUCCESS in 16m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c3adf33ec6254e26932b635d95497f30 : FAILURE in 7m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c9d903c363af4c99ab67949e21e35bcb : FAILURE in 17m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3cef5fd6929a4649a81474fd25dd22af : FAILURE in 7m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03eca813b1ef4ff89770b3fb5e597e52 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba429cc77c1947a3807b8d5583edb2f0 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0236da090a0b465a941645b993512a4d : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0716eb522c54a3699861c5cae65b597 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed4fe1f726e4e08ab25591063dcf749 : SUCCESS in 9m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf7ea4365e0c4e3395a55d7a9c3f4659 : FAILURE in 16m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/09528b41e2e545d7b9dfd5e9a83054bf : FAILURE in 4m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb8b87e93dd641a8ba085c7a641d6039 : FAILURE in 7m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/19089de9972648ca8c0af9dc4d225e6f : FAILURE in 14m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/abd7efbb12db4727b4178bd7a1002cd1 : FAILURE in 16m 13s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"9e346ee8c5688a46f1038a5a68206484325b2332","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-06 09:34:02.000000000","message":"Abandoned\n\nthis is not applicable","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"131d31fbf3e2f1903e04085a03bae288ad6fc135","revisions":{"131d31fbf3e2f1903e04085a03bae288ad6fc135":{"kind":"REWORK","_number":1,"created":"2021-04-28 18:49:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/28/788528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/788528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/788528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/788528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/788528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/788528/1"}}},"commit":{"parents":[{"commit":"56a570ddcdc287644e31c950f7591d355f61d6f4","subject":"base: configure curl before first use","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56a570ddcdc287644e31c950f7591d355f61d6f4"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-26 16:54:11.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-28 18:49:55.000000000","tz":0},"subject":"Pin erlang","message":"Pin erlang\n\nNow that we run without erlang pinned by the repository itself,\nlet\u0027s pin it on our side.\n\nChange-Id: Ifa491b102ad0db7f6dc645ca552d83b53bcf3b46\n(cherry picked from commit b54789d093d8f9e35265d313e95f9f89aceb4f4b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/131d31fbf3e2f1903e04085a03bae288ad6fc135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/131d31fbf3e2f1903e04085a03bae288ad6fc135"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
