)]}'
{"id":"openstack%2Fkolla~788550","triplet_id":"openstack%2Fkolla~master~Id32dd40854f728ea5e058c391d70f9f3a2a0bd37","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Id32dd40854f728ea5e058c391d70f9f3a2a0bd37","subject":"CI: Remove kolla_python_version from Debian/Ubuntu jobs","status":"MERGED","created":"2021-04-28 16:24:05.000000000","updated":"2021-04-28 19:19:20.000000000","submitted":"2021-04-28 19:16:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788550","meta_rev_id":"f207b29ebd4f94f94242289bbbf618f32405e435","_number":788550,"virtual_id_number":788550,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 19:16:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2021-04-28 16:56:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-04-28 16:56:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 16:25:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-28 16:26:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 16:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c40d882fdaa0caf057f87695a59abbf53aa756dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-28 16:24:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5fcbb08463f4a65ad7856d69964cda5e1a61b7b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 16:25:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b43aa4da0f8d8be066f72070fdcc5d1c0b8498bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 16:26:25.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][2][\u0027kolla-ansible-debian-source\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-debian-source\n          - kolla-build-debian-binary\n          - kolla-ansible-debian-source:\n      check-arm64:\n        jobs:\n          - kolla-build-debian-source-aarch64\n          - kolla-ansible-debian-source-aarch64:\n      periodic:\n        jobs:\n          - kolla-publish-debian-source-quay\n          - kolla-publish-debian-source-aarch64-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-debian-source-dockerhub\n          - kolla-publish-debian-source-aarch64-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-debian-source\n\n  in \"openstack/kolla/.zuul.d/debian.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"564533eab1e5a52e7c98c23b1f96727b547e18df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 16:26:54.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][2][\u0027kolla-ansible-debian-source\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-debian-source\n          - kolla-build-debian-binary\n          - kolla-ansible-debian-source:\n      check-arm64:\n        jobs:\n          - kolla-build-debian-source-aarch64\n          - kolla-ansible-debian-source-aarch64:\n      periodic:\n        jobs:\n          - kolla-publish-debian-source-quay\n          - kolla-publish-debian-source-aarch64-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-debian-source-dockerhub\n          - kolla-publish-debian-source-aarch64-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-debian-source\n\n  in \"openstack/kolla/.zuul.d/debian.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"a6b45a7fbe42b5bac62ace69f9b65fa2c3ab865d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 16:55:39.000000000","message":"Patch Set 1:\n\noh, you left the colons hanging","accounts_in_message":[],"_revision_number":1},{"id":"add52fcec80d00740147cdbac8ed55f34eb407a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 16:56:04.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][2][\u0027kolla-ansible-debian-source\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-debian-source\n          - kolla-build-debian-binary\n          - kolla-ansible-debian-source:\n      check-arm64:\n        jobs:\n          - kolla-build-debian-source-aarch64\n          - kolla-ansible-debian-source-aarch64:\n      periodic:\n        jobs:\n          - kolla-publish-debian-source-quay\n          - kolla-publish-debian-source-aarch64-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-debian-source-dockerhub\n          - kolla-publish-debian-source-aarch64-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-debian-source\n\n  in \"openstack/kolla/.zuul.d/debian.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"1521a29b35dad4e97d3875ecdd31c0fecf3c487c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 16:56:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"aa3fd207614e3f6008df61bac6de1a4904b0ff42","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-28 16:56:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"523ef5b14715b264842cdde3615fe082c8d9cbfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:08:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e6b2c8053abf47ec88c8f0f27f0460b9 : SUCCESS in 46m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8932ea2ac8d04a04a3b5d2d6a71af32b : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0a7a0ea5e104e4cbd7c23d12c0f5cb9 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/834b301b242648908e91df70b751496c : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f310c87a667246fdbad87d06c21327ec : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aca070763b984cb698361ace8ab3ff07 : SUCCESS in 6m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bfeffd9373a54ac3b2b6d1c506711f0c : SUCCESS in 58m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/da4274688c044e3a8f66f63bc7fa1875 : SUCCESS in 49m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a5bb07d668904cae966e5266087a2592 : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":2},{"id":"5427e9c57a2118ed51184788ea7788f326da7668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:09:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"584467a1e5d9ab154a6a2c785ba1b600731a2095","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:23:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2700cc401b2844fc855a7fe2759b2e1a : SUCCESS in 1h 22m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4f37d8f70ef3005b93fb82645b1224594dd7c80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:16:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee26bf6422174e48a0801d2a8aa9903f : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be11fe1fadd24c10a3fad79db2ce4be6 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a011a95931648298cf6f5ace26fd6b9 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771c65eaa5d342ca9b888fc41a7e17e1 : SUCCESS in 6m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f7b7cbecadf4826992508bdc3ff09f7 : SUCCESS in 55m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8111bb3aafa6485b9b5e2f400bd38efb : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":2},{"id":"bdb1803fa4f7a1cc2e207e0010a32ebc8a1d68b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:16:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f207b29ebd4f94f94242289bbbf618f32405e435","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:19:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f586ece7469a4c43adeb53aaec72c75c : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"29b16b8a8dad14259465c5fe9944a148e702f770","revisions":{"f640d41cf378b9508d5bb38943e1e9d0d512a5b3":{"kind":"REWORK","_number":1,"created":"2021-04-28 16:24:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/50/788550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/788550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/788550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/788550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/788550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/788550/1"}}},"commit":{"parents":[{"commit":"a4f6a8f05b45838338ee1c6a78f278d49486ac03","subject":"neutron: Remove vmware-nsx and networking-mlx from binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4f6a8f05b45838338ee1c6a78f278d49486ac03"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-28 16:23:20.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-28 16:23:59.000000000","tz":120},"subject":"CI: Remove kolla_python_version from Debian/Ubuntu jobs","message":"CI: Remove kolla_python_version from Debian/Ubuntu jobs\n\nThese are leftovers from Focal and Bullseye upgrades.\n\nChange-Id: Id32dd40854f728ea5e058c391d70f9f3a2a0bd37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f640d41cf378b9508d5bb38943e1e9d0d512a5b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f640d41cf378b9508d5bb38943e1e9d0d512a5b3"}]},"branch":"refs/heads/master"},"29b16b8a8dad14259465c5fe9944a148e702f770":{"kind":"REWORK","_number":2,"created":"2021-04-28 16:56:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/50/788550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/788550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/788550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/788550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/788550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/788550/2"}}},"commit":{"parents":[{"commit":"a4f6a8f05b45838338ee1c6a78f278d49486ac03","subject":"neutron: Remove vmware-nsx and networking-mlx from binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4f6a8f05b45838338ee1c6a78f278d49486ac03"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-04-28 16:23:20.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-28 16:56:06.000000000","tz":0},"subject":"CI: Remove kolla_python_version from Debian/Ubuntu jobs","message":"CI: Remove kolla_python_version from Debian/Ubuntu jobs\n\nThese are leftovers from Focal and Bullseye upgrades.\n\nChange-Id: Id32dd40854f728ea5e058c391d70f9f3a2a0bd37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29b16b8a8dad14259465c5fe9944a148e702f770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29b16b8a8dad14259465c5fe9944a148e702f770"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
