)]}'
{"id":"openstack%2Fkolla~788614","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I1d68a47f78b925af76c44c61af7da1465371e721","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I1d68a47f78b925af76c44c61af7da1465371e721","subject":"CentOS on AArch64 has only Erlang 22","status":"MERGED","created":"2021-04-29 12:00:10.000000000","updated":"2021-04-30 09:23:45.000000000","submitted":"2021-04-30 09:22:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788614","cherry_pick_of_change":788615,"cherry_pick_of_patch_set":2,"meta_rev_id":"cb9a060db15db1025639d03423d21b3387bcffad","_number":788614,"virtual_id_number":788614,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-30 09:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-04-30 08:06:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-29 13:53:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-04-30 08:06:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 13:10:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 13:53:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-30 08:06:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"f2280930f2156fa750bec52c9d7142ae1275de3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-29 12:00:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"3db8eaba5f2da44065de91ab57e8e65aa760f973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:10:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/829a93b644a543afaaf19ca259632347 : SUCCESS in 28m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b9f6306394524b74ba75920976b4aa82 : SUCCESS in 29m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ba5fbd702bb74b7ab8d9426608856674 : SUCCESS in 1h 04m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/37418f0bc9184bffa239a4fc22ebef0a : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/be242d7bf25048ea8e45bd30f48ab867 : SUCCESS in 1h 07m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2fc894d849bc474a9eb07808af29d8bd : SUCCESS in 26m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fbb8ce6935064109bab3c95d4ca31720 : SUCCESS in 23m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1f1ccfe997a2411398eb2a16744d6a6f : SUCCESS in 58m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37dec3dc7809485f88b2750dbb46cd51 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/184590411216404a8a97d294845b40da : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/610c2a320c2e4513877664daa7185bc9 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23cfa7cec72744dc8825da7db00ece11 : SUCCESS in 6m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d0eb44533344a21a39d3d6887fbfeed : SUCCESS in 25m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a0bb1cd44db489ba242e9566145f375 : SUCCESS in 25m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/41572bdd51a5474fb0de44851cd90345 : SUCCESS in 53m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6288bbad7d5940d6ae8c0d761bb0ec81 : SUCCESS in 51m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c9612f41cbc141dbb7a91090e12b3149 : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ca106d3307f6873f5cea4f456e0070230e27f3a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:42:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/70dae2c8807f4bbb92ecd50a3ce42f19 : SUCCESS in 1h 01m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f077cc4b96274ac89ca5cf39afc8bf86 : SUCCESS in 1h 29m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ff3b23c6a334812895f1e9b7e8fa789 : SUCCESS in 1h 38m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e435eedac6d4cb3a3974345e59d5e5c : SUCCESS in 1h 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c333db77b3162e4dbdc2eb4de1582dced6604684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-29 13:50:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7e94bafae0c3c4c3e41fd8ded9ef84e12cb079f7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-29 13:53:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b45cac1e10949d6cf8192ef3e497d618472bf217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 15:11:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/882e00b685284ebbaabcd64b8a2d301b : SUCCESS in 25m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ea74e3b5cf3f4e028fc76133c194b513 : SUCCESS in 29m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/99b629f798e34b518466d2adf789c882 : SUCCESS in 1h 02m 33s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/eab6a90a13a3430fb6776c2e6f3800c2 : SUCCESS in 1h 03m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6cedbcd102094468a121de6d83c6ddb2 : SUCCESS in 1h 05m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6e7f82e596a9419585d346b5f3341f15 : SUCCESS in 25m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7cf1242a20244f088e840c33de95a950 : SUCCESS in 20m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fbbb8cf72345483d8e48cf54a3865938 : SUCCESS in 56m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e6d392648fd41af91f1c819cc20d774 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e15d5c9da654fd690e1d8afdce22378 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/350ff46c461e44fba8a323ffb35a0914 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c87b0bd70a4637b0fd8903b6f89ca1 : SUCCESS in 6m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c5b7fbaed9c949d9a15a89201ac4e466 : SUCCESS in 21m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b7a7f5111f6435ba387bc946807f3a9 : SUCCESS in 31m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e19981ac8664cc085181b0f01c89566 : SUCCESS in 1h 01m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f3b1a76e346e40c4b52ad754e842083b : SUCCESS in 52m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf03dc7c1d8c4b7eb29ee988a893ae92 : SUCCESS in 1h 13m 48s","accounts_in_message":[],"_revision_number":2},{"id":"5585a5d68eb818e901acb8ee452a3e7324d49736","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 16:44:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b96f02ac0e7e4aafb43cd6a0df56ec8c : SUCCESS in 2h 13m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d1c5439b402499e8f43f1c2ecbd8a10 : SUCCESS in 1h 45m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/12ad48ee9aa24fcead385e9335e5a0e9 : SUCCESS in 2h 50m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f75b0b615eb4a8ba760f07198191741 : SUCCESS in 1h 39m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a250d2e795630db266d5c1877f0d495217a7d7f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-30 08:06:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"68e33d7c12f0b60c12bfc0b1a41350a5a3c79782","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 08:06:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9a73f7988ad7ed97d110cac07a9a1f498952389e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 09:22:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b00ae55d50fd4c16b94e7057f4ef3543 : SUCCESS in 23m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8f045daea4de458aa012faf84b8d849b : SUCCESS in 1h 01m 44s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a6aa9b7a2979489aa07639d98be40a39 : SUCCESS in 55m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecdbde5e150042fb96c1023a9f0733ed : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba9c29718eaa4d0fa56e5aa157fb7c85 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cbd62cd641643698f0a79cf676610c5 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6b1ca6d66f48beb1ecb64e0e7f28f9 : SUCCESS in 6m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4fa827e5b4f4af68ee57204b056d60a : SUCCESS in 30m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ae9a97ee01144afbd612ecf24c52e5a : SUCCESS in 57m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/34842396bff34160800e75e929e1794c : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f9e7bbdf70d963a39c3bfcb6ceedaf52b16e27fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 09:22:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cb9a060db15db1025639d03423d21b3387bcffad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 09:23:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c8185b0eabe49528b94ce54453b4d28 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf2e0d62e60cc363fb699e6bd3e3f40b90af7d23","revisions":{"bfe8f408a1e0782db0040b39beda978d81cc463a":{"kind":"REWORK","_number":1,"created":"2021-04-29 12:00:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/14/788614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/788614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/788614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/788614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/788614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/788614/1"}}},"commit":{"parents":[{"commit":"c366a5ca1799d5c85df60fdda21167806bcca6e6","subject":"rabbitmq: Move to packagecloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c366a5ca1799d5c85df60fdda21167806bcca6e6"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-29 09:12:21.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-29 12:00:10.000000000","tz":0},"subject":"CentOS on AArch64 has only Erlang 22","message":"CentOS on AArch64 has only Erlang 22\n\nRabbitMQ 3.8.14 is the last version with Erlang 22 support\n\nChange-Id: I1d68a47f78b925af76c44c61af7da1465371e721\n(cherry picked from commit 0ec6049a73ef7e322c06ee0e7b8582e5e69b8815)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfe8f408a1e0782db0040b39beda978d81cc463a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfe8f408a1e0782db0040b39beda978d81cc463a"}]},"branch":"refs/heads/stable/ussuri"},"bf2e0d62e60cc363fb699e6bd3e3f40b90af7d23":{"kind":"REWORK","_number":2,"created":"2021-04-29 13:50:35.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/14/788614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/788614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/788614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/788614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/788614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/788614/2"}}},"commit":{"parents":[{"commit":"f4a5d52df21bf7aa30c80d4dc65f740a2780b6c5","subject":"Pin erlang","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4a5d52df21bf7aa30c80d4dc65f740a2780b6c5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-29 09:12:21.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-29 13:50:35.000000000","tz":0},"subject":"CentOS on AArch64 has only Erlang 22","message":"CentOS on AArch64 has only Erlang 22\n\nRabbitMQ 3.8.14 is the last version with Erlang 22 support\n\nChange-Id: I1d68a47f78b925af76c44c61af7da1465371e721\n(cherry picked from commit 0ec6049a73ef7e322c06ee0e7b8582e5e69b8815)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf2e0d62e60cc363fb699e6bd3e3f40b90af7d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf2e0d62e60cc363fb699e6bd3e3f40b90af7d23"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
