)]}'
{"id":"openstack%2Fkolla~788615","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I1d68a47f78b925af76c44c61af7da1465371e721","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"I1d68a47f78b925af76c44c61af7da1465371e721","subject":"CentOS on AArch64 has only Erlang 22","status":"MERGED","created":"2021-04-29 12:01:38.000000000","updated":"2021-04-30 09:29:28.000000000","submitted":"2021-04-30 09:27:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788615","cherry_pick_of_change":788613,"cherry_pick_of_patch_set":2,"meta_rev_id":"96627e77ad4c31cdced65c14239c7ec40110adb0","_number":788615,"virtual_id_number":788615,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-30 09:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-04-30 08:06:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-04-29 13:50:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-04-30 08:06:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 13:21:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 13:50:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-04-30 08:06:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8a4a41656cc5859656d653b4bb62f66f82d148a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-04-29 12:01:38.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"6008857d18345b041c8315005a15f233b7658a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 13:21:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6da73adb703b47f5bf2e71e4244ed36f : SUCCESS in 25m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a5876fdfe1074458a2176d340b933d8a : SUCCESS in 34m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7c6a3c5de08d4bfca5c29b1af899a30f : SUCCESS in 1h 08m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1b5c426afab54bedb70887c080e1311f : SUCCESS in 1h 00m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/372750dc893d4355a494757a3938350d : SUCCESS in 1h 15m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4a4c630a06f54a5c97a8b0c143b0b875 : SUCCESS in 19m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/707c9e11cb3c47e987b2c2ded9357fd8 : SUCCESS in 20m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ba0e9dd709ef4537896075e288f19b33 : SUCCESS in 52m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fbc24b831f04ecd9d061394c5ec3af8 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2122e9b57ccb4ba4b2a317f88bc8c7e1 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68fb0f854eca428e9c041065043c2253 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dbb53aca9934de680b8c4a7cf025a1a : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e92269c8597c4fb1b23bfec0b2b7636e : SUCCESS in 19m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/28ffc2f720224db28cb811812d94640d : SUCCESS in 22m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d024e65c714341d7b3f8608a40bb8fd6 : SUCCESS in 46m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3511482cbc13406faac0a569ff7c71d2 : SUCCESS in 46m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/93c18be943be438fb9dc2eb467c9f5ff : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"eab14bc52509444c7721a3ca42c40f2f8e4ba52e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-29 13:49:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"152a6ea9627780976469069ac30954fc4d117d58","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-29 13:50:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e5be460d6b67b022ecf815ab435124e87e4b8a10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 15:20:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8c507820183040c5a01f2042798ceb4b : SUCCESS in 20m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/db0ffaecdf524242a6aff29b79e34fad : SUCCESS in 38m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c9a40fb5d129450c805112db7772beab : SUCCESS in 52m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b4425fbe2a16497caf0233455ae24e8b : SUCCESS in 57m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d8395555f0004e44a2a39bc21192614e : SUCCESS in 1h 14m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/81175fe4e58a4361be2f218e54931bf1 : SUCCESS in 22m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4eababaed1e54853833c92516ed61d0a : SUCCESS in 21m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/507ab58d6ce442d9b78ac7e28d86a8a2 : SUCCESS in 51m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2e66f65aafb4867bccf9c7ee806505c : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83022513a6b848d09e94919e79ea0a84 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/939744ad96414931bd0b649d66caea8c : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ecc534fcfd84630a5e836422ca2b02f : SUCCESS in 7m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e7f641973c74a6eb8583919e843bc18 : SUCCESS in 20m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e46537c4c71a4f17856160c90687e34f : SUCCESS in 21m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36a4eff2a4d040de811d6332a550274f : SUCCESS in 54m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/13525ab4d3d94b7c9589cc756b5f69a9 : SUCCESS in 53m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e8d3bc924b104b63a2588d2e19d2b4e1 : SUCCESS in 1h 17m 36s","accounts_in_message":[],"_revision_number":2},{"id":"01ef8b3b16562cdead79253e183e7f9d2cdf929d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 16:20:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ce125579fd440549ec85fcf4e9d77fc : SUCCESS in 2h 17m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1442c5c94c964712bdec10a4eb2570b7 : SUCCESS in 1h 41m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0bbca8a741a94f9e917e147b1b29d794 : SUCCESS in 2h 26m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/291533794dae4f16b6800d0577af0365 : SUCCESS in 2h 02m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf0d08fc3f9160afdb03cdff963f8a70ac72d334","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-30 08:06:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"04b3af8bcf7c90b29330373e2bbe5714450ed2e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 08:07:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f6dcc4ea3fd2533f5cbb56c5f2f91f2a1d30dc01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 09:27:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9a23fde4b6cd47938fdfd1a925e1fd13 : SUCCESS in 27m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1c34a351ebe24548bcc30976b9350940 : SUCCESS in 1h 00m 21s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4c37e8f604b34c3dae2f2e7c16b4c535 : SUCCESS in 1h 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019936c6f63649a8a6ce0bca86409d2b : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfafaf4dde7b46ddb2e48a6a04173e02 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c0dd51826de436d96ccc187a51a581c : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42ce1e2181b46928a64c4428a9782c0 : SUCCESS in 6m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/330045053e18499fa5c8e14d76679f24 : SUCCESS in 30m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0efed7f3a10a4d379b2138f7da68de8a : SUCCESS in 50m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3415dee8e96b4b1a8e0233bdf3934588 : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9cb77be6a1ff7bf2fdaac039063579cd996f6a9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 09:27:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"96627e77ad4c31cdced65c14239c7ec40110adb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 09:29:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3e89c774b2460088a862170602c28d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0c20a44f9ad764492d98a6c6e21218dbee67e2a7","revisions":{"9dd473aa4a9ddf1ac1d5e51149553c8530a99a8d":{"kind":"REWORK","_number":1,"created":"2021-04-29 12:01:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/15/788615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/788615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/788615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/788615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/788615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/788615/1"}}},"commit":{"parents":[{"commit":"b886f74fe8516e480696d338988201198a3979fd","subject":"Merge \"Disable trove for Debian/binary\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b886f74fe8516e480696d338988201198a3979fd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-29 09:12:21.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-29 12:01:38.000000000","tz":0},"subject":"CentOS on AArch64 has only Erlang 22","message":"CentOS on AArch64 has only Erlang 22\n\nRabbitMQ 3.8.14 is the last version with Erlang 22 support\n\nChange-Id: I1d68a47f78b925af76c44c61af7da1465371e721\n(cherry picked from commit 0ec6049a73ef7e322c06ee0e7b8582e5e69b8815)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9dd473aa4a9ddf1ac1d5e51149553c8530a99a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9dd473aa4a9ddf1ac1d5e51149553c8530a99a8d"}]},"branch":"refs/heads/stable/victoria"},"0c20a44f9ad764492d98a6c6e21218dbee67e2a7":{"kind":"REWORK","_number":2,"created":"2021-04-29 13:49:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/788615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/788615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/788615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/788615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/788615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/788615/2"}}},"commit":{"parents":[{"commit":"3ee3c7d7c9702bb2f76d0e01f8bb6d4638e5b78e","subject":"Pin erlang","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ee3c7d7c9702bb2f76d0e01f8bb6d4638e5b78e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-04-29 09:12:21.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-29 13:49:14.000000000","tz":0},"subject":"CentOS on AArch64 has only Erlang 22","message":"CentOS on AArch64 has only Erlang 22\n\nRabbitMQ 3.8.14 is the last version with Erlang 22 support\n\nChange-Id: I1d68a47f78b925af76c44c61af7da1465371e721\n(cherry picked from commit 0ec6049a73ef7e322c06ee0e7b8582e5e69b8815)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c20a44f9ad764492d98a6c6e21218dbee67e2a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c20a44f9ad764492d98a6c6e21218dbee67e2a7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
