)]}'
{"id":"openstack%2Fkolla~789269","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I11256c674c42e35496146ff64fee1e8e04c30d2c","project":"openstack/kolla","branch":"stable/victoria","topic":"debuntu-rmq-fix-key-stable/wallaby-stable/victoria","hashtags":[],"change_id":"I11256c674c42e35496146ff64fee1e8e04c30d2c","subject":"Install RabbitMQ repo key on Ubuntu and Debian","status":"MERGED","created":"2021-05-03 14:14:33.000000000","updated":"2021-05-03 18:48:41.000000000","submitted":"2021-05-03 18:47:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789269-debuntu-rmq-fix-key-stable/wallaby-stable/victoria","cherry_pick_of_change":789268,"cherry_pick_of_patch_set":1,"meta_rev_id":"e3e1f11ce7c55afcedcecca1f5aeec76e743f940","_number":789269,"virtual_id_number":789269,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-03 18:47:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-05-03 16:57:29.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-05-03 14:14:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-05-03 16:57:29.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-05-03 14:14:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-03 15:59:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-03 16:57:29.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"fc85c2b87ea69bdeb973756f08d41a83f75cd2fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-03 14:14:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"7952c34fd3d83c304f3ee3640e2d9520e1f8cd1a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-03 14:14:54.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"2366ad21c69a66308c68f2bc24f16f1b996665ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 15:59:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f1c5ad13f03546dc87641047f3a7e2f7 : SUCCESS in 29m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ce8bcd64caa642f6a80cf0e85dbeda44 : SUCCESS in 24m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8a572535a7624ca5b7279c2477f7ea8e : SUCCESS in 1h 08m 24s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/13bf198be60b4691b141f6143f737778 : SUCCESS in 57m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/63926b7768f74a5eb5d603056a497243 : POST_FAILURE in 1h 42m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/32b910871e414d149c0846a22050e26f : SUCCESS in 24m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/187c37ca47604151853ba026c6ae1d76 : SUCCESS in 23m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7c740bb99f1f4993bb87981b4b9b7eaa : SUCCESS in 56m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/756b9cd704e2461682bc281146b5c1f9 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49468f439a834f79bf2f3cf19db19d28 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/733682db6f4a466a88809b6ed8b23a61 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e87c36dd49f4a04b08a322834b03eb2 : SUCCESS in 6m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bd5481b1495c4b01bb2ee9748c67efed : SUCCESS in 22m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0343d4d273344d02b179176702377d0b : SUCCESS in 22m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/161331572b3c49dd893db1058abc5f2c : SUCCESS in 53m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be2298dcaaff4d819f377206cb6a4137 : SUCCESS in 56m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1175087ea86c4b87811ed025bd127eaa : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2538f2a9f4e9f46f4c36bb3c7f3f4d7025aa1b21","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-03 16:01:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd3fff276acebbb6646d92a2d238eb8964c4fbec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 16:36:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/33bd24846f664993b3592ca669030e1c : SUCCESS in 2h 10m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/67083cb6fc3049ffb8a1b75be0d5cfb8 : SUCCESS in 1h 49m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c2a917c62f843f9a26285fcac7971b1 : SUCCESS in 2h 19m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4dffb7559c9346df8a4d5ae9b42ca51e : SUCCESS in 1h 59m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99be5dee542569f0177c30f47808a27b87cf4644","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-03 16:57:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cac849299c962dd2dde13a8edb09ae5ce46dbc7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 17:12:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/98376b52a0754bfcbaf578f45b73f720 : SUCCESS in 25m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e0497fba63df4b4f894324e928711b54 : SUCCESS in 27m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/165eaf5430a64903993daf435dc4eff7 : SUCCESS in 1h 07m 41s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6cfcfeedb31f429ca3689738f5326a39 : SUCCESS in 1h 01m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5c0b11ec915a4496b70d9e1328e5909c : SUCCESS in 1h 06m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b3f9e95a97484b55b7ae9802aa552ab5 : SUCCESS in 21m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6b5e07949ba8481a90f609de1e434b81 : SUCCESS in 17m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e162ec1cf08d422e85cfacaaa88e946a : SUCCESS in 53m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a95ce90ecaa247bf8b6dcb0bc576fe38 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e21feb2c5694bf7bf5ecc23b2a4c11a : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb15af0b2d064359942c0ae349a94256 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415efd28ca284117aedfdef6c07d3cdf : SUCCESS in 6m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/14845e6558be450687631529608e6b8b : SUCCESS in 21m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3895410fe39249b59bd6a8a80d61f315 : SUCCESS in 23m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf031d0b9fe84ea091337872eca4cbbd : SUCCESS in 53m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/39107b5f9a9245a4a859ec9d6b6a7ded : SUCCESS in 52m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/afae8165f75e4b329b39ceaf7c50efd3 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f357a4fb70bd3c1bbdc7e64fb2f7256d63128377","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 17:12:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbf5714a14622c7d8f9bcd45f9ed8efd10674594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 18:47:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/510c6c7974ef4fe69b45c9f2309ff3a9 : SUCCESS in 35m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9ae7fa32f92141f395b2cb654b33c433 : SUCCESS in 45m 01s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f35343036e1d4cc58d00bcae8b165468 : SUCCESS in 58m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a5d6fd23af340e1b9ba35db10a7049d : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff8c95dad9f942af9204ec7227d456c5 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7a30ef7e6d14fc58f61e6707e5d1993 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfa131921ec4170a0a9687bf203d8db : SUCCESS in 5m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/049ff0a839fd4169bba9380dfaa7d5ca : SUCCESS in 1h 09m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7589586124a4560b1a8da51ba88c559 : SUCCESS in 52m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/13524aac2ce046b291e87644fb173bf8 : SUCCESS in 1h 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"52508b4b16e7212df150747582ac9b78ce835e16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 18:47:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e3e1f11ce7c55afcedcecca1f5aeec76e743f940","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 18:48:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fca4357ff9d48c981b43bcb5950f2cc : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29bb43d53b611aa66d470c27550fae5cebced6ce","revisions":{"29bb43d53b611aa66d470c27550fae5cebced6ce":{"kind":"REWORK","_number":1,"created":"2021-05-03 14:14:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/69/789269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/789269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/789269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/789269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/789269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/789269/1"}}},"commit":{"parents":[{"commit":"0c20a44f9ad764492d98a6c6e21218dbee67e2a7","subject":"CentOS on AArch64 has only Erlang 22","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c20a44f9ad764492d98a6c6e21218dbee67e2a7"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-30 18:41:11.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-03 14:14:33.000000000","tz":0},"subject":"Install RabbitMQ repo key on Ubuntu and Debian","message":"Install RabbitMQ repo key on Ubuntu and Debian\n\nIt was missed previously.\nCI did not detect it because it ignores apt security issues.\nThat gets fixed in Ic5abc4b87fd76f87aba383abf43e95ba70629fcb\n\nChange-Id: I11256c674c42e35496146ff64fee1e8e04c30d2c\n(cherry picked from commit 444313975e6099055c5573363eea8af5a035fdbd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29bb43d53b611aa66d470c27550fae5cebced6ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29bb43d53b611aa66d470c27550fae5cebced6ce"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
