)]}'
{"id":"openstack%2Fkolla~789550","triplet_id":"openstack%2Fkolla~master~I639765c400b74518bea6c6c54cba97491dcd50b2","project":"openstack/kolla","branch":"master","topic":"ci-trust-only-infra-mirrors","hashtags":[],"change_id":"I639765c400b74518bea6c6c54cba97491dcd50b2","subject":"base: use proper keys for influxdb repositories","status":"ABANDONED","created":"2021-05-04 11:07:30.000000000","updated":"2021-05-04 19:31:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"987adcf9e47e4ecfa1f115d30f3e99b3d107564f","_number":789550,"virtual_id_number":789550,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-04 12:44:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 12:44:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b326559481d9575a2bc839f9381ecada61ef0261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-04 11:07:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b61c4fb10f418fd374b60974e2f68eddb8855136","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-04 11:30:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4ca570517bdc60e9aa3c0c1162ef2a3c708b44d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 12:44:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/db6b1ac0a185462cbf3f5193c6621918 : FAILURE in 23m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/182b20069f4b44c7a8d5860746b17abb : FAILURE in 34m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f2f22b8cf1224931923464739b1f9c08 : SUCCESS in 1h 02m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a00f39a3c3d44f30822fe2c07e7afbce : SUCCESS in 1h 02m 18s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c5e0edc7f5914959850dd99b492095a5 : SUCCESS in 1h 12m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ce2b6e08315a428eaf78d32efcfce1ca : SUCCESS in 33m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/02d262fa7fdc4f0db6bd0dd58254aee0 : SUCCESS in 17m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/733ca30d4c664517b4e5a6a3834de9b4 : SUCCESS in 54m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7778476f25247a5b90dbb6f48fa7559 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4da0d9a9694b4e27b4d4057b2b3c48d2 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c5ee9b25d0d49dbb3ae55cb9a1201b2 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2a6a8b793cb419abb23308d3fc34eb3 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ca82fbfbf34ee0ad0c21f247482a4a : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/929e270933854c529dbfd31cdc422140 : SUCCESS in 21m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d990cd11dfb4294943f5984606aa55e : SUCCESS in 32m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6663eea4f18342968575e03f7d1b6d2d : SUCCESS in 51m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f515ee44e70f471fb78c59bd60532f0e : SUCCESS in 50m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/56f22dc4c8b748ae81c250b882041564 : SUCCESS in 57m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d119921b406ae53df335bb2ead135526f0370574","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 13:37:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5aa660a6f17047138ba31b8eed3b043a : SUCCESS in 46m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6f0e4d2424546509a33b386b982867f : SUCCESS in 50m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ace0f55bef14c74bfc2402afbc6be33 : SUCCESS in 2h 05m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"987adcf9e47e4ecfa1f115d30f3e99b3d107564f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-04 19:31:01.000000000","message":"Abandoned\n\nLooks like upstream sorted out back","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b5209ee9bdb8c1f89dae27cfebd63ca3a7849de8","revisions":{"0792a035d21526161b62be543383b1a16af8152a":{"kind":"REWORK","_number":1,"created":"2021-05-04 11:07:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/789550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/789550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/789550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/789550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/789550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/789550/1"}}},"commit":{"parents":[{"commit":"39000f90b78e80b68cd32e006ba72c76df7e02a2","subject":"Merge \"Revert \"CI: temporarily disable CentOS/AArch64 testing\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39000f90b78e80b68cd32e006ba72c76df7e02a2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-04 10:40:33.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-04 11:06:48.000000000","tz":120},"subject":"base: use proper key for influxdb APT repository","message":"base: use proper key for influxdb APT repository\n\nChange-Id: I639765c400b74518bea6c6c54cba97491dcd50b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0792a035d21526161b62be543383b1a16af8152a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0792a035d21526161b62be543383b1a16af8152a"}]},"branch":"refs/heads/master"},"b5209ee9bdb8c1f89dae27cfebd63ca3a7849de8":{"kind":"REWORK","_number":2,"created":"2021-05-04 11:30:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/50/789550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/789550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/789550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/789550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/789550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/789550/2"}}},"commit":{"parents":[{"commit":"39000f90b78e80b68cd32e006ba72c76df7e02a2","subject":"Merge \"Revert \"CI: temporarily disable CentOS/AArch64 testing\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39000f90b78e80b68cd32e006ba72c76df7e02a2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-04 10:40:33.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-04 11:29:54.000000000","tz":120},"subject":"base: use proper keys for influxdb repositories","message":"base: use proper keys for influxdb repositories\n\nLooks like InfluxDB team is working on changing their repos.\n\nChange-Id: I639765c400b74518bea6c6c54cba97491dcd50b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5209ee9bdb8c1f89dae27cfebd63ca3a7849de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5209ee9bdb8c1f89dae27cfebd63ca3a7849de8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
