)]}'
{"id":"openstack%2Fkolla~789741","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I51c7c6e2f520a582de0409025eb5d1a6da2952a2","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"I51c7c6e2f520a582de0409025eb5d1a6da2952a2","subject":"swift: fix group membership in debuntu/binary","status":"MERGED","created":"2021-05-05 09:39:25.000000000","updated":"2021-05-06 15:13:54.000000000","submitted":"2021-05-06 15:11:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789741","cherry_pick_of_change":763947,"cherry_pick_of_patch_set":1,"meta_rev_id":"aa530613cbf5529b7a142f7eac6fc776888e338f","_number":789741,"virtual_id_number":789741,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-06 15:11:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-05-06 13:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-05-05 11:30:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-05-06 13:20:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 10:58:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 13:20:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d71ce8e02ce9980a4ded9ed1463e47f21c1069fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-05 09:39:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"48d73cbce1c49e1cd43327cdd9cf5cf4f07973f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 10:58:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2f88c9c4d2e2499d85af98a8dcb5e502 : SUCCESS in 21m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f3e56a1268cd4d0ab000bb7892e0c89b : SUCCESS in 28m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a458e35dd8124762aa2efbe72fd8a610 : SUCCESS in 52m 12s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1c0f388fa600486b84d855a982c3ae80 : SUCCESS in 1h 06m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3977238f6a2c400da74659fbe79ca117 : SUCCESS in 1h 13m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/005e396af69445efbbc8ec289ef0222d : SUCCESS in 23m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dcf21d6c02f544fe846f41a2f1363ca2 : SUCCESS in 19m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b3879ce3cd1d4496858a7917984a73db : SUCCESS in 57m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9705d9ef5f0640e38d5714c113aa3362 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a4b5d594bc84e858320ed6fd4cba484 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8918c08efca3480b9e9fd3f8b0f0cace : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e2b6982abc4ee6826153e167624d20 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a5a9ad3213b4a16934465109fea59a8 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47577cb3d59d4f3c8167a6e0f88fe0ab : SUCCESS in 27m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/199b16aa322c414691c636ab6510ccc0 : SUCCESS in 22m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07ac1616fb9a420d9d0d05933579cd66 : SUCCESS in 46m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2e89fa465c34433f84572043c06562f1 : SUCCESS in 57m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/740d3cc633404dd090b80162143d9d25 : SUCCESS in 1h 05m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3389255b7d96039f2c46163323be396696599d96","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-05 11:30:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b48135b5f93c6aaaf30fd9b5433f53f94123dc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:51:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c39661e06a9d44088a5ad57a402f9bb9 : TIMED_OUT in 3h 02m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb2cc868f5974a0f9bafb2499b94cafa : TIMED_OUT in 3h 02m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/907959031e314a9f88e2c578bc642d58 : TIMED_OUT in 3h 03m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b07077c3564645efb1db6aa3d74dfec8 : SUCCESS in 1h 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47f5d9681a2f346e15c94a085b08449a9b72d33e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-06 13:20:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3aed55d47e44f66fe323a799413882324842407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:20:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"878fa45d6415c8d032008addef0e461ff9329895","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:11:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5ca9be5fc54a449ebbaeddf770185453 : SUCCESS in 31m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/12139da68c1c48c78c53b9da39f05030 : SUCCESS in 1h 00m 58s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6187f6c84c14429da389c8fd735f9f47 : SUCCESS in 1h 47m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc9a0697c444d8283ef8f8772747c32 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e761138db62647ef9039992e9580f4de : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11b50c339bc544739fdd2e0b33fd21ca : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58dec18bd53b4c1f997d0a06fd27178c : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cfb7df77c814b82866f04837931d331 : SUCCESS in 5m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/397233c7c0fb46699b8c6c50fee1dbd5 : SUCCESS in 24m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4913388dbb14e3683d3f102405e1c23 : SUCCESS in 49m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f8b1043eccd24097af900d9290b8b53f : SUCCESS in 1h 09m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5a3312f33cdb4472b8910feff9d686852b838e36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:11:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aa530613cbf5529b7a142f7eac6fc776888e338f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:13:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5252d54ec88c4d49847cdfcacf8f8917 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/018ef62b2fff467cbb260afbbc2e111b : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc97aa584689469eacdb3b80f3d2706f9c6bff96","revisions":{"bc97aa584689469eacdb3b80f3d2706f9c6bff96":{"kind":"REWORK","_number":1,"created":"2021-05-05 09:39:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/41/789741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/789741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/789741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/789741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/789741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/789741/1"}}},"commit":{"parents":[{"commit":"29bb43d53b611aa66d470c27550fae5cebced6ce","subject":"Install RabbitMQ repo key on Ubuntu and Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29bb43d53b611aa66d470c27550fae5cebced6ce"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 09:27:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-05 09:39:25.000000000","tz":0},"subject":"swift: fix group membership in debuntu/binary","message":"swift: fix group membership in debuntu/binary\n\nThe UCA packages seem to be removing the swift user from the kolla\ngroup. Explicitly add it after installation.\n\nCloses-Bug: #1905279\n\nChange-Id: I51c7c6e2f520a582de0409025eb5d1a6da2952a2\n(cherry picked from commit aa3208ea0f10889469f864aaccdb593547412d76)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc97aa584689469eacdb3b80f3d2706f9c6bff96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc97aa584689469eacdb3b80f3d2706f9c6bff96"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
