)]}'
{"id":"openstack%2Fkolla~789742","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I51c7c6e2f520a582de0409025eb5d1a6da2952a2","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I51c7c6e2f520a582de0409025eb5d1a6da2952a2","subject":"swift: fix group membership in debuntu/binary","status":"MERGED","created":"2021-05-05 09:39:32.000000000","updated":"2021-05-06 14:34:01.000000000","submitted":"2021-05-06 14:32:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789742","cherry_pick_of_change":763947,"cherry_pick_of_patch_set":1,"meta_rev_id":"fc52db3c145180466d6c40c960c22641eab3a57e","_number":789742,"virtual_id_number":789742,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-06 14:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-05-06 13:20:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-05-05 11:30:32.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-05-06 13:20:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 10:56:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 13:20:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"22204c91adea3e4c401c9457a554ab7601cd3b1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-05 09:39:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ed1051ab621dc5b184effe24392c9318e0b0ac39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 10:56:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d96728b1eb5348019de0fcfde1c901a8 : SUCCESS in 34m 11s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0103fc5e22ad486e9dedc1425c9ad78d : SUCCESS in 39m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/20f4d3a120b64125aa9772b34ce823e9 : SUCCESS in 52m 43s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/26b2d5b9fbc143bc8c01a8d9ae1bc8ad : SUCCESS in 54m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/dbeff994497b478e8d0a823dbfaf40d7 : SUCCESS in 1h 13m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d9f4e992928e4b84851ea43867055767 : SUCCESS in 24m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8fcff7ecb9e14fbe9a73c12ef2e0a018 : SUCCESS in 29m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/77bfdacc31924d25bffb6b942cd26156 : SUCCESS in 1h 03m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada41596555449ffa7e713ba2af90946 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39bae7f9e6a8437f9673f3266c5807ec : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41974cf6a9ba4fee92b8c680a1142e42 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31cfe2f97144dd1bab790e1e64ab3cc : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbee8e120f064d5aabf027f494b0d4f2 : SUCCESS in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e992e95744a4343976c98955750a63f : SUCCESS in 21m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c22735c4931481da3fd5c9e37567425 : SUCCESS in 44m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c4de960232a4a12bf55304984ad3373 : SUCCESS in 56m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9202ce96e2e5463aa805f914cd007de3 : SUCCESS in 55m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3d04e6f0062841fead4f9798a849bbf5 : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5c98cdfe37a4bac0c6c4fe70ad7ffb9defecc13e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-05 11:30:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3d337e6fc32a19abbdbbed3272aca72ca1cddfae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:54:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c361b7c0e6d406fb76ef5930d2c713b : TIMED_OUT in 3h 02m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8945e12a1d094e94a5514dc1eef5e4d4 : TIMED_OUT in 3h 01m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fbf3629ad04646a18f7e07b1cefbeaa7 : TIMED_OUT in 3h 02m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ea3bfc9c0574398bfbcc86b7c20778d : TIMED_OUT in 3h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"273551c4024949a3507db8e0024169929ec7e896","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-06 13:20:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2543a5cc66f1c01ae5ad6abb48cebfb968c6d57c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:20:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7f35f7112d2307fc0159962befa3df346363382","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 14:32:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d74d110db7f04dbb84d96038f01b73ca : SUCCESS in 25m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/54d87e29d03a4547b666de9f5a277746 : SUCCESS in 1h 07m 38s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b87c1f43d8534f70b8da0f6c2c190e72 : SUCCESS in 1h 09m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dad0d1b1ba0e455daffd819883bf67bf : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd967aa05a6d4850be3b2f07d7956e5f : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e4823eae15f434fa3cd89b2ee64b8c0 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a282b30bcc0d40ebab49a3c904c76ab5 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a173dcf1ae214bf49aeeb6dda19f61ce : SUCCESS in 4m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b32a0282c2147f2ae1586040576dbe4 : SUCCESS in 29m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/88641141eea042588d6115fab40c7a34 : SUCCESS in 47m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5c81df50afad407c87ee67904c65c19e : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f75ade5811730c2ece9ab0b9ec667962aca7e47f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 14:32:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fc52db3c145180466d6c40c960c22641eab3a57e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 14:34:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66b1724930442fdaf6f39dc0202603b : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85eaa65e512d408cb33fc9676650c0d5 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8f23bf71c464d18868e4c456a8a45fd53cd74fa","revisions":{"b8f23bf71c464d18868e4c456a8a45fd53cd74fa":{"kind":"REWORK","_number":1,"created":"2021-05-05 09:39:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/789742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/789742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/789742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/789742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/789742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/789742/1"}}},"commit":{"parents":[{"commit":"f3704f99f3f9484805ff05b0a6c00288f0c0bb9a","subject":"Install RabbitMQ repo key on Ubuntu and Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3704f99f3f9484805ff05b0a6c00288f0c0bb9a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-24 09:27:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-05 09:39:32.000000000","tz":0},"subject":"swift: fix group membership in debuntu/binary","message":"swift: fix group membership in debuntu/binary\n\nThe UCA packages seem to be removing the swift user from the kolla\ngroup. Explicitly add it after installation.\n\nCloses-Bug: #1905279\n\nChange-Id: I51c7c6e2f520a582de0409025eb5d1a6da2952a2\n(cherry picked from commit aa3208ea0f10889469f864aaccdb593547412d76)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8f23bf71c464d18868e4c456a8a45fd53cd74fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8f23bf71c464d18868e4c456a8a45fd53cd74fa"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
