)]}'
{"id":"openstack%2Fkolla~789743","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Iad84dd12432befe9250c50d4094b1e5e2e0cf8ee","project":"openstack/kolla","branch":"stable/victoria","topic":"ovs_dpdk_kolla_ansible_crudini-stable/victoria","hashtags":[],"change_id":"Iad84dd12432befe9250c50d4094b1e5e2e0cf8ee","subject":"Add `crudini` to ovs-dpdk containers","status":"MERGED","created":"2021-05-05 09:44:21.000000000","updated":"2021-05-06 14:28:17.000000000","submitted":"2021-05-06 14:26:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789743-ovs_dpdk_kolla_ansible_crudini-stable/victoria","cherry_pick_of_change":775821,"cherry_pick_of_patch_set":1,"meta_rev_id":"e11a03d885ef3b6919cf995b11dde4998f454b27","_number":789743,"virtual_id_number":789743,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-06 14:26:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-05-06 13:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-05-05 11:30:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-05-06 13:19:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 11:04:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 13:19:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bdd839cbb038a18935ee4c2d6ceca42eeb91d8e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-05 09:44:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"577dd28df8060b94649003728010d90d1576410f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 11:04:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/88f40422ac894e6794d5ab443f1d598a : SUCCESS in 30m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2ccb5ab4aed94bfba9e32e816150c5f0 : SUCCESS in 25m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/15b4da46feec44f9ace28038744dfad9 : SUCCESS in 1h 06m 47s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a3a0a5782c984c5b962958f58219069a : SUCCESS in 1h 11m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a3cbfa4448d94984800f9987c4b2f3bc : SUCCESS in 1h 16m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/08f2bdabfc70435f8411faf49ed3f45c : SUCCESS in 25m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cfac9781cde046d1a3ded93b3c5fc9b5 : SUCCESS in 25m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/377fdf90847f4b1d9feb13fbc14fc8c7 : SUCCESS in 55m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828b4454b7a749569de205e9de5d2ff7 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21a9e573569b4dbf88f2a3db791652fa : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4883d6eee434342810103b85a72d503 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b47b8e88904dc38715e61959502951 : SUCCESS in 6m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/60bbb3a24c23495a946d48f1411295ed : SUCCESS in 32m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb144d57c2744d379c90c96da812ab9c : SUCCESS in 25m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa9f70d580634d62bf320d05f7ac32e7 : SUCCESS in 56m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/da107e1025644965970182cc3c338daa : SUCCESS in 45m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a652f217b78247b3a4acaf7441e0ed7a : SUCCESS in 52m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d7f49b989d932c60bbf79a6166be1996f63e9e82","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-05 11:30:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed968040a4b7a8dd17514632c39eb60e1678d555","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:21:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/68dad6f2bb1b4681a29fb3c0169e8216 : SUCCESS in 2h 32m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2fd1b238230e421099d31f83a72919c5 : TIMED_OUT in 3h 05m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1b9860baf24488384834608f5a2bb29 : TIMED_OUT in 3h 02m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/90a0432cf87e4215b68dad14f8daa82e : TIMED_OUT in 3h 02m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2c5335068f0048d2c84e435b5d74bcba0f78069","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-06 13:19:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6458a727a99f55e7bf8ed8d1c46b503ecefe8c7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 13:20:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44782e90d42174cfcf6e38d2236c6c8cf3c7bc74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 14:26:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7062903b7232488bb5b496a089ddfca3 : SUCCESS in 34m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2e74bdc0417a483fb3adea62ae2d7652 : SUCCESS in 1h 02m 40s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3983537bc24a46bc9ee13ba877211ac6 : SUCCESS in 1h 01m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e57e93eb317450e934eb9d91b79dc9b : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff0b9d365e474b799fece48727047575 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d07bcccc9203403b85052f3f41c6ea5e : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77866f572635464092498aedfb199e87 : SUCCESS in 6m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e68e60884d824127aff9ac28d9ddd60b : SUCCESS in 29m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/398a9b0661f740afa0d671067b7deeed : SUCCESS in 56m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9295e9f394344d15bdbf2cd565d42c60 : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":1},{"id":"061c4357a2ae2faa1a4e4fd137582f9c175e8e5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 14:26:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e11a03d885ef3b6919cf995b11dde4998f454b27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 14:28:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9498c1b71341e8821a4fa915d76500 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a596231e3e12f0b5bbc35a6541c4f0b8a01db98e","revisions":{"a596231e3e12f0b5bbc35a6541c4f0b8a01db98e":{"kind":"REWORK","_number":1,"created":"2021-05-05 09:44:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/789743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/789743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/789743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/789743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/789743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/789743/1"}}},"commit":{"parents":[{"commit":"29bb43d53b611aa66d470c27550fae5cebced6ce","subject":"Install RabbitMQ repo key on Ubuntu and Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29bb43d53b611aa66d470c27550fae5cebced6ce"}]}],"author":{"name":"Alexander Evseev","email":"alevseev@sbercloud.ru","date":"2021-02-16 12:32:01.000000000","tz":180},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-05 09:44:21.000000000","tz":0},"subject":"Add `crudini` to ovs-dpdk containers","message":"Add `crudini` to ovs-dpdk containers\n\nkolla-ansible runs ovs-dpdkctl.sh inside container, and ovs-dpdkctl.sh\nrequires crudini.\n\nChange-Id: Iad84dd12432befe9250c50d4094b1e5e2e0cf8ee\n(cherry picked from commit 6585332a63b277ff0d63c6c7bd0e77d640579a0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a596231e3e12f0b5bbc35a6541c4f0b8a01db98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a596231e3e12f0b5bbc35a6541c4f0b8a01db98e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
