)]}'
{"id":"openstack%2Fkolla~789803","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I4cd3d7f67068157b0de2c0a61dff622a1968b71d","project":"openstack/kolla","branch":"stable/wallaby","topic":"deprecate-tempest-and-rally-stable/wallaby","hashtags":[],"change_id":"I4cd3d7f67068157b0de2c0a61dff622a1968b71d","subject":"Deprecate tempest and rally","status":"MERGED","created":"2021-05-05 11:52:55.000000000","updated":"2021-05-06 10:17:41.000000000","submitted":"2021-05-06 10:16:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789803-deprecate-tempest-and-rally-stable/wallaby","cherry_pick_of_change":789731,"cherry_pick_of_patch_set":1,"meta_rev_id":"576ee93feb01a169ce9f50be606e613854ab4a84","_number":789803,"virtual_id_number":789803,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-06 10:16:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-05-06 08:40:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-05-05 14:49:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-05-06 08:40:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 13:04:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-05 13:39:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"80bbec3f6f34afc7d2953780ba8bfb1a11ea682f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-05 11:52:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"80e0468a378bddbb0015103e7511c5ac7c048b8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 13:04:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/195a5383278e456c8f68f7ad84e125e9 : SUCCESS in 3m 23s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1ddcd7a22af540049a2d3e18f07c2792 : SUCCESS in 39m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b89f53de3f6747e3a1ec8d0af7a72b7d : SUCCESS in 30m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6c15501d21554b6d8f063fad00f7129d : SUCCESS in 55m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/78513400ebc8449fabfa59961df2afa3 : SUCCESS in 1h 05m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4d161ba7082f4f9dbcc746cacde27565 : SUCCESS in 1h 05m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e58acebdc7db452a90a6e8253c1451d2 : SUCCESS in 28m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0287466ec2364fc09643e501ee47c02e : SUCCESS in 17m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/abc70c57e8214e88b7e12d9cc5808803 : SUCCESS in 56m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54bd43559c514d95a84e2a7b38fde03c : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/349ab36fa3e241ab9c091089f75a80e2 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1d851e8002c4f3cad7d829e36236e53 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1e3d0bf10e54fde906d255df1b7372b : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2b33ae42ca4841b6c9e8276269b875 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7899584d63ab42a1ad847039c5608eb0 : POST_FAILURE in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/74ce29a4ab264ef483319ef365fcdd19 : SUCCESS in 20m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/089d652aa436481aa5cfeec1cf5adda2 : SUCCESS in 41m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cbe7cd38109493cb830e15096317bc0 : SUCCESS in 53m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6bc4570b2f964dd0a1307c9391057603 : SUCCESS in 48m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/073eeb0057354622a2c5db755a2fb0de : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":1},{"id":"016237e6cc8a4694d180feac0a37e9ba1532acfc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-05 13:06:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18969fe1b4791a52c9a1ef9cd83ca5651761ca0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-05 13:30:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d2ca570e3428550b00160a6316719466c0b66ea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-05 13:39:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c2311e3d26f3ca2124bef598faad2c278ae033ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 14:38:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5fb7dc100ed948d4828535a35ce71327 : SUCCESS in 3m 23s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e32109a3538040d1a92052565c5ca887 : SUCCESS in 22m 01s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ba1db0e13a604c04856c2ea7cf1ee621 : SUCCESS in 38m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6521da96edd74807831c4227cf5d6316 : SUCCESS in 1h 02m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/981f672dab924065afbba84f4ae839b2 : SUCCESS in 53m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/182a70b2cc0240bb8ead0e4170e2de96 : SUCCESS in 1h 05m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/732026e7a6bb4e45aa006316ce531b4d : SUCCESS in 35m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8653e23076f344c99e2b6aa7bd3e5bed : SUCCESS in 22m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0895f2b1e7134bb3a60c338abf197346 : SUCCESS in 1h 05m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05bddef4644d440d9695b2ea6ae496a2 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f5ff1f8c7344fe6b3c507a8db473a2a : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8cd8b0d45ab40f8ac334cf6d1035e97 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a3090bf0fd54c83b28f1630fa8f3f96 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3939f8fa60641398ea26c5b466570ee : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/867007569c8540ab81551383e74a078a : SUCCESS in 4m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4bba36cb49f74359babfaab5014d7c09 : SUCCESS in 23m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/00617902676e4d0caef18ecf4f1790ba : SUCCESS in 38m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0d44a0552b74d318298fdc8e0ac0b35 : SUCCESS in 1h 04m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5785a15b51344ba98e9e43560af2d5cf : SUCCESS in 52m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/09ea4cae19b344f6958e1ea3c5708416 : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":2},{"id":"5cf6915bb4016ccf47f221e8422382205a58ad58","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-05 14:49:08.000000000","message":"Patch Set 2: Code-Review+2\n\nwas my backport not good enough? 😄","accounts_in_message":[],"_revision_number":2},{"id":"14826b68a0c0d5337e82cb2ef18a54e748058d68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 16:00:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b4fcda7c8fd46888da4f7d981def36f : SUCCESS in 2h 26m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6886122fc2614d8fbe520e4409cfaa90 : SUCCESS in 52m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/84a52b5c58234a639bce041bbfca0044 : SUCCESS in 1h 26m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/06fd230f63074f2aa3c006328236ca29 : SUCCESS in 1h 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b876219dad228b0b69d7310c936df2a6faec98a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-06 08:40:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5281f12b73fe87f732f49ef9647d21874a539f5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 08:41:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c593a686824b22a2eadc8c547484eaa7d462cd87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 10:16:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0895bcc377434737b7ada91b6966c64a : SUCCESS in 2m 56s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d8e2908f380b4ee4a3d8d82b4b4f2cfa : SUCCESS in 39m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b1c4c64c3a8e46b8aa1de0b9bba1189a : SUCCESS in 1h 04m 35s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b3af06cab0db4e9eadfcd3c193606d66 : SUCCESS in 1h 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a44fbf6c81645ae913c3f3a2d57380d : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4fb2175e4fa45798e0ac5ecf0a05d53 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea99de1a3ae64cee807b19e7d96402b4 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de86937d2ede46c7b1670f28ebcc4ba7 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01bec15a911844b38c8e450a61c3bfc5 : SUCCESS in 5m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c26961b842746788b14419319b1112e : SUCCESS in 41m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fda78935637949fda766f0c75c057722 : SUCCESS in 48m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8b0ebed231fb45878529f1b8500e6553 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3460f0bdb447ba532378135d1911847338d44496","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 10:16:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"576ee93feb01a169ce9f50be606e613854ab4a84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 10:17:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba18b9933bb4781a4bf300b4dd393ca : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89fb80e199a94f6d8f54c8f241200474 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f885c6c1d4cc7dd8d204578b3b8f4e724ab435cf","revisions":{"c17d95064d947c66d4edf2f15089dddf626a471a":{"kind":"REWORK","_number":1,"created":"2021-05-05 11:52:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/03/789803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/789803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/789803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/789803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/789803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/789803/1"}}},"commit":{"parents":[{"commit":"c01b017e59ecb62dd59101fcf957964563afdf40","subject":"Revert \"CI: drop ubuntu-aarch64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c01b017e59ecb62dd59101fcf957964563afdf40"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-05 08:30:11.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-05 11:52:55.000000000","tz":0},"subject":"Deprecate tempest and rally","message":"Deprecate tempest and rally\n\nPer [1].\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-04-28-15.00.html\n\nChange-Id: I4cd3d7f67068157b0de2c0a61dff622a1968b71d\n(cherry picked from commit e01a490c97286a7aac66244aee7b79836d371422)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c17d95064d947c66d4edf2f15089dddf626a471a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c17d95064d947c66d4edf2f15089dddf626a471a"}]},"branch":"refs/heads/stable/wallaby"},"f885c6c1d4cc7dd8d204578b3b8f4e724ab435cf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-05 13:30:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/789803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/789803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/789803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/789803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/789803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/789803/2"}}},"commit":{"parents":[{"commit":"2e5e8ba8580c2bcc4d82add8a3c4c5f05d73d6fb","subject":"Merge \"[CI] Trust only infra mirrors\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e5e8ba8580c2bcc4d82add8a3c4c5f05d73d6fb"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-05 08:30:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-05 13:30:44.000000000","tz":0},"subject":"Deprecate tempest and rally","message":"Deprecate tempest and rally\n\nPer [1].\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-04-28-15.00.html\n\nChange-Id: I4cd3d7f67068157b0de2c0a61dff622a1968b71d\n(cherry picked from commit e01a490c97286a7aac66244aee7b79836d371422)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f885c6c1d4cc7dd8d204578b3b8f4e724ab435cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f885c6c1d4cc7dd8d204578b3b8f4e724ab435cf"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
