)]}'
{"id":"openstack%2Fkolla~790261","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ic5abc4b87fd76f87aba383abf43e95ba70629fcb","project":"openstack/kolla","branch":"stable/train","topic":"ci-trust-only-infra-mirrors-stable/wallaby-stable/victoria-stable/ussuri-stable/train","hashtags":[],"change_id":"Ic5abc4b87fd76f87aba383abf43e95ba70629fcb","subject":"[CI] Trust only infra mirrors and fix APT mirrors","status":"MERGED","created":"2021-05-07 18:25:41.000000000","updated":"2021-05-10 18:41:11.000000000","submitted":"2021-05-10 18:39:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790261-ci-trust-only-infra-mirrors-stable/wallaby-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":789754,"cherry_pick_of_patch_set":1,"meta_rev_id":"b40cdd33f2f7af6ae713ad5fe5eabcd8c7f1773c","_number":790261,"virtual_id_number":790261,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-10 18:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-05-10 17:29:41.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-05-10 14:31:11.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-05-10 17:29:41.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-05-10 18:41:11.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-07 18:53:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-05-07 18:53:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-05-07 18:53:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-05-07 20:19:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4bc97b9f391f3d54abb4753d9b13ab87384b6f2f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-07 18:25:41.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.\n\nThe following files contain Git conflicts:\n* tests/templates/template_overrides.j2","accounts_in_message":[],"_revision_number":1},{"id":"4e6606f5bed73af5285879a40aabdec88a517000","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-07 18:40:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"079e54678353701747454f8a06f8dcc0c8d064ce","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-07 18:44:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"41635cb70b78726fc4010f05031c604144d00e6c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-07 18:48:04.000000000","message":"Patch Set 3:\n\neh, still conflicts due to other backports going unclean before and amended differently","accounts_in_message":[],"_revision_number":3},{"id":"58c5d33738a5168242b45eb016dd88950ec428b6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-07 18:50:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cc2098011258c658f33c9cae6b37abe34ec070a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-07 18:53:21.000000000","message":"Patch Set 4: Review-Priority+1\n\nneed to merge the whole stack to get a clean situation here on train","accounts_in_message":[],"_revision_number":4},{"id":"415c8f43646fd918749e1950680bc8e138682dbf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-07 19:29:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"221367666a26a39b86add1d3934e923d23288613","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-07 19:53:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"902434d574ddf719bb58b51ac2471b032088f47c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 20:19:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/52387220c35e4a3e935e94015f043612 : SUCCESS in 41m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4e546853bc6744e5a665aea6477249e0 : SUCCESS in 47m 59s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7807377c072c41dbb0742613f9eecb2a : SUCCESS in 29m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c22489c409244229a2bfa1c45a77df36 : SUCCESS in 33m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6d1e0f9538284fcaad899aa5236f1553 : SUCCESS in 1h 17m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8f7dd840071d4ea79263bcfc4afec114 : SUCCESS in 57m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f730e9bcff284640a1978229ab13e0ae : SUCCESS in 1h 14m 41s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bdc6da548b224eeab5e40e10f40e61d4 : SUCCESS in 1h 00m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/39d23f9ff9c4443cb0534c21218f10ff : SUCCESS in 1h 27m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ad09e1425a454e73b0471cbee84c6cf0 : SUCCESS in 24m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f920c0e799bb4af284a8b86f60680851 : SUCCESS in 25m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/124aa90e9aac4457a34d0e0757743b89 : SUCCESS in 1h 05m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482d4411efb449ceb5871eec6f9a0106 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d576a7e238a3408d896e780b39e85566 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71006a6f4aa7456bbcf780325ba02f2a : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7857ebfaec134bfeb56ec50faa5283ec : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/173849fe13ca480a96af14e364d40b90 : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c893f1a0c1a94969b859a71f10573b8c : SUCCESS in 32m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e661577e5064d9f95bd44cfa693b53d : SUCCESS in 28m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4726888b23d447f2b4994553574e20a7 : SUCCESS in 56m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8709b5688c9e41b682609ffa2279abd8 : SUCCESS in 50m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6280189dbf124d00888cabbe70d0b7ce : SUCCESS in 1h 03m 23s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"4458c5c42571ded72b7b6c2f75493732cec8b736","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-07 20:36:05.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b0c043faf096f92a1c79ce8fe05b8681d7f60877","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-10 12:16:46.000000000","message":"Patch Set 4:\n\nok, let me squash them all instead to lessen the load","accounts_in_message":[],"_revision_number":4},{"id":"533b6b8ededb5df84f74342e12bf3e6de8ac3197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-10 12:23:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4f097d7d0941531ce5e339910cefaea28fc294b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 13:53:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ee4edd1dd485452a9b1be6d051d99d47 : SUCCESS in 44m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/671ee04785b14dc0849c8b81c0b94539 : SUCCESS in 1h 02m 21s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ae1cce3aea66416bb25ec84b3a27857e : SUCCESS in 28m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cb2fe528618a46cfb172b40e6e67a033 : SUCCESS in 38m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c92ef7961e6f48199775d0945c773841 : SUCCESS in 1h 15m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f4e1e14155db4df299a803e847190880 : SUCCESS in 53m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/59e31e5c00a94fb5b2f98daf35db3e6a : SUCCESS in 1h 24m 42s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ac6fd969a8f64544b4229642c308da63 : SUCCESS in 53m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/68a3e539fd73467996f1c86f8f023ba1 : SUCCESS in 1h 22m 11s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4efa6dbaa1a1478dbf601ff39151dfb2 : SUCCESS in 57m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dee9ba19bb144ad6967c93b7331377f3 : SUCCESS in 22m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4cac113f9cf44f4ab013c1de27b6bc19 : SUCCESS in 24m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f46749a683941d2bda89501e0928e35 : SUCCESS in 1h 23m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fa818b98fe640deac6378f41ab6feef : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e60b12a47c57490cb41179150e7bc1a1 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4da2fd9cc3404f8f9f0a425c0209a8e3 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb6768b774004f22bab737c505be8b03 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1084d443807a477291b4c30c0425098b : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/117dae11d07644d69e1d9af2d4c6694f : SUCCESS in 7m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d83b3645edb34ec98ec59a20916b119b : SUCCESS in 27m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/77b8e9affaf54d2ebc281e010181a22f : SUCCESS in 38m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f15a78b6e6d4411ab992558b20e059f4 : SUCCESS in 43m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3a95c71f76449b38d0a5d01d82d6961 : SUCCESS in 56m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/39dee61b8d9f4ce48277e4ceceb21e43 : SUCCESS in 1h 05m 51s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"b7c82fe3ad6082dce5c94aac673fe99cadd0a84d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-10 14:31:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0b95c0e4050e4917b4345579d8063d4c91e43138","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-10 17:29:41.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"7b6594f8edc2175123f43254779058266f581a16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 17:30:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"22af0142348441f5b45958068f46643bf07e60af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 18:39:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d1cf4535ba184b28944aa6f5e93534a3 : SUCCESS in 39m 15s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d5caed6fbe8b4dffa70a785155e18bd7 : SUCCESS in 26m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0701ad11b3b34c25ad7ca80acd8b40c4 : SUCCESS in 1h 03m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/91a57997b3f14490a9238eb99e7608be : SUCCESS in 44m 46s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ef9478d3a4e44f3db536131d8070de34 : SUCCESS in 1h 03m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/14f18ba1a2fd49faa9274ff003cf9b4c : SUCCESS in 51m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc37710b78a643dfa3387ff240b85417 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c9a4152330a4e32a91c00e831bf9b11 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82f7ffaf491e4a5d88b8f5a97a11212d : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/425c6e1c177745518b34f55deb463c08 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b72778087a949d6bee4782d57dbbad5 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49c2bfe447cc413eac53c136a7a1437d : SUCCESS in 4m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c188de7ce944753997ad8ed22bd3982 : SUCCESS in 24m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa0319fb1b404f91804ffd4f760ae6be : SUCCESS in 52m 59s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/39b2250e942b4032be7bc8d2fc8e8fb0 : SUCCESS in 1h 01m 59s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"22a2fe2ede3134118d9b1850dbdb29c62fd1a921","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 18:39:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"05f37eafeb25fc56a560bd83e9e7e34792e32e59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 18:41:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c0dc8b1a5147a28459690ffb0b6099 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e829d350c2f94dbb9b0c6d2a9f74ffd4 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3b378b0a10570b91549a23055863d14e869768b5","revisions":{"c10f8af5daf96bfd3ccbeef9f301c4e66b0e533b":{"kind":"REWORK","_number":1,"created":"2021-05-07 18:25:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/61/790261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/790261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/790261/1"}}},"commit":{"parents":[{"commit":"c9b0e2ae402345a606a90878a643ee5560615c1a","subject":"rabbitmq: Move to packagecloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9b0e2ae402345a606a90878a643ee5560615c1a"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-30 15:12:53.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-07 18:25:41.000000000","tz":0},"subject":"[CI] Trust only infra mirrors","message":"[CI] Trust only infra mirrors\n\nInfra mirrors get their indices rebuilt to avoid broken\nindices (due to partial update).\nUnfortunately, this wipes out the cryptographic signatures.\nOur approach so far was disabling apt security features globally.\nHowever, this is not a valid choice for external repos.\nIt hid an issue we introduced with new RabbitMQ repos missing\nproper keys installed in the image.\nThis caused permanent failures outside of our CI.\n\nOur process should be as close as possible to users\u0027 experience.\nThis patch makes CI trust only the mirrors that have their indices\nrebuilt (so infra mirrors).\n\nChange-Id: Ic5abc4b87fd76f87aba383abf43e95ba70629fcb\n(cherry picked from commit c364c8be85f4653bf7af1c7ac0378823553e7441)\n(cherry picked from commit 4a482380408d107c59119b379add34f4b8e87ad7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c10f8af5daf96bfd3ccbeef9f301c4e66b0e533b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c10f8af5daf96bfd3ccbeef9f301c4e66b0e533b"}]},"branch":"refs/heads/stable/train"},"0b924fb65a8f6edc34cd2466232d2d3ba48f91db":{"kind":"REWORK","_number":2,"created":"2021-05-07 18:40:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/61/790261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/790261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/790261/2"}}},"commit":{"parents":[{"commit":"afa1e286100646fe5daa9a9e60dabc5ec6fb95f5","subject":"CI: use Debian/Ubuntu mirrors in base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/afa1e286100646fe5daa9a9e60dabc5ec6fb95f5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-30 15:12:53.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-07 18:40:47.000000000","tz":0},"subject":"[CI] Trust only infra mirrors","message":"[CI] Trust only infra mirrors\n\nInfra mirrors get their indices rebuilt to avoid broken\nindices (due to partial update).\nUnfortunately, this wipes out the cryptographic signatures.\nOur approach so far was disabling apt security features globally.\nHowever, this is not a valid choice for external repos.\nIt hid an issue we introduced with new RabbitMQ repos missing\nproper keys installed in the image.\nThis caused permanent failures outside of our CI.\n\nOur process should be as close as possible to users\u0027 experience.\nThis patch makes CI trust only the mirrors that have their indices\nrebuilt (so infra mirrors).\n\nChange-Id: Ic5abc4b87fd76f87aba383abf43e95ba70629fcb\n(cherry picked from commit c364c8be85f4653bf7af1c7ac0378823553e7441)\n(cherry picked from commit 4a482380408d107c59119b379add34f4b8e87ad7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b924fb65a8f6edc34cd2466232d2d3ba48f91db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b924fb65a8f6edc34cd2466232d2d3ba48f91db"}]},"branch":"refs/heads/stable/train"},"14152de981be61f615d292477310974a892fc01a":{"kind":"REWORK","_number":3,"created":"2021-05-07 18:44:37.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/61/790261/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/790261/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/790261/3"}}},"commit":{"parents":[{"commit":"f4d7c2ee279cfb76e3a19a044bfc2aec7dce7c46","subject":"Ubuntu: use local APT mirrors if possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4d7c2ee279cfb76e3a19a044bfc2aec7dce7c46"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-30 15:12:53.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-07 18:44:37.000000000","tz":0},"subject":"[CI] Trust only infra mirrors","message":"[CI] Trust only infra mirrors\n\nInfra mirrors get their indices rebuilt to avoid broken\nindices (due to partial update).\nUnfortunately, this wipes out the cryptographic signatures.\nOur approach so far was disabling apt security features globally.\nHowever, this is not a valid choice for external repos.\nIt hid an issue we introduced with new RabbitMQ repos missing\nproper keys installed in the image.\nThis caused permanent failures outside of our CI.\n\nOur process should be as close as possible to users\u0027 experience.\nThis patch makes CI trust only the mirrors that have their indices\nrebuilt (so infra mirrors).\n\nChange-Id: Ic5abc4b87fd76f87aba383abf43e95ba70629fcb\n(cherry picked from commit c364c8be85f4653bf7af1c7ac0378823553e7441)\n(cherry picked from commit 4a482380408d107c59119b379add34f4b8e87ad7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14152de981be61f615d292477310974a892fc01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14152de981be61f615d292477310974a892fc01a"}]},"branch":"refs/heads/stable/train"},"79cd44156b5d26ae2c12fc3b9bbcd8695271704e":{"kind":"REWORK","_number":4,"created":"2021-05-07 18:50:53.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/61/790261/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/790261/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/790261/4"}}},"commit":{"parents":[{"commit":"f4d7c2ee279cfb76e3a19a044bfc2aec7dce7c46","subject":"Ubuntu: use local APT mirrors if possible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4d7c2ee279cfb76e3a19a044bfc2aec7dce7c46"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-04-30 15:12:53.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-07 18:49:55.000000000","tz":0},"subject":"[CI] Trust only infra mirrors","message":"[CI] Trust only infra mirrors\n\nInfra mirrors get their indices rebuilt to avoid broken\nindices (due to partial update).\nUnfortunately, this wipes out the cryptographic signatures.\nOur approach so far was disabling apt security features globally.\nHowever, this is not a valid choice for external repos.\nIt hid an issue we introduced with new RabbitMQ repos missing\nproper keys installed in the image.\nThis caused permanent failures outside of our CI.\n\nOur process should be as close as possible to users\u0027 experience.\nThis patch makes CI trust only the mirrors that have their indices\nrebuilt (so infra mirrors).\n\nFor Train, fixed missing override parts as well.\n\nChange-Id: Ic5abc4b87fd76f87aba383abf43e95ba70629fcb\n(cherry picked from commit c364c8be85f4653bf7af1c7ac0378823553e7441)\n(cherry picked from commit 4a482380408d107c59119b379add34f4b8e87ad7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79cd44156b5d26ae2c12fc3b9bbcd8695271704e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79cd44156b5d26ae2c12fc3b9bbcd8695271704e"}]},"branch":"refs/heads/stable/train"},"3b378b0a10570b91549a23055863d14e869768b5":{"kind":"REWORK","_number":5,"created":"2021-05-10 12:23:15.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/61/790261/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/790261/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/790261/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/790261/5"}}},"commit":{"parents":[{"commit":"c9b0e2ae402345a606a90878a643ee5560615c1a","subject":"rabbitmq: Move to packagecloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9b0e2ae402345a606a90878a643ee5560615c1a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 15:40:50.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-10 12:21:59.000000000","tz":0},"subject":"[CI] Trust only infra mirrors and fix APT mirrors","message":"[CI] Trust only infra mirrors and fix APT mirrors\n\nInfra mirrors get their indices rebuilt to avoid broken\nindices (due to partial update).\nUnfortunately, this wipes out the cryptographic signatures.\nOur approach so far was disabling apt security features globally.\nHowever, this is not a valid choice for external repos.\nIt hid an issue we introduced with new RabbitMQ repos missing\nproper keys installed in the image.\nThis caused permanent failures outside of our CI.\n\nOur process should be as close as possible to users\u0027 experience.\nThis patch makes CI trust only the mirrors that have their indices\nrebuilt (so infra mirrors).\n\nFor Train, fixed missing override parts as well.\nAnd it includes also:\n  - I8209770ca1353fd89aacef0859ca3a373645c6b2\n  - Icb745f628c1c50d9f20df022b08ddc46fccc270c\n  - Id3b55ac470a977be7ee7decdef3b49e8d89a130f\n\nChange-Id: Ic5abc4b87fd76f87aba383abf43e95ba70629fcb\nCo-Authored-By: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n(cherry picked from commit c364c8be85f4653bf7af1c7ac0378823553e7441)\n(cherry picked from commit 4a482380408d107c59119b379add34f4b8e87ad7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b378b0a10570b91549a23055863d14e869768b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b378b0a10570b91549a23055863d14e869768b5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
