)]}'
{"id":"openstack%2Fkolla~790274","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I2b084ea78b236623b174473f411bae04e624a3e9","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"I2b084ea78b236623b174473f411bae04e624a3e9","subject":"CI: drop bashism from template overrides","status":"MERGED","created":"2021-05-10 09:05:38.000000000","updated":"2021-05-10 13:52:48.000000000","submitted":"2021-05-10 13:51:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790274","cherry_pick_of_change":789541,"cherry_pick_of_patch_set":1,"meta_rev_id":"ae5aa11f63d25e41a32c6921ac249c47759e015b","_number":790274,"virtual_id_number":790274,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-10 13:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-05-10 11:22:37.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-05-10 11:12:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-05-10 12:35:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-05-10 11:22:37.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-10 09:05:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-05-10 10:39:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 12:35:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"893dbaf49eadeaf0559155816c27804804e527f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-10 09:05:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"12d4edf714b93893142609eb630b8c55c3aa1431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 10:39:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/576723e8c1304cf093b193ab486ede91 : SUCCESS in 40m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1310352c631c498e90e2ba1ba2427dd2 : SUCCESS in 54m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/903c24f9f2ce4f81832ff00e5462cadf : SUCCESS in 1h 00m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/56a3d8707e454b3b9209e426af5dbba9 : SUCCESS in 1h 01m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f5e03e62a3f142d585ced07f91520015 : SUCCESS in 1h 32m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d15a630538504fcc8bbf3458956ef962 : SUCCESS in 38m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dd775ef9fce34d6aaea51b61d93dc9ed : SUCCESS in 39m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eeb4891a80644c32a69c22f26dc3b5d2 : SUCCESS in 53m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb365a57a0494d17807d0e071661d413 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef65f8a4ac5a48cfb5ede31111fe923b : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/820ed0f04edf465eb5495aa90b86541f : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848eab7a99284b489e2464e714533548 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b6115a010c34d44bad801732b271b54 : SUCCESS in 48m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/17b116b57baf4e7b8151394a3bd49686 : SUCCESS in 40m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fb0a4c5ac8548ab935b0c2920b15932 : SUCCESS in 55m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/19f0df0062704a5c898b8062bf3854d9 : SUCCESS in 1h 04m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5165adf3f91c434d886899601f3e02bf : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":1},{"id":"941981a87682aeb112919dd9922984c99dfcfac4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-10 11:12:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82ad1900a5eafc4684efdce9653add7c30c13c3f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-10 11:22:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c2ba41874ec498b0a16b2e759f768132ac22af5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 11:23:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44ba16350e535a7cdfe87c98d5ea4ff60f97681b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-10 12:35:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"84c3be6e965576edc9f081bdbfea21cf5c963cae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 12:56:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c249b97a35bb43b4b6ff22722b412789 : SUCCESS in 2h 37m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/facd8ddfae6e466c867577b109b01355 : SUCCESS in 2h 15m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/30b70f0c24704045adc681dadc203078 : SUCCESS in 1h 20m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/40db14c79b1e44dfa929d880af2c8b19 : RETRY_LIMIT in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"949472b3d6d28a4e23fff7be6cde4dc43f4d87ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 13:51:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e8259d7b757a48f489a5e8207afbb2b6 : SUCCESS in 51m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/aca91e8059b04ac4a1d79e7955b32686 : SUCCESS in 50m 40s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5fe8ed65e97f482b96f56d035207a772 : SUCCESS in 2h 27m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67b6730a2c4240afbc7ca7576c102f63 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e0b5358c0ad42ee9d1d91880225920d : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff16137848c1411c960b4b84dfa45dd8 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c23e55fa297c43ac9b28d2d4b84e3359 : SUCCESS in 6m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/69196b5dc2884d7a943f295a1a206e75 : SUCCESS in 1h 09m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6cccf7c3067643e3a8ec884522739953 : SUCCESS in 1h 08m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/88374a499d61411ab7cc710a40d60baf : SUCCESS in 2h 20m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e296bb38f349cf5c99f0ab416b78465708cda370","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 13:51:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ae5aa11f63d25e41a32c6921ac249c47759e015b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 13:52:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497e75b32c844edea0c9760ab0db676e : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ec3f577a0c7e35d1c318ff3827ca7bbb0e9b261","revisions":{"8ec3f577a0c7e35d1c318ff3827ca7bbb0e9b261":{"kind":"REWORK","_number":1,"created":"2021-05-10 09:05:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/790274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/790274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/790274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/790274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/790274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/790274/1"}}},"commit":{"parents":[{"commit":"c8e0b4b7bd93b0ff6be113b3804906be1fbd4212","subject":"Merge \"swift: fix group membership in debuntu/binary\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8e0b4b7bd93b0ff6be113b3804906be1fbd4212"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-04 09:30:28.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-10 09:05:38.000000000","tz":0},"subject":"CI: drop bashism from template overrides","message":"CI: drop bashism from template overrides\n\n[[ is bash only\n\nIn many distributions /bin/sh \u003d\u003d bash but not in Debian and derived\nwhere it is dash (very simple POSIX shell).\n\nChange-Id: I2b084ea78b236623b174473f411bae04e624a3e9\n(cherry picked from commit f0c3567740be342355b39a7c7d3ca341d5a4ec0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ec3f577a0c7e35d1c318ff3827ca7bbb0e9b261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ec3f577a0c7e35d1c318ff3827ca7bbb0e9b261"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
