)]}'
{"id":"openstack%2Fkolla~790436","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I2b084ea78b236623b174473f411bae04e624a3e9","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I2b084ea78b236623b174473f411bae04e624a3e9","subject":"CI: drop bashism from template overrides","status":"MERGED","created":"2021-05-10 12:34:33.000000000","updated":"2021-05-11 12:14:37.000000000","submitted":"2021-05-11 12:13:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790436","cherry_pick_of_change":790274,"cherry_pick_of_patch_set":1,"meta_rev_id":"19bb24f4ab671b47e4128986bb8ee1a17d173642","_number":790436,"virtual_id_number":790436,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-11 12:13:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-05-11 12:14:37.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-05-11 12:14:37.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-05-11 05:39:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-10 12:34:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-05-10 12:45:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-05-10 14:08:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b177d3e326ea69dd4002fc203b2a06b4f9e803d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-10 12:34:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"4dd96ebebbffefd5cd7c6a2ccc7a51916992f885","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-10 12:34:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"39a702d5dff2432f5f6630d51ad6c9fa974790dc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-05-10 12:52:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3e0ebabd67311fcb5a62905b8d93814571a876f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 14:08:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/45f7da0a0e3d4e848f8fb78b085619ec : SUCCESS in 43m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/162704cce1674a6ab0a6face5a08f492 : SUCCESS in 54m 52s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1f5218067cc2455db2273d53e16c6346 : SUCCESS in 29m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7f3be403a4964235923478aecd13ffd1 : SUCCESS in 35m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f86fb6d94bb04387b90a23b7d76e7e20 : SUCCESS in 1h 16m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0d59ba6d2574473db34d190436236b7c : SUCCESS in 58m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3e59da6e16f740d69a315c965d0f1fbc : SUCCESS in 1h 14m 39s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9bca07f62f7c4aaea8525517c7536508 : SUCCESS in 57m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d9787f681933428b8e4116c685b76ddc : SUCCESS in 1h 31m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7dff2277a45c4404b6105a1e25f01d4b : SUCCESS in 18m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7b3610b0267241c0b103bd6fcbdb4c60 : SUCCESS in 24m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b839c8507a784f59845f9ce19c90fd8b : SUCCESS in 55m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/061c091f0fa347928c61778b8fbf05de : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8bdfb0ecbbaf419eb60405838ff1505c : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e6537fda02d4f61837a8f16730e91c5 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/647a732aa6bc4343b314496b31f2485f : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4872d5cec847459804fdefbc09f946 : SUCCESS in 5m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/872113271c734e5da732df16042a16ff : SUCCESS in 29m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e838058c9cad41fb8b1e3653100b3571 : SUCCESS in 21m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/87dd521c4e5c4f2f94a5a53007c37329 : SUCCESS in 59m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd0e0ad55de84fc1886e6b26e65fb98d : SUCCESS in 53m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5185930c8abd4c638db7a37418f5c842 : SUCCESS in 58m 43s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"3417bb64fa6eae18c52436f079f406e0869b9b90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 14:09:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43662acb199fd8111ef4b6e168f186d93504bdbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 17:17:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/961e19b818ac4a20b4fa715e8ccfe038 : TIMED_OUT in 3h 03m 41s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fa77c10858134bcdb3924c7a6be6c7b8 : SUCCESS in 33m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e0fdaba52c0547c5a00b62df841990bb : SUCCESS in 1h 10m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ccb838c6e75e43858dc469cac934e50c : SUCCESS in 56m 21s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2f1d214ae179484c91834bb2661d8d1a : SUCCESS in 1h 23m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f866bc23324a84b33ea2138b530065 : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d885e3de6834a0a9b5236c0230112e3 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d33193677984eaba5fbf695f7d533ec : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec8dfff393084fc69b7f04cc1941d790 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ec65abc7864d20a86f31c45648d7a3 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8717c806799e406e931def849438c0fe : SUCCESS in 31m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7eab219ab3f4c20991d97de6b5033b0 : SUCCESS in 49m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/41c20bf95485442f82073ef583f10da7 : SUCCESS in 1h 31m 41s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"99e60fed5a01c2efed90b5d28b563be7a9b7987e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-10 17:18:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"12eecc7fab190338050d9e7c43076e3f63ce065a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 20:27:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ed86df69bc5f4b07aab9715b5b72805d : SUCCESS in 59m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fa8ba61300384358a3d85534d80a7729 : TIMED_OUT in 3h 03m 01s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/300c5a4543914728a2827950187065bc : SUCCESS in 44m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/09d16b75279c4e98973be8c34647698a : SUCCESS in 47m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/71e6bc7a133a42609ded17131657aec0 : SUCCESS in 1h 32m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/80792a9ee6bb49488c6a6dbb93cac4fa : SUCCESS in 1h 20m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dfacce8d09b74cd3bf1053a3a9855b1c : SUCCESS in 1h 26m 28s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4b162107148446988ccac3188f5c8ee3 : SUCCESS in 1h 14m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/28cecc24c17e4151a980e16cf6b43d25 : SUCCESS in 1h 51m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a30f047e0d746ff90a6a987cf30b8db : SUCCESS in 27m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fb981f0d2abf4c82abac489dc531673e : SUCCESS in 19m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d31d6fda95b34029b55d7dc1de6291e8 : SUCCESS in 1h 00m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b151dbea99764989b96e43ce9cbd1556 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b02de72aab284af9ad6b7f62837132d7 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe47ce5afee04b95b1143e8ec4cf4a17 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77bb82fdbcad48b196578a7eb6b3b7db : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e313cadac7e74a3f8e55799ef8732622 : SUCCESS in 6m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18eb8ef11e184de6b23f01763ab7c726 : SUCCESS in 41m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b5cfece707448ab995f16c61de60c76 : SUCCESS in 42m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c57b145c74d54e729c7a261c360270ad : SUCCESS in 1h 06m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85a7b5c8078b42f89b3c17391813f970 : SUCCESS in 55m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5c1abb5487354985805d782beed8f596 : SUCCESS in 1h 24m 20s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"9e3739f80a3d4c1df975d847b56d680ce7f94b42","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-11 05:39:13.000000000","message":"Patch Set 1:\n\ncentos so slow","accounts_in_message":[],"_revision_number":1},{"id":"eae762bde1e36be394872e1cda8b73264d3dd83b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-11 05:39:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"aff41ae023c9215f6fa472e0748f8dcd10a425bc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-11 05:39:26.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eb2756585201fe24f1b74ff882cf107da16b0735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 08:47:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/09adaf6235d64025bba684d641452a08 : SUCCESS in 40m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/371e38530a454288b8efba9a5a3ec012 : TIMED_OUT in 3h 02m 42s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8465e4a614b744b6ae5a4ec349e49c95 : SUCCESS in 26m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6a09e969a4f146b88d79ef97e23d6ca4 : SUCCESS in 32m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f2889ce93f154b9a96bfdf760c8d9638 : SUCCESS in 1h 15m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/835c3839f6b04448a702a1e0ac0f5da6 : SUCCESS in 59m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f0eb8433a37a4e36a2b9ff2fa8d2905a : SUCCESS in 1h 14m 42s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/105f18acffdf4da5bb3dae50e128a4ab : SUCCESS in 1h 02m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/06455b2da8a44c28b5ad54b000709936 : SUCCESS in 1h 36m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/340cae30d6aa4b568b01e38fd6dd80e3 : SUCCESS in 23m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/45e0c488d9e54f97ab2b4cbd51e7af76 : SUCCESS in 23m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c44aee1b710c45d2b20c888e311e00a6 : SUCCESS in 54m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962424a29e59459ab3314896b4f91edb : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e90b5913d45d4e68818af8fe2905b2d1 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33b4cf59056746b1be26e90f7e49731b : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fda3aed924654232a0824fe22c8103ae : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1791c7d324cf4a6298a341e294021f3a : SUCCESS in 6m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ff912ccaf2f4c5d897000f85eb36650 : SUCCESS in 25m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/496e216518dd4fcabf999085c0f60d94 : SUCCESS in 23m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d83e2459b484203b0d6844f263a9c23 : SUCCESS in 52m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0e3f41b50c774fc8bcf3b82ff342f0a8 : SUCCESS in 1h 04m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2c9b70a450054eb1b1a3bb2b3e952fc3 : SUCCESS in 1h 01m 32s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"d8103bc020bc7f071d3b3b900214dbf49f07d1a6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-11 08:55:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"acf54451d31a50cbe52a06de14224811acd5f2b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 10:29:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/dcec73b9c7db4d2d96d24d7e7351624d : SUCCESS in 44m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e41278c29d58432d909af918d94136ea : SUCCESS in 52m 52s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a29b83c6cd384287a85b7498380d7825 : SUCCESS in 25m 50s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7647a924520142998e4dd3555fd77cb6 : SUCCESS in 33m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2b7d2d9eef6b4f45bb5630845f1c809a : SUCCESS in 1h 17m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a41b29257ccc459e9603f4c19aaafec1 : SUCCESS in 1h 03m 07s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0d9cdc2d73aa4b2abc0ccc1c7a61c350 : SUCCESS in 1h 09m 25s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/58d998a124544f1ab0134674e4d66e4a : SUCCESS in 59m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6ddf1ff9a32f4078bcfebe6ce7dafc55 : SUCCESS in 1h 28m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/17bb516f942844c6980f1563c9de491f : SUCCESS in 24m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ce5d1ad2157b460d99fc57edb5612679 : SUCCESS in 30m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e6dbd020ff11475db419eb5ca5400e5e : SUCCESS in 49m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db29b8370fe448f181ea7fd709fd6c05 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a473610afdb4db187c00ba631faa526 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c746995e34e24b84be364750d0af6901 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb0a171edaa4441fac6aeb17787f1e46 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d331f9cb5b6c427aa937fdae68398268 : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/02c6d3583fa14b33aada5c2bd08a6901 : SUCCESS in 29m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c49df7342e9544118f08ade39b3a4446 : SUCCESS in 40m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0423dc5c9ee440808db8b9b444fedb39 : SUCCESS in 43m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/940792244b68446d904d8b86b7d8fc0f : SUCCESS in 1h 00m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92658e209c194fed9c2965b33ce601cd : SUCCESS in 1h 02m 41s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"3d416bbebbd0f493429bfa053b3caa36857ea9c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 10:30:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6dba2807e0b26dacd27140e014b07c1ab00c393c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 12:13:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/faa0694bd3b347cfb6480ae3bf38df26 : SUCCESS in 58m 39s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2f607ca2f3e7492da7fe9594999f6e5d : SUCCESS in 35m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c4a3308c6b7a499b9b444a79029b51f2 : SUCCESS in 1h 16m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e864529e79ed43cf8c4cf83cff6aeb28 : SUCCESS in 57m 58s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/21b4283b02074e958360ad148ce31932 : SUCCESS in 1h 38m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f111ffd9648c4fb9b79415d63c895cfd : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50997104982645c4bc10579b0298b018 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7458745764cb4ea09d66a77f1435f29d : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8984bd647e1b4c1c8693ae4213e9f99e : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc89d8edd5f4362ac3e13086950f415 : SUCCESS in 5m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/59a47b5ca10f408088091f356f442534 : SUCCESS in 25m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/62ba47bd586f4ec1bc9d096eb0bb601d : SUCCESS in 52m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b4727eab67ee43bab72e65180928d6f6 : SUCCESS in 1h 01m 41s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"024f0fb5fc5247eaecef655dd5aef500b77261f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 12:13:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c8ac792109c97bfede575b703123767c5d5ca72e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 12:14:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05d96d9a8b7a4e77a05ed1e415c14fa6 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"326201932f93fc18371c2c5385fbf55ceb5d882d","revisions":{"a545fb473d7cd252cd3ce7fbe573faa81eeffedc":{"kind":"REWORK","_number":1,"created":"2021-05-10 12:34:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/36/790436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/790436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/790436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/790436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/790436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/790436/1"}}},"commit":{"parents":[{"commit":"bb1b8c37d4336c136fbe33d262006771debfa100","subject":"Merge \"Fix Mistral source images to respect upper-constraints\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb1b8c37d4336c136fbe33d262006771debfa100"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-04 09:30:28.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-10 12:34:33.000000000","tz":0},"subject":"CI: drop bashism from template overrides","message":"CI: drop bashism from template overrides\n\n[[ is bash only\n\nIn many distributions /bin/sh \u003d\u003d bash but not in Debian and derived\nwhere it is dash (very simple POSIX shell).\n\nChange-Id: I2b084ea78b236623b174473f411bae04e624a3e9\n(cherry picked from commit f0c3567740be342355b39a7c7d3ca341d5a4ec0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a545fb473d7cd252cd3ce7fbe573faa81eeffedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a545fb473d7cd252cd3ce7fbe573faa81eeffedc"}]},"branch":"refs/heads/stable/train"},"326201932f93fc18371c2c5385fbf55ceb5d882d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-11 05:39:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/36/790436/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/790436/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/790436/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/790436/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/790436/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/790436/2"}}},"commit":{"parents":[{"commit":"5687029e38aec76bff18c27e2421825ac431ad30","subject":"Merge \"[CI] Trust only infra mirrors and fix APT mirrors\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5687029e38aec76bff18c27e2421825ac431ad30"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-04 09:30:28.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-11 05:39:20.000000000","tz":0},"subject":"CI: drop bashism from template overrides","message":"CI: drop bashism from template overrides\n\n[[ is bash only\n\nIn many distributions /bin/sh \u003d\u003d bash but not in Debian and derived\nwhere it is dash (very simple POSIX shell).\n\nChange-Id: I2b084ea78b236623b174473f411bae04e624a3e9\n(cherry picked from commit f0c3567740be342355b39a7c7d3ca341d5a4ec0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/326201932f93fc18371c2c5385fbf55ceb5d882d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/326201932f93fc18371c2c5385fbf55ceb5d882d"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
