)]}'
{"id":"openstack%2Fkolla~791744","triplet_id":"openstack%2Fkolla~master~I7ed0202dfa644f7e85161b2ebaa999069fc1dcc2","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I7ed0202dfa644f7e85161b2ebaa999069fc1dcc2","subject":"ubuntu: unpin OVS","status":"MERGED","created":"2021-05-17 12:22:58.000000000","updated":"2021-05-27 10:39:02.000000000","submitted":"2021-05-27 10:37:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791744","meta_rev_id":"fb203083ad3e3b774eb7359a39a9746d924120e6","_number":791744,"virtual_id_number":791744,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 10:37:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2021-05-24 16:13:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2021-05-25 07:58:52.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-05-25 07:58:52.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-05-24 16:13:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-05-24 16:13:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 13:31:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 16:06:35.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-05-24 16:13:20.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-05-25 07:58:52.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-05-25 09:37:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"}],"messages":[{"id":"b1db01a61d8fc5b16e0c96c89c06d901428958b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-17 12:22:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77d2f698b812eba03d02a1fc3e8ff47b5a1a204f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 13:31:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bcdf8b316799444e9e49c00c522288df : SUCCESS in 21m 01s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/46bc87a377954f4f97ab47af8619b0cd : SUCCESS in 30m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b4d85c172b2c4b44937b6d7b01a07bfc : SUCCESS in 1h 00m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1d405231c8f0454cb52ed2591583f3a4 : SUCCESS in 58m 54s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5b2d4cebb6784bd3b84f10cdfab6ffd5 : SUCCESS in 58m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e9b0459ed6c143b5ac2740e4048dffb9 : SUCCESS in 41m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/350728cee2eb45458d412bdd09ee9ef7 : FAILURE in 18m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6cd75279491a4cdbaf1e42cde97136f6 : SUCCESS in 57m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c7284c822664048b119159e4cdc8078 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da8739083d884c22bfe9275f7442bc7b : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f561b59dc234ed49912999ad696dd54 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3b4b1d5bacd4bb7afd1f34128a2ff60 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d123e87f578443c996677344ed990be3 : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/19d97597dfa74f99a146e28a38807119 : SUCCESS in 21m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/64354bd726524a22b103343610cf4bf5 : SUCCESS in 29m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc832d1f112c46f3908333835b16a143 : SUCCESS in 46m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abeb34b9041c4e60aca69e0d4e6cad2f : SUCCESS in 50m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8773b5e44a8d4fa88a76cbc9898a719d : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":1},{"id":"23c9bbbec3fd12738dc5c132a285388d8b1346cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 18:33:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a408f818aaf64631bca494b094c94764 : TIMED_OUT in 6h 01m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/de749908a8204066b3b4c6007035d5f5 : SUCCESS in 1h 51m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80ceb27768dc403694c029cc830ef104 : SUCCESS in 2h 40m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4916bbee18ed46adbb9e71a48fbd75e2 : TIMED_OUT in 3h 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14047e779b14a8b45fc4d32089c7fd86e2f9ac5e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-17 19:44:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d3967e5459e0c9bb549988a2d84ca8cdb5cc8005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 21:08:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b0329395f7f342f1adaff5576dadb73b : SUCCESS in 20m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bcb9dd807770423c83e50b6bbe8eedaf : SUCCESS in 34m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1ac02cd132164aababf33cbcbe0f4350 : SUCCESS in 1h 03m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/760e2df180f747f194729c96c0c2e08d : SUCCESS in 1h 01m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/43b214055abf4d09857fe11c33cddce1 : SUCCESS in 1h 20m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a0ea35bae6f548b08f9eda11d2ac37a4 : SUCCESS in 36m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8ac88d0f77ed4692977334005e837bc7 : FAILURE in 19m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/378e3255ac1141b79be3bb1c5f267f27 : SUCCESS in 57m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/164f04d9fd274400b0919806de12faa1 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2de6659a4c81476eb0283d34b865e71c : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b80bb374092d4458a0a73bb0e9d64aad : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be186b609e0f4b16a336c393cde3db03 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d02a59287245db8e23ab9be8b616fb : SUCCESS in 6m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8c20809664ba48fd850b5bef294de574 : SUCCESS in 21m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac553743169d4c9399c3770bf6dfaeb3 : SUCCESS in 29m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/565f362cc1214e0db7b9096fbe417661 : SUCCESS in 53m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b55b489465584a91a34199d2a2fcaef1 : SUCCESS in 50m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/50d55c317c1c41479cc3b27297133302 : FAILURE in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9576e0c830c20cebb21bd8ae87ace50bc19592af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 21:12:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dfe6d6ffe29948ca96ee29e28188876f : SUCCESS in 1h 25m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6202dbf62b5f4612ab7f53b2137de92c : SUCCESS in 43m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/494520afcef74c32a8a7069e74476999 : SUCCESS in 1h 22m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b31ccb6e4bc4173933f685baa018516 : SUCCESS in 42m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d8a66610dd2cd0c20eeda99e5a0c6e6ca147b42","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-20 16:06:35.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"b9c4428e19d892a8926aa31e48ae09b60f2d25ab","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-20 16:06:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4768e019955896d4b98d6c45ae258b86655ecb7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 17:39:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/398f1069a4654e758c2b287a0725ce22 : SUCCESS in 23m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d0221d1c2fb84be68d8c825e76373e37 : SUCCESS in 30m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/abf26d9327dd4b0780bd8afb511a7397 : SUCCESS in 57m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/112b6a4f5d40493abb3f9f3d9337fdb9 : SUCCESS in 53m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9c0a543ebcb4418e8a9c8cc3001e8633 : SUCCESS in 1h 05m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6717d84c443d4f618116fc3d7d323a3c : SUCCESS in 36m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/18dbb4cf3a4349a483b733ab042bc64c : FAILURE in 18m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/050007f6f69541dfb1d32fa4f275fd05 : SUCCESS in 56m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d95c9b96ee1463b9fb9364426d5dbee : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ba101d7de5b4814ae468593915084ce : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b53830071c74be8aa997c19ca388891 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86af0c621f0640009ad64c48504017f5 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2066c425aa24429a61a6cc4d48e530a : SUCCESS in 7m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/32088e26201c4b449f0a807c48a86feb : SUCCESS in 23m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/22c668db55554c10b2beecb3b142ecd3 : SUCCESS in 41m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c85bcd2e0994ed696f4e147b4fc1911 : SUCCESS in 51m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b909dbdb73524de784da9fe29c654a2d : SUCCESS in 48m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/261122cb8a9d422b99bbc8a21c088b9d : FAILURE in 1h 07m 12s","accounts_in_message":[],"_revision_number":1},{"id":"fec0cb7973ac02e18667edbfd7e856786800ec19","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 18:08:58.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/db1f9a9c75ee4adf85a11d050231a9c3 : SUCCESS in 36m 10s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/395d22343de940108f7aff84e1092aab : SUCCESS in 27m 20s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/048a2c56d87745479ed98a2ea69b24d0 : SUCCESS in 1h 06m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b145a152654b447486e028e4c38d6d8b : SUCCESS in 1h 24m 51s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/0ad594e244494f85af1708375ee049af : SUCCESS in 1h 19m 28s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/cff22a18a4cf486fb4bdd07a90c1cdf7 : SUCCESS in 1h 27m 36s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/e0ebd95e925148b185e98245d4447306 : SUCCESS in 1h 03m 37s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/192df58d64c84bd7a23d2a6538f1497e : SUCCESS in 1h 25m 41s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/8fd0501be76d452b86c97048d37d8919 : SUCCESS in 35m 27s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/79a2e5546009420f9eb9908b180bd9fa : SUCCESS in 1h 19m 27s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/844f8ca903e74855add035ed1c7248db : FAILURE in 1h 45m 34s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/5941d2fec94340ddb6e3568e0ad16cac : SUCCESS in 32m 40s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/cac624ef1d8b4326a404f5e0268876a1 : SUCCESS in 32m 29s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/94756aa58bbd49ed910690ac48ecb021 : FAILURE in 40m 58s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/172731cf5c3441eda664245593aa06dc : SUCCESS in 1h 00m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/09f6421700914f339493962cedefe259 : FAILURE in 15m 29s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d9a7cdde3f3a4817a8d884dcd91dc070 : SUCCESS in 1h 06m 39s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/4d02481d8ed64cafba9abd15de20455a : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/d8a03c02a7514338b50d9b45400c2138 : SUCCESS in 1h 18m 40s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/d36c2f85f0354f14a50fa503f3302f61 : SUCCESS in 29m 36s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/9bf3171ae9ce4399b186e57870968a1b : SUCCESS in 53m 50s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/098754e868f04ebc87c425aaa0bd3f00 : SUCCESS in 48m 42s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/c6379736b0434af2a6c688fc8eb0ebf7 : SUCCESS in 1h 01m 19s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/a50742985ed34f3babbbbbdf933c38bf : SUCCESS in 30m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aefe41221f1f93106a5b45c3af31e3ff3191b11b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 18:28:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/65fb0f1dc05b4aa89342746cea417df1 : SUCCESS in 2h 19m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f897c0e1176342149e65fde0e56c0536 : SUCCESS in 1h 01m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/65a3a89af6f14b8cb408cb165081c267 : SUCCESS in 1h 28m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e0b0bc367e04d468e5931ed76788f3e : SUCCESS in 1h 45m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b011995dccb9b7089224cef721668be0c85a7bde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-24 11:10:02.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8bbf5653cbd99744a50dd403fb43ae23e8b7e3d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 13:10:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e17017038fd4e3983c39d0f11e12151 : SUCCESS in 1h 20m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e077347065cb45a2a886e994e31edec7 : SUCCESS in 1h 20m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5f6e16f579d34a34aaf64298a9895105 : SUCCESS in 1h 52m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2ab8913764f426b9889407fb2c0d617 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1d81f6c97acb46b8b239edbe3f8bf226 : SUCCESS in 6m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7e7050ef0fda47e2a735dea9a116d45b : SUCCESS in 1h 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9905ef0bec7f4e50e9ed327e319a37ff864b9fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 14:26:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c542fa61523c411c919701758926263b : SUCCESS in 23m 21s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/407e428918944eaf85457715aded079b : SUCCESS in 39m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/85640e23c5ac4d5e9af75ab486f14305 : SUCCESS in 48m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b6d3669e71684c208d8af8b664e79ef8 : SUCCESS in 51m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1ac16408bb184096a911760139b51fbd : SUCCESS in 1h 04m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a494a6b207e64be9bd7dd846ef5f2a8e : SUCCESS in 31m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/21e5dff50b64459d9e9c2761379e07a5 : SUCCESS in 22m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ecddc03e11de43bc8e229b940460c342 : TIMED_OUT in 2h 01m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7905ff91c5624783bd1d89f1c5345514 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f62f9d3361c64d45a6f85ca2895325c2 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5bc2f09f2ff4928bdee87cda8fd686c : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06c298f4f20a48c8873c955434453ef6 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb2a017587cf460fb5bd0e75f3071a4d : SUCCESS in 6m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/68ead41143d24275bffdc0256d4c8f67 : TIMED_OUT in 3h 02m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c76af12543714874804ae89b80bb5220 : TIMED_OUT in 3h 14m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/894b175ea17e49f5987aa551eaebae70 : SUCCESS in 52m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce14b9f9d1044658a94693a8b286a968 : SUCCESS in 58m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/210bff578a264019bdd7c88fac4a2bd4 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":2},{"id":"1fdd40f73cafc560556e8034e9fbbd63f0ea68c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-24 14:39:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5e20222bf72037b05a8fe83fce5cbe4061a82007","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-24 14:40:03.000000000","message":"Patch Set 2:\n\ny so slow","accounts_in_message":[],"_revision_number":2},{"id":"3360093f21c7171bbd29eb9ca83000465203d6e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 14:40:43.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"d5d09567373786fc677f72bce849adf08fd1e4ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 14:40:52.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"52a5e30e2e21f0c26a68792777a1e96e5e8a4da5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-24 14:42:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"eb1000e49297aa7ddd09d1ad4b2d847add67258f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 16:06:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b1afb7b42f2148aca85e31685c16dda1 : SUCCESS in 26m 16s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7ddb7e6d4c334341891bcea52431e62e : SUCCESS in 36m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a7b8e4fc98bf420cb9c416e109f84724 : SUCCESS in 1h 06m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bcfa28b096f5436c9b91f1781c89adcd : SUCCESS in 55m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4ac21881f9084f9cb41a36cf2d16f7a7 : SUCCESS in 1h 20m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/27d8e35ad86548aea160b5b0d12b23c4 : SUCCESS in 31m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/da16da04f2a8474c9763eaf24f32dc92 : SUCCESS in 23m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9c0d22c0c51e4d87879aa48029b98caa : SUCCESS in 48m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efd498c6ba794218a7efaf29bf341959 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c82336c5ba404a9cbb0b501f62dd97e0 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf12d2c35c2a4439bfca89813f028d5b : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba44efd575704bf293c640c54c61aae7 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7859c000d4f141fa9747c243607ea59e : SUCCESS in 8m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ee36aaa1ce84717a2949fb59436aefc : SUCCESS in 26m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5abd76f8196240978c692631094d8cec : SUCCESS in 47m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e537fd52aa3e46b096b40933da1c16b6 : SUCCESS in 44m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e174408829c4207ada43951d0fc902e : SUCCESS in 53m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f848f59a0d12406d9aecea56771898f9 : SUCCESS in 1h 20m 43s","accounts_in_message":[],"_revision_number":2},{"id":"27b0c8367252b949fb00a6d6cef9968c367ef8d7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-24 16:13:20.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"16944070820c8e5ad165096d2ffecabc8dc35ad8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 16:45:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ad7bbed969942d99008280d023a0c01 : SUCCESS in 59m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0026c18f627d4d54916f79f3448f5ed9 : SUCCESS in 53m 15s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0afad0108c9c4dddb6ab62cee92ee7a3 : SUCCESS in 2h 01m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/692140e9cbaf401b996ce857fba80512 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a04ab3a661524024955905ab6f7f7247 : SUCCESS in 7m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c4b05939f5a444919419ec2ebd42523e : SUCCESS in 46m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6be92c23ff59d63079a0a23185fc543cbf5b8441","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-05-25 07:58:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f527a83abfc774492b57494d9e073ad806847cf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 07:59:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"69a64fee597804abecd79ec38e09a91be1b58bbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 09:12:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/01de1f2e30d84b7f9c063b2d12542cd1 : SUCCESS in 1h 11m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/40ba78163e3d4db0894c94b7ac91f7de : FAILURE in 47m 56s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7592e050ddca42a8b6f7588219b8d5d4 : FAILURE in 25m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/995da93156904bb7a2aa915d2bfe3776 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2563b9f023db4cf7bf80c0d33ced660b : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a5e7b60392d4ead838bf6fee676b051 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d9b9612d31b49e788d2c5d1b553a75c : SUCCESS in 7m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/253c9ee23ceb4b55a6cd497d484a9dbe : SUCCESS in 41m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b30de5b76f24486abc3a7a21e2e90ee8 : FAILURE in 37m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ab32399f3dad44579b41abf703f443d0 : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":2},{"id":"ad62fbce889dfafbba00a8cecd50fa48f409d0dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-25 09:37:22.000000000","message":"Patch Set 2:\n\nhttps://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_40b/791744/2/gate/kolla-ansible-centos8s-source/40ba781/primary/logs/kolla/nova/nova-compute.txt\n\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899] Traceback (most recent call last):\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/nova/compute/manager.py\", line 2370, in _build_and_run_instance\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     accel_info\u003daccel_info)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/nova/virt/libvirt/driver.py\", line 4152, in spawn\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     cleanup_instance_disks\u003dcreated_disks)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/nova/virt/libvirt/driver.py\", line 7220, in _create_guest_with_network\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     cleanup_instance_disks\u003dcleanup_instance_disks)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/oslo_utils/excutils.py\", line 227, in __exit__\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     self.force_reraise()\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/oslo_utils/excutils.py\", line 200, in force_reraise\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     raise self.value\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/nova/virt/libvirt/driver.py\", line 7188, in _create_guest_with_network\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     post_xml_callback\u003dpost_xml_callback)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/nova/virt/libvirt/driver.py\", line 7120, in _create_guest\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     guest \u003d libvirt_guest.Guest.create(xml, self._host)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/nova/virt/libvirt/guest.py\", line 149, in create\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     encodeutils.safe_decode(xml))\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/oslo_utils/excutils.py\", line 227, in __exit__\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     self.force_reraise()\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/oslo_utils/excutils.py\", line 200, in force_reraise\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     raise self.value\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/nova/virt/libvirt/guest.py\", line 145, in create\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     guest \u003d host.write_instance_config(xml)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/nova/virt/libvirt/host.py\", line 1203, in write_instance_config\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     domain \u003d self.get_connection().defineXML(xml)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/eventlet/tpool.py\", line 190, in doit\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     result \u003d proxy_call(self._autowrap, f, *args, **kwargs)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/eventlet/tpool.py\", line 148, in proxy_call\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     rv \u003d execute(f, *args, **kwargs)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/eventlet/tpool.py\", line 129, in execute\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     six.reraise(c, e, tb)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/usr/lib/python3.6/site-packages/six.py\", line 703, in reraise\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     raise value\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/var/lib/kolla/venv/lib/python3.6/site-packages/eventlet/tpool.py\", line 83, in tworker\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     rv \u003d meth(*args, **kwargs)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]   File \"/usr/lib64/python3.6/site-packages/libvirt.py\", line 4074, in defineXML\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]     if ret is None:raise libvirtError(\u0027virDomainDefineXML() failed\u0027, conn\u003dself)\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899] libvirt.libvirtError: internal error: unknown feature amd-sev-es\n2021-05-25 08:45:46.956 8 ERROR nova.compute.manager [instance: c802f1d5-8be9-434e-ab84-1ead86339899]","accounts_in_message":[],"_revision_number":2},{"id":"91685287b8d57d7336bafbbd7856f39c250fad01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-25 09:37:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"04319da333cea6bd80598cd9cc9c6b7ea62f6738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 11:33:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c9f2e5f693a64b9c857b098331f962a9 : SUCCESS in 23m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/eb430bdf18ef42be8c2c20cfa0b46fcc : SUCCESS in 34m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bfd7e8d6641443099ac979dcdd119da6 : FAILURE in 37m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b689dc9286bb4450a0dcf1e8217c3cba : FAILURE in 46m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2ea992c9ddd94a839433d197714a5fa1 : FAILURE in 1h 14m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7b5b4d934c584632b9debb1212fa0ec6 : SUCCESS in 42m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dd1e1fe9c8914fffb69e179393798079 : SUCCESS in 1h 51m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dd74b9a95fea4bb0a599d762787211b6 : SUCCESS in 1h 11m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f07fc930bb4c36bd3c8d25364aab36 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4da960c897c54e969f98c9c843feb1d6 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af3ae359a5ff44df9d3c3db6bd53bcdf : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c262598dd09b49e893990c3bf7415260 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/838ad0304eef466b8d8c8886d4ded13a : SUCCESS in 5m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d55c6328e646441a9a940f29abe11932 : SUCCESS in 20m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/44cb8afaa464469ea74ffaf2b4971f26 : SUCCESS in 38m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/838673fe6c5a465498e0446e0b9073f3 : SUCCESS in 50m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2a210e1f7e9643918d0c4ab292312ac0 : SUCCESS in 40m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c9e8f76bb68b4d3fa76cfd5615c580a4 : SUCCESS in 54m 08s","accounts_in_message":[],"_revision_number":2},{"id":"8c0517761e82261434511fd587a882cfbffa60f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 12:25:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3884ea3383c24a589da30930699976ab : SUCCESS in 1h 49m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/29bedd6eece9499b8efea0a8451db459 : SUCCESS in 45m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b9bacf971c14e98bcf7bc9a8e28a0a3 : SUCCESS in 2h 45m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b570d33a6ae245a8a139ba217d24b1e1 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e9cc38e33aa844cd837adae14d77d77b : SUCCESS in 5m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b8821c9571c4cb59e8714e19d6e66f2 : SUCCESS in 44m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"829a9666ad76d9057a7f480e8be39336e797cf00","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-27 08:19:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9101d7b22f3d275d0f4bb300da7c564431df172b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 09:23:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0271aeb258d2469a86d62bf4a86699ee : SUCCESS in 20m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1c680e9e8d2c457789c49ace81fb82c9 : SUCCESS in 36m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5cf4ae5485a04da2b48ce716971f67fd : SUCCESS in 52m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d5f51c3052a240c6a1a19efed5a24f67 : SUCCESS in 56m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/756b61172a11469fb16603d56d7d4071 : SUCCESS in 58m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/189f504d323f4f67b2f32f1cefe6b4b3 : SUCCESS in 29m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/36b09c624aef4710b1b590cf24aaeb8e : SUCCESS in 17m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f2c3a0ffd8684aea8fc956857ef3ff8b : SUCCESS in 54m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b449e38cef488287db65143668fd1d : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eec3169aca6f463795acddbb06f590fd : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2563aba6d80141eca481274a02019416 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1757abf639e04ee295f76ce2421c3fe7 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38b1ca8672fe4733bcc469a5e6beceab : SUCCESS in 8m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf61d8dbdd3c4d4bb1aeefd01a853192 : SUCCESS in 16m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d745e8a07d21483790ec7f6fde38dcb0 : SUCCESS in 29m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8171e0c7c10a4338811487620bdfd7c6 : SUCCESS in 50m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/07089f85bbc646ecbf8ce4f64927fa27 : SUCCESS in 47m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/617d5d08a48e40aaa0faf7adc4491e81 : SUCCESS in 1h 02m 34s","accounts_in_message":[],"_revision_number":2},{"id":"fc3e3c458fe3ab65e0576ae210216ca36c0f8cb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 09:23:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5f6407e819e4fc4ba29bd01d978e789baf97292f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:08:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5ace83a46f447d892de637c15ab5efc : SUCCESS in 57m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/61ff45d678c84c09aaca698e635dcb34 : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b3109c0205894e72b95751c88b55c49a : SUCCESS in 1h 44m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99a2090510ff4d55abd694ad2ffaa7fe : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f58813348e9428592cbf3b19aeddd2b : SUCCESS in 5m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/73cb408e624e4a7cb3a3ade9a556914a : SUCCESS in 51m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b7e60373a933731447aa79e4f7ef5fcbbb5926f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:37:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4b2ce651fe284a939150ba96e5b75b54 : SUCCESS in 35m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fef35a36b2594b8fa1cb312537767640 : SUCCESS in 56m 55s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/077ba3588ca048068e74dae2dc098731 : SUCCESS in 1h 08m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b68912b831e24e1390b57e9e9903cefc : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/638dacd9f5034c4eafb77e96500ef1c8 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/947c3c4caf1e43799cb3fc54dc5103b0 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0b51bcc382c4b09870ebb4a77765d9b : SUCCESS in 9m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8503af3a621f47349d5e28cab865a487 : SUCCESS in 26m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/46c399c5adb14219b178a35a59cf1297 : SUCCESS in 47m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/702f078a1e314556b3193d8369ebf3bc : SUCCESS in 1h 13m 20s","accounts_in_message":[],"_revision_number":2},{"id":"210b67ff0546720fa3cefd620446bb473b53a1be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:37:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fb203083ad3e3b774eb7359a39a9746d924120e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:39:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21816e15af234f2c8f6435bb640829bc : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6035f57a353cfd05f1017b6763e60f0632e65c2e","revisions":{"0156979b7b6172bf35771525b2558be420e674a4":{"kind":"REWORK","_number":1,"created":"2021-05-17 12:22:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/791744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/791744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/791744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/791744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/791744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/791744/1"}}},"commit":{"parents":[{"commit":"0506ede84f97159a8156ad51aea395d0868e6141","subject":"Drop ppc64le support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0506ede84f97159a8156ad51aea395d0868e6141"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-17 12:22:35.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-17 12:22:55.000000000","tz":120},"subject":"WIP: ubuntu: unpin OVS","message":"WIP: ubuntu: unpin OVS\n\nChange-Id: I7ed0202dfa644f7e85161b2ebaa999069fc1dcc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0156979b7b6172bf35771525b2558be420e674a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0156979b7b6172bf35771525b2558be420e674a4"}]},"branch":"refs/heads/master"},"6035f57a353cfd05f1017b6763e60f0632e65c2e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-24 11:10:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/791744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/791744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/791744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/791744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/791744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/791744/2"}}},"commit":{"parents":[{"commit":"0506ede84f97159a8156ad51aea395d0868e6141","subject":"Drop ppc64le support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0506ede84f97159a8156ad51aea395d0868e6141"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-17 12:22:35.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-24 11:09:58.000000000","tz":0},"subject":"ubuntu: unpin OVS","message":"ubuntu: unpin OVS\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/792352\n\nChange-Id: I7ed0202dfa644f7e85161b2ebaa999069fc1dcc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6035f57a353cfd05f1017b6763e60f0632e65c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6035f57a353cfd05f1017b6763e60f0632e65c2e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
