)]}'
{"id":"openstack%2Fkolla~792468","triplet_id":"openstack%2Fkolla~master~Ibeed1e13a38e722ba483a6d00dbc905f8c585a7e","project":"openstack/kolla","branch":"master","topic":"dnm-ubuntu","hashtags":[],"change_id":"Ibeed1e13a38e722ba483a6d00dbc905f8c585a7e","subject":"[DNM] Test ubuntu upgrade","status":"ABANDONED","created":"2021-05-20 18:19:18.000000000","updated":"2021-05-21 17:00:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e49c8571132729909a6003a8aa3070dec000b32a","_number":792468,"virtual_id_number":792468,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-21 12:05:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 20:37:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"350986361fa1a90b60c96a5fcda729e62ad36772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-20 18:19:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe6a1eb187d6e8792e3aa07f5e89d4f160cb8fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 20:37:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1bf1e7a1dd024eae94fac010be4bf7e5 : FAILURE in 2h 15m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c84fd1fabf7db9bee41796e957805d6e3c261055","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-21 08:14:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6833c951bb4bff4e9172b0f47c5d5a42c93a34af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 09:53:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/83a9366e1eb8482a96112c9b447eb550 : SUCCESS in 1h 19m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3126d4d0638bfe5419766fc555ea2c4f660b4267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-21 10:56:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a286cea1967f1b9a2ff4b149c5b167569c9657cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:05:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/97e54e0b683142d08a7d06a0ab782f7d : FAILURE in 57m 45s","accounts_in_message":[],"_revision_number":2},{"id":"7df991e47c7b3374dcb9808658f13a0f1d777f83","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-21 12:09:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"98cc82de63513c5d935503e9fb352a66d52e309c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:51:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2f6c3e4017db4db4bc333ae9837dd59a : FAILURE in 40m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6067aaf5ede4a82c15a157acc682ad1a6d6e7809","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-21 15:24:56.000000000","message":"Patch Set 2:\n\ncryptography 3.1 depends on cffi!\u003d1.11.3 and \u003e\u003d1.8\n    The user requested (constraint) cffi\u003d\u003d\u003d1.14.2\n\nThose look compatible to me. Are we getting \"even better\" (TM) resolver again?","accounts_in_message":[],"_revision_number":2},{"id":"a6c36099cc9d71f68d2f4b113761825cc6763774","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-21 15:25:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"abc19780800e5552f04cf77ed9cc9104cce2d49a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 16:59:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ffdf62a41b4741459d1486efb48e9d1d : FAILURE in 1h 29m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e49c8571132729909a6003a8aa3070dec000b32a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-21 17:00:44.000000000","message":"Abandoned\n\nfinally the failure I was looking forward to...","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"899fda1f7727113cbe922cd72cfec0f757cf210a","revisions":{"f41e4c36a5064d7c47382ba1645b4d68d4835733":{"kind":"REWORK","_number":1,"created":"2021-05-20 18:19:18.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/68/792468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/792468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/792468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/792468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/792468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/792468/1"}}},"commit":{"parents":[{"commit":"0156979b7b6172bf35771525b2558be420e674a4","subject":"WIP: ubuntu: unpin OVS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0156979b7b6172bf35771525b2558be420e674a4"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-20 18:19:03.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-20 18:19:03.000000000","tz":0},"subject":"[DNM] Test ubuntu upgrade","message":"[DNM] Test ubuntu upgrade\n\nChange-Id: Ibeed1e13a38e722ba483a6d00dbc905f8c585a7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f41e4c36a5064d7c47382ba1645b4d68d4835733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f41e4c36a5064d7c47382ba1645b4d68d4835733"}]},"branch":"refs/heads/master"},"899fda1f7727113cbe922cd72cfec0f757cf210a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-21 10:56:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/68/792468/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/792468/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/792468/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/792468/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/792468/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/792468/2"}}},"commit":{"parents":[{"commit":"0156979b7b6172bf35771525b2558be420e674a4","subject":"WIP: ubuntu: unpin OVS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0156979b7b6172bf35771525b2558be420e674a4"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-20 18:19:03.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-05-21 10:46:53.000000000","tz":0},"subject":"[DNM] Test ubuntu upgrade","message":"[DNM] Test ubuntu upgrade\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/792563\nChange-Id: Ibeed1e13a38e722ba483a6d00dbc905f8c585a7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/899fda1f7727113cbe922cd72cfec0f757cf210a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/899fda1f7727113cbe922cd72cfec0f757cf210a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
