)]}'
{"id":"openstack%2Fkolla~792516","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I483db6d6eddd84b376251b38509e74d4a5c2a10b","project":"openstack/kolla","branch":"stable/wallaby","hashtags":[],"change_id":"I483db6d6eddd84b376251b38509e74d4a5c2a10b","subject":"Revert \"CI: add monasca-grafana to allowed-to-fail images\"","status":"MERGED","created":"2021-05-21 08:11:39.000000000","updated":"2021-05-21 12:16:39.000000000","submitted":"2021-05-21 12:14:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792516","cherry_pick_of_change":792288,"cherry_pick_of_patch_set":2,"meta_rev_id":"588d16992707b5fd743b7bdbd61a8edfc60d5c8d","_number":792516,"virtual_id_number":792516,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-21 12:14:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-05-21 08:13:10.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-05-21 08:15:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-05-21 08:15:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 08:11:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-05-21 08:15:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-05-21 09:58:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-21 10:43:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45d9d8eb31693f9e26faf6755ab857ca10195213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-21 08:11:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0d7c76e2e14a960d15f862a8e2606de40ac156b0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-21 08:13:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5ed9c11d92513bb6f3eda4ddd2d984b7492c6fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-21 08:15:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"13f9d916953b8843dc8e06fd091416797b1d0456","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 09:58:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b032754b7834c0bad4efc4055bae869 : SUCCESS in 1h 17m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/14397b1ea7f04ea4864c8e235772c03d : SUCCESS in 1h 16m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/488dc224944a4f8783504583b32e4f11 : SUCCESS in 1h 43m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c87eadb038814a03b494e1804006e06e : SUCCESS in 1h 05m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cc2dad2630f632709e94d7147e296ee57cf50a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 10:43:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e44464bca7804dccb73c13d9460e02c9 : SUCCESS in 27m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c294c6258d204735b971680b95a0ea90 : SUCCESS in 2h 27m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/44066eacb6e24825aa1f65e3dfc9577d : SUCCESS in 1h 13m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/30b91e3cb90743ffabcb7bde1c65f52d : SUCCESS in 56m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a8089ea82c4447f1ba17c7a830e0093a : SUCCESS in 1h 06m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dcac5a12fdaf4c848bdb8d36ce352276 : SUCCESS in 41m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5753c8f2e8524b7eb1ca709e0b3052e5 : SUCCESS in 20m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3abb4e4edb75478ba507cdbeb72852a7 : SUCCESS in 59m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3214bf47cb604b859c8afc4be2c4d170 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeaa5c20f0a949c59645628ec641820f : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03dd460858ad4830af00829cd5c875dd : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fd3ad50c1704aeabb82b20ee09716c6 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7cd9eac8f04d50ba0fae98f152f024 : SUCCESS in 14m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b774fb54a83e4071a184f15a1a1a78cd : SUCCESS in 1h 57m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4237bbde38294c129409a4f320b18c1e : SUCCESS in 42m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/02c288c47c274e8d8cec8e2b7217144d : SUCCESS in 56m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1cb3f0058b24811891c41cc09232687 : SUCCESS in 47m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b5302177d97e486287aa77afd627f78e : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2aa8366414da51b3f859f957891ff76a34530ae6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 10:44:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca2715d93e1b9a6094a4011af454b1bffbd47255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:14:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3bd1efcc6d4647549bf31cc10471ab96 : SUCCESS in 43m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/65a066e7426d422f94a1df68e435d5e7 : SUCCESS in 1h 07m 31s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8c25e64c9efd416d91787240f648209a : SUCCESS in 1h 24m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84dede697ba3444abcc5400cc958af58 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a966eeac91f847c9bb41550caf1addf5 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2eaa1f0207d431c849535a0b9d81dda : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc821880f68e406ca84fc8397d75fe57 : SUCCESS in 6m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5cf7f5567e29477fa1a3839ddffb6e99 : SUCCESS in 36m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0dd253a26cf543368f98b64fd6e3c979 : SUCCESS in 57m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/afee3963fd5b4911a340ac249a4c9148 : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ecc3c54a005c8004b40cf0e6132e79f49502cd46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:14:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"588d16992707b5fd743b7bdbd61a8edfc60d5c8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:16:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b2544c827bf4c63afd900aa9a1ec813 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe1c8e882c0ac4ae4640b0e80721a758fa8f6a02","revisions":{"fe1c8e882c0ac4ae4640b0e80721a758fa8f6a02":{"kind":"REWORK","_number":1,"created":"2021-05-21 08:11:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/792516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/792516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/792516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/792516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/792516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/792516/1"}}},"commit":{"parents":[{"commit":"66b5abf672ed0fb8646f0378bfdaba65ab16fdf0","subject":"Drop monasca-grafana image.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66b5abf672ed0fb8646f0378bfdaba65ab16fdf0"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-05-20 15:53:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-21 08:11:39.000000000","tz":0},"subject":"Revert \"CI: add monasca-grafana to allowed-to-fail images\"","message":"Revert \"CI: add monasca-grafana to allowed-to-fail images\"\n\nThis reverts commit 931d401887252d8830ebfb4d0b5d352f77094e9a.\n\nReason for revert: monasca-grafana is no more\n\nChange-Id: I483db6d6eddd84b376251b38509e74d4a5c2a10b\nDepends-on: https://review.opendev.org/c/openstack/kolla/+/784902\n(cherry picked from commit 1e7d1f3bd3368e6fa6514efef8d9612d8d69961e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe1c8e882c0ac4ae4640b0e80721a758fa8f6a02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe1c8e882c0ac4ae4640b0e80721a758fa8f6a02"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
