)]}'
{"id":"openstack%2Fkolla~792751","triplet_id":"openstack%2Fkolla~master~If8f0730031435dda4235b7f2d2c23e5f5f767f87","project":"openstack/kolla","branch":"master","topic":"bug/1808805","hashtags":[],"change_id":"If8f0730031435dda4235b7f2d2c23e5f5f767f87","subject":"monasca-thresh: Allow topology check and removal in storm","status":"MERGED","created":"2021-05-24 03:50:00.000000000","updated":"2021-08-10 13:04:50.000000000","submitted":"2021-08-10 10:48:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792751-bug/1808805","meta_rev_id":"35ae3526afd202cd9b28432ede41abc377ee229a","_number":792751,"virtual_id_number":792751,"owner":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 10:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2021-08-10 13:04:50.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-08-10 13:04:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-08-10 13:04:50.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-08-10 09:33:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-10 13:04:50.000000000","post_submit":true,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 04:59:34.000000000","updated_by":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2021-05-24 05:01:22.000000000","updated_by":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-24 05:01:22.000000000","updated_by":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-06-30 16:16:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-07-22 11:18:03.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2021-07-29 10:05:05.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-08-10 09:33:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"c6fad2866d54342f13ae46cfe7c55ce29396191d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 03:50:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bab8b16980a6b04fe68dd9849f7bcc275f1d8202","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 04:59:34.000000000","message":"Assignee added: Doug Szumski \u003cdoug@stackhpc.com\u003e","accounts_in_message":[],"_revision_number":1},{"id":"e1d616fb1c1680a9bd013ae859c348d795fd70a9","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 05:01:22.000000000","message":"Patch Set 1: Backport-Candidate+1\n\nMoved monasca-thresh topology script to Kolla (must be enabled to do anything).","accounts_in_message":[],"_revision_number":1},{"id":"c54bf5516554d67094de1a5ad0ffd1a22200926e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 05:05:53.000000000","message":"Patch Set 1: Verified-1\n\n(26 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2fbeaeae129e49e9883c9dd1faea41c5 : SUCCESS in 26m 08s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2c34f0fc7ab24aa0be947350ba127b63 : FAILURE in 1h 07m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b603f09e86ca4616b200d566aa114ca4 : SUCCESS in 1h 06m 57s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9fe594bea6844ecd8a742fd73b9f8434 : SUCCESS in 1h 04m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9b3642de38804fe6996feab2b0b309e6 : SUCCESS in 1h 14m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a298901a76e04212a9a575e05d9b030f : FAILURE in 26m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fde01adfffaa4b03a784b7248c5ce94d : SUCCESS in 17m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/50db4ea48f314a0e8a54bb65f2a590ce : FAILURE in 29m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77248c5986c14e59bce59c93eea287d7 : FAILURE in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84812b3bfabd4ff7af85db8daa301640 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/775286575f39492288cc769272be26ce : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d482d420b4094334a8ac3daadc834e78 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b24a20128c34b70b0893f007792fbd4 : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de3814597b61409799ef6a09bbb2f77e : SUCCESS in 21m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e00c325d3dda44a7b201ef2f3b1e02d4 : SUCCESS in 38m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/34d64628c78e4f21ad1d27239c4a15c7 : SUCCESS in 55m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc1e1656e2e64bf388c7efb68aeff8a3 : SUCCESS in 50m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f9ed2bec9d50418dbf9d53ff0272cf06 : SUCCESS in 47m 14s","accounts_in_message":[],"_revision_number":1},{"id":"349bca9db64e9540646b072e3150259c4dad5082","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 05:59:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0bf2f123904e4c27957a1185a20e587e : SUCCESS in 1h 38m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/21137b2fc20142ac90d950c704c76d45 : FAILURE in 36m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/75f3c5b5bd274df6b27d9252eaaeb7b6 : SUCCESS in 2h 07m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/603b9d2ee73846d69281aab0c638d40e : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32fe2763a29940b1bebb251166f3002a : SUCCESS in 6m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/57ad975f778e41d39dd95b9e687deeb0 : SUCCESS in 1h 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0e98436c831b496fec615b148b9df0d302ac034","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 16:17:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c07ecbc581b07e37fb48fbbd4a68d797a11adec1","tag":"autogenerated:gerrit:deleteAssignee","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 16:19:09.000000000","message":"Assignee deleted: Doug Szumski \u003cdoug@stackhpc.com\u003e","accounts_in_message":[],"_revision_number":2},{"id":"3770bd532f1408b133fe4903bd616e7eab2da644","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 16:19:15.000000000","message":"Assignee added: Scott Shambarger \u003cscott-openstack@shambarger.net\u003e","accounts_in_message":[],"_revision_number":2},{"id":"86911ac83e75c5bfd00c54043eceaeba5db03763","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 16:20:30.000000000","message":"Patch Set 2: Backport-Candidate+1\n\nUpdated script indentation","accounts_in_message":[],"_revision_number":2},{"id":"5a64f9883d5600a0ed937afcc69597b8e02def5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 17:21:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/785a75f2154d4de3a0d8432ac4102a82 : SUCCESS in 20m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6f02256e04c544c2ace3406597f35e75 : SUCCESS in 33m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1d220a055a494e9287fe22023b50b480 : SUCCESS in 53m 57s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/88a7c3756cab4ac7b7620320a2577fae : SUCCESS in 1h 02m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/01a70ffb54bd45939b7afee3ef6f0947 : SUCCESS in 1h 00m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c3472f6884444acfa5acabeda47bc7df : SUCCESS in 36m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/610c365cc6304cde8287a9cf99f88e69 : SUCCESS in 17m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9dba68252f74518a75ee45281dde398 : SUCCESS in 51m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c747bab1194942c7b1434f04b9514aa6 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b54c93c256943a6a81527ccc2d77024 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/296d07b331074eb192e7d101984e2a6c : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0511ad3ca5524be0be6de73e5da2ef8f : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25ea15adfe594e4095d277f30a69aee0 : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fbe337fe136541bc8fa13de51b44ba35 : SUCCESS in 15m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2bfb993253f74b44825662d5aca16923 : SUCCESS in 28m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d92e097cbee8496597a09c63fd4a5c59 : SUCCESS in 53m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/837ea63062834dae9092bc51f7055ffd : SUCCESS in 45m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f0d224b02d0d495eacf205e4cb66133f : SUCCESS in 49m 14s","accounts_in_message":[],"_revision_number":2},{"id":"3f5bb3ccd3b6fc1b0a39324edadf06fa9b02554c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 18:07:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/199d4c9373114afa928c42b47ce05f50 : SUCCESS in 1h 47m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/93aad8270d524a62b3af41a893623f9a : SUCCESS in 42m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f4cba9de76884d5a938fb6c0339bdec3 : SUCCESS in 1h 24m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b89d752d89c74505af353a760a5bf287 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/232321ab553144b7ad1ef1845d6e311d : SUCCESS in 4m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/43b51cf2ef234ad685d42ae52a05a869 : SUCCESS in 1h 07m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56c190e7b6964bdc4ebca27ad8c19f07aafd66b3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-06-03 16:03:58.000000000","message":"Patch Set 2: Code-Review+1\n\nThis looks good to me (I\u0027ve also tested locally). Thanks Scott!\n\nOne last thing, please add a release note. It can just say that functionality has been added to manage Storm typologies via a bootstrap flag or something similar.","accounts_in_message":[],"_revision_number":2},{"id":"9454627f7369029b38d01b7e66a9de2727e4f864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-06-05 01:34:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8f0f9bbcbdfe2a0d686818207da7e4fbbe9fed2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 02:48:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/dfe47d6fdd9b40e99a891064b81b0c6b : SUCCESS in 25m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0eb226a87ec64cb3b38025c47b998e32 : SUCCESS in 37m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/49bc1d9526014390a46366266ed5a73f : SUCCESS in 1h 05m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4d1dfff9bf214eb39ae5eb17671cba68 : SUCCESS in 1h 02m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/55bb8724251c4b3d8612c9e87944f73f : SUCCESS in 1h 12m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0112beeb34c74e9795358b7fe6dc075d : SUCCESS in 34m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ac84ded743e040119680d24261f9f523 : SUCCESS in 19m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5c022dcbcf674e5f9d8007b009527a69 : SUCCESS in 1h 03m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbdcc8e8a4a4408e8f4e10a7af6b41f4 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d584b6c438641d698cdb2ef28f52b8d : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f47ed8005d847c097ca976be405b391 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff0c54d9475b45ed887d463c34728189 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a43155468054474dad808d78f7bacb19 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee6a582f02864048924eeab57b98cfb5 : SUCCESS in 4m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d12abbc2747247a28b51fdc431f1e791 : SUCCESS in 17m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4fe6d0b9f3b46048ea3dca133c2568a : SUCCESS in 31m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f2b5ff50a1c4850999fe18a3bcf484d : SUCCESS in 49m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e8bfd10b89e0458d904e601a8dab409d : SUCCESS in 49m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2433691fd66b44a5ba7d56a5c3352184 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":3},{"id":"ceb9aec31cd3802773cc0a073d1ee77bd6d9085d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 03:11:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/104b2b1f66614555876b063e99ea5795 : SUCCESS in 1h 29m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/66ee2f9c214c4f9aa8c018499beceabd : SUCCESS in 40m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c22e11d32c447828a622e924a4ea6aa : SUCCESS in 1h 29m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc0d8b79a8ac466097b3040f3aef18b3 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ac4109520964f989775073e547d9957 : SUCCESS in 6m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e5abb58e1a75492eac7d0022ec746cdd : SUCCESS in 1h 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"358557a5708f611b7f5435879f6bc54d285f9e33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-06-11 04:52:57.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"4a0ee287432429c6270b1d187520aa80cb4fe820","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-06-11 04:53:33.000000000","message":"Patch Set 4:\n\nAnything else needed for review?","accounts_in_message":[],"_revision_number":4},{"id":"64ee1771b6e111f77c97a67142560693c229c96e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 06:08:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8e1120647cf743a2a827ab572f92264d : SUCCESS in 28m 36s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2fb3a0c329eb4f0a874c2609ed60b802 : SUCCESS in 34m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/948e9b2a4d1f4bd2bc25f4270ae843bc : SUCCESS in 1h 02m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9a6d47f383934b75a8b380ae439d2606 : SUCCESS in 1h 06m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/de7cd5413f664b29a7ac4752c9181ef8 : SUCCESS in 1h 13m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4eab8f096aeb46b6904fc14a4353fa07 : SUCCESS in 35m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dfebde09ceb54d13a0a897489db3dfd9 : SUCCESS in 21m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/38cc1ba511f846fbb9d761133f36576d : SUCCESS in 55m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26589a5e3ab24dfa933740458e3536b6 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbf75e4016da45878c9c65237a0b1e2e : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2912ceaf9ac140a2a804d7fc364f25fc : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0d3cd4e85b14dffa29121058be018aa : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a3675685e8468f9c97ace1619217e7 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fe1d33fca6440c3973559bd18f5b2ca : SUCCESS in 4m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/035e07d3c018488caea602ea1ae4d0b1 : SUCCESS in 21m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6dde7f22943947f4a28d7f536e2e1493 : SUCCESS in 33m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/14c267e830b045f08ffa2110a8ae1556 : SUCCESS in 52m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/420824776489418995be69002695b51f : SUCCESS in 49m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4285e839e3ec4fb1a5fb4b0b8dda6bc6 : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":4},{"id":"09a3184183b2813265e2d9f673924d9d8465533f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 06:48:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/29066f7ace414d8e99be41fe993b8087 : SUCCESS in 1h 36m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f9d4354e0924736a7f5c6c70b8e4dbc : SUCCESS in 1h 08m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f8ba2d527514495bf0eed3113656c84 : SUCCESS in 1h 49m 24s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49393597bdfd4245a19558fced662adb : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0f44f1604b04fb191b758e3967fc639 : SUCCESS in 10m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8f851d84d31341d4b039b49d4af4113e : SUCCESS in 59m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7d61cd4fafa09799ab81971c4feb5b4221c6df4c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-06-28 09:38:09.000000000","message":"Patch Set 4: Code-Review+1\n\nI don\u0027t have +2 in Kolla, but this is good to go pending the patch in Kolla Ansible IMO.","accounts_in_message":[],"_revision_number":4},{"id":"3307c08b07552ff1543769e855cd71d62c6c93d9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-30 16:16:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"91f2b7e6aaf9f30f48529253477c20eada71ce45","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-07-12 17:29:21.000000000","message":"Patch Set 4:\n\nIs there some action I need to take to trigger a merge for this?","accounts_in_message":[],"_revision_number":4},{"id":"0308f1b4c28460cea5953819bb26e8669648fb21","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-07-19 08:17:17.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e Is there some action I need to take to trigger a merge for this?\n\nIt\u0027s worth raising at a weekly meeting (you can add it to the agenda), we just need another +2 which I can\u0027t give in Kolla.","accounts_in_message":[],"_revision_number":4},{"id":"15ff933dcd43136c8082a489ecfb56efdedcdc52","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 10:05:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9ccaa8da87ed3a7a415b3bacc7f6fc1c6abd8434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:05:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a3887de2cd816a2541231209ec69bd9347d7c402","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:33:38.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a106eb64a98041f6a7fa6e7061236ba6 : SUCCESS in 43m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/aee98bb50ba9478d81e113bb2f4be452 : SUCCESS in 1h 04m 35s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b2d6fa55fc414b01859e6ba53614831f : SUCCESS in 1h 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af4304e517e4799a2ffc4c73e8b69a2 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dcdc83a1d9647caa8fe8bcc61a26575 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16e0b3cb9c74413ca20a143543b6810c : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69fc20f875624301989e4513b3c39d6d : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f7710e57cbe441c9ac10abf7eff7e5e : SUCCESS in 5m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c35c84667a81487891664987cbf7639b : SUCCESS in 37m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7ea91fa4cd3417d988212b44ea27072 : FAILURE in 44m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/49d0831509074e74b7996b812dabbcb6 : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":4},{"id":"c985d828fa74bdabd066a8f0db7454ca263b1f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-08-05 16:51:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"eaf2234971f105f3677beff8db0bbfbd4a95d80d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 18:00:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/07b5f593b0f14f958a1541b50c768000 : SUCCESS in 23m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9ea8b36b7dcd43bd9f59a10914cb3ee9 : SUCCESS in 34m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/42e1a21e3da34fb2840098041a32a5e4 : SUCCESS in 1h 00m 34s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9802ff886857447a83f2f6bff46bd585 : SUCCESS in 59m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b73b175df3a24b73a7c20dad81fa48d1 : SUCCESS in 1h 06m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/502d7ae4369d4a2fa24f4e25cc11ed10 : SUCCESS in 20m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e76ab75c66724d4b897cec2b47902296 : SUCCESS in 17m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aef366c417c244d3a7cb18fda136ddaa : SUCCESS in 49m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eea5600546ac45c69de0b918dba03215 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1981f13b890447a83b6da14f743cbf1 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6242d25654434944b67ef13f52409e25 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9a69e0bb4d147dc84e16101a020a32f : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de951061b984cc68b27bf99796ce224 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/317f0d7cbd5c41e490ce951a6716ec90 : SUCCESS in 4m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a6ec8d368b24c11ab8e6f1524e5d0d1 : SUCCESS in 19m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e8ba3ed729f407d9f25148db24a832b : SUCCESS in 24m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/151c4472d6fc4d5a98cf7bfc7ac66cff : SUCCESS in 1h 04m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d15b60ea9a6c4e7da6f11033e2025da4 : SUCCESS in 53m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ce25b2b219144fe983ae73de70c89482 : SUCCESS in 50m 37s","accounts_in_message":[],"_revision_number":5},{"id":"ea58e8e3716da16151d9471be52c1589a34ebf5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 18:22:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/098b354d5a194823b9adb779a6493080 : SUCCESS in 51m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2bbace47ea5547c59fe6e1cd4f9dc3cc : SUCCESS in 42m 09s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3489ee56eee42d3a50c7423f31ed8ed : SUCCESS in 1h 24m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b3e8d3ecb7d4dbfa439bb722257f8b8 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ae294969068413bb70e9be7a7628f2d : SUCCESS in 6m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c2a1347fc7a54d42bbb64dc0b1b43e7a : SUCCESS in 42m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5ddef778e52375570e29cdfb5060c1c2f77e8904","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 09:33:09.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"26423096bf73ea34e8e983b6c56e1dfd3f828d63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 09:33:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ea08a9583b15fe3154a8305e8277294f8de7d565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 10:48:10.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d405ad92c6d14ae289a6f80088c1bb28 : SUCCESS in 35m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d1802c59fa0e41de9274ccf9cd645415 : SUCCESS in 1h 05m 13s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/afa91132027f44d6bb3eab37d04cdc54 : SUCCESS in 1h 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7ff6f1680d4e9a9a84fe913752d703 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9e814993c274be7b5683d32cf2a7150 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c4602afd7a54a288d5e719f1f7dba2f : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61e0e5d43dcc4e70b75f80e0abd5c7f5 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0145e5a0d9af401399dced2b5f10fa0d : SUCCESS in 4m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/22709904cc53446aae43932d3a974cd7 : SUCCESS in 29m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e36170b3f594b76be3d036a5f5b7859 : SUCCESS in 50m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/75693ae66eb544a387aac4e8f26b8935 : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":5},{"id":"df8ea8670120c6b6919df48c7f2fda361b066c0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 10:48:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"37b80fda3a1caab348dec62afb321846e00e6024","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 10:49:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/901f62dffd8047f99445bd0661a9bba3 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5339d27999614e8dbc253006b52ae45d : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5},{"id":"39e574e91dc84486cc89e44802cdf22f86acbf4b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 13:04:50.000000000","message":"Patch Set 5:\n\nBackported to ussuri.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0a410a54609104579dbbeb96c10e60e7d86dc165","revisions":{"7fd506d5f6151c71b249e74893d160d96ddd8ace":{"kind":"REWORK","_number":1,"created":"2021-05-24 03:50:00.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/51/792751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/792751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/792751/1"}}},"commit":{"parents":[{"commit":"a80de4a7926ab9188191505a86c24fe6db3fbb75","subject":"Merge \"Install Monasca datasource plugin for Grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a80de4a7926ab9188191505a86c24fe6db3fbb75"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 03:34:46.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 03:34:46.000000000","tz":-420},"subject":"monasca-thresh: Allow topology check and removal in storm","message":"monasca-thresh: Allow topology check and removal in storm\n\nPatch adds a script in the monasca-thresh image that can be used\nto check if a topology exists in Storm, and optionally kill it.\n\nThis is part of a bug in kolla-ansible where topologies were\nnot submitted to Storm, but run locally.  This patch includes\na topology check script enabled by KOLLA_BOOTSTRAP which will exit\nkolla_start if the topology exists, and optionally enables topology\nremoval (to allow replacement) enabled by TOPOLOGY_REPLACE.\n\nTopology names and various timeouts may be customized.  If the\nnew env variables are not set, existing behavior is unchanged.\n\nPartial-Bug: #1808805\nChange-Id: If8f0730031435dda4235b7f2d2c23e5f5f767f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fd506d5f6151c71b249e74893d160d96ddd8ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fd506d5f6151c71b249e74893d160d96ddd8ace"}]},"branch":"refs/heads/master"},"4e9ce230bd6a493d3c69f496aed98cb3999d9738":{"kind":"REWORK","_number":2,"created":"2021-05-24 16:17:23.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/51/792751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/792751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/792751/2"}}},"commit":{"parents":[{"commit":"1167be2d2f1ba91cdb6ece78d785b6721dc8e6c9","subject":"Merge \"add openstack-python3-xena-jobs-arm64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1167be2d2f1ba91cdb6ece78d785b6721dc8e6c9"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 03:34:46.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 16:17:03.000000000","tz":-420},"subject":"monasca-thresh: Allow topology check and removal in storm","message":"monasca-thresh: Allow topology check and removal in storm\n\nPatch adds a script in the monasca-thresh image that can be used\nto check if a topology exists in Storm, and optionally kill it.\n\nThis is part of a bug in kolla-ansible where topologies were\nnot submitted to Storm, but run locally.  This patch includes\na topology check script enabled by KOLLA_BOOTSTRAP which will exit\nkolla_start if the topology exists, and optionally enables topology\nremoval (to allow replacement) enabled by TOPOLOGY_REPLACE.\n\nTopology names and various timeouts may be customized.  If the\nnew env variables are not set, existing behavior is unchanged.\n\nPartial-Bug: #1808805\nChange-Id: If8f0730031435dda4235b7f2d2c23e5f5f767f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e9ce230bd6a493d3c69f496aed98cb3999d9738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e9ce230bd6a493d3c69f496aed98cb3999d9738"}]},"branch":"refs/heads/master"},"21ab72743b402290c162b77abd82bd8ae21e3de5":{"kind":"REWORK","_number":3,"created":"2021-06-05 01:34:50.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/51/792751/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/792751/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/792751/3"}}},"commit":{"parents":[{"commit":"1167be2d2f1ba91cdb6ece78d785b6721dc8e6c9","subject":"Merge \"add openstack-python3-xena-jobs-arm64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1167be2d2f1ba91cdb6ece78d785b6721dc8e6c9"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 03:34:46.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-06-05 01:34:16.000000000","tz":-420},"subject":"monasca-thresh: Allow topology check and removal in storm","message":"monasca-thresh: Allow topology check and removal in storm\n\nPatch adds a script in the monasca-thresh image that can be used\nto check if a topology exists in Storm, and optionally kill it.\n\nThis is part of a bug in kolla-ansible where topologies were\nnot submitted to Storm, but run locally.  This patch includes\na topology check script enabled by KOLLA_BOOTSTRAP which will exit\nkolla_start if the topology exists, and optionally enables topology\nremoval (to allow replacement) enabled by TOPOLOGY_REPLACE.\n\nTopology names and various timeouts may be customized.  If the\nnew env variables are not set, existing behavior is unchanged.\n\nPartial-Bug: #1808805\nChange-Id: If8f0730031435dda4235b7f2d2c23e5f5f767f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21ab72743b402290c162b77abd82bd8ae21e3de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21ab72743b402290c162b77abd82bd8ae21e3de5"}]},"branch":"refs/heads/master"},"077660e812226da46fd95bc3fec9fd0507cfefd9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-11 04:52:57.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/51/792751/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/792751/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/792751/4"}}},"commit":{"parents":[{"commit":"b9b60f62621e84e4861b526078b62d5636518be1","subject":"Merge \"Enable Debian updates repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9b60f62621e84e4861b526078b62d5636518be1"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 03:34:46.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-06-11 04:52:57.000000000","tz":0},"subject":"monasca-thresh: Allow topology check and removal in storm","message":"monasca-thresh: Allow topology check and removal in storm\n\nPatch adds a script in the monasca-thresh image that can be used\nto check if a topology exists in Storm, and optionally kill it.\n\nThis is part of a bug in kolla-ansible where topologies were\nnot submitted to Storm, but run locally.  This patch includes\na topology check script enabled by KOLLA_BOOTSTRAP which will exit\nkolla_start if the topology exists, and optionally enables topology\nremoval (to allow replacement) enabled by TOPOLOGY_REPLACE.\n\nTopology names and various timeouts may be customized.  If the\nnew env variables are not set, existing behavior is unchanged.\n\nPartial-Bug: #1808805\nChange-Id: If8f0730031435dda4235b7f2d2c23e5f5f767f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/077660e812226da46fd95bc3fec9fd0507cfefd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/077660e812226da46fd95bc3fec9fd0507cfefd9"}]},"branch":"refs/heads/master","description":"Rebase"},"0a410a54609104579dbbeb96c10e60e7d86dc165":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-05 16:51:42.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/51/792751/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/792751/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/792751/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/792751/5"}}},"commit":{"parents":[{"commit":"ea6ceef686c018d810a2ff20ca4ba81aa77164e5","subject":"Merge \"drop leftovers of RHEL support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea6ceef686c018d810a2ff20ca4ba81aa77164e5"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 03:34:46.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-08-05 16:51:42.000000000","tz":0},"subject":"monasca-thresh: Allow topology check and removal in storm","message":"monasca-thresh: Allow topology check and removal in storm\n\nPatch adds a script in the monasca-thresh image that can be used\nto check if a topology exists in Storm, and optionally kill it.\n\nThis is part of a bug in kolla-ansible where topologies were\nnot submitted to Storm, but run locally.  This patch includes\na topology check script enabled by KOLLA_BOOTSTRAP which will exit\nkolla_start if the topology exists, and optionally enables topology\nremoval (to allow replacement) enabled by TOPOLOGY_REPLACE.\n\nTopology names and various timeouts may be customized.  If the\nnew env variables are not set, existing behavior is unchanged.\n\nPartial-Bug: #1808805\nChange-Id: If8f0730031435dda4235b7f2d2c23e5f5f767f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a410a54609104579dbbeb96c10e60e7d86dc165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a410a54609104579dbbeb96c10e60e7d86dc165"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
