)]}'
{"id":"openstack%2Fkolla~792763","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a","project":"openstack/kolla","branch":"stable/victoria","topic":"centos8_stream_victoria","hashtags":[],"change_id":"Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a","subject":"Add support for CentOS 8 Stream","status":"MERGED","created":"2021-05-24 13:08:16.000000000","updated":"2021-08-11 14:34:52.000000000","submitted":"2021-08-11 08:39:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":13,"has_review_started":true,"submission_id":"792763-centos8_stream_victoria","cherry_pick_of_change":772841,"cherry_pick_of_patch_set":19,"meta_rev_id":"5d1d7bb70062c2594c385373b34b502126361e53","_number":792763,"virtual_id_number":792763,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-11 08:39:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-08-04 09:14:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-11 07:23:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-08-06 07:48:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-08-11 07:23:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 13:08:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-05-24 13:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-01 10:55:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-08-06 07:48:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-08-11 07:11:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-08-11 07:23:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"f40e8271b5df26e94a5413eba2bc8c072a5861b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-24 13:08:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* .zuul.d/centos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"a8b7dda204ddeec4229c4f65beab6a72f0e8195b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 13:08:40.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/victoria.  The error was:\n\n  while parsing a block collection\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 2, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 25, column 1","accounts_in_message":[],"_revision_number":1},{"id":"805bb246b4266b93a9bf12cdc5487b7aa44e6e2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 13:08:51.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/victoria.  The error was:\n\n  while parsing a block collection\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 2, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 25, column 1","accounts_in_message":[],"_revision_number":1},{"id":"046e8e0d7acc2c93cdd04e99c1e502db4f0d657c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-24 13:10:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a9b6a68057278e19cfa7f7089f5dca540f52f803","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-24 13:17:54.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b5ef275acc63209bd4e2947c74f937c0c9193da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 13:18:13.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/victoria.  The error was:\n\n  expected a list for dictionary value @ data[\u0027check-arm64\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          #- kolla-build-centos8-binary\n          #- kolla-build-centos8-source\n          - kolla-build-centos8s-binary\n          - kolla-build-centos8s-stream\n          #- kolla-ansible-centos8-source\n          #- kolla-ansible-centos8-binary\n          #- kolla-ansible-centos8-source-upgrade\n      check-arm64:\n        jobs:\n          #- kolla-build-centos8-source-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8-source\n          - kolla-ansible-centos8-source\n          - kolla-ansible-centos8-source-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-centos8-source-quay\n          - kolla-publish-centos8-binary-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8-source-dockerhub\n          - kolla-publish-centos8-binary-dockerhub\n      release:\n        jobs:\n          - kolla-publish-centos8-source-dockerhub\n          - kolla-publish-centos8-binary-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos8-source\n          - kolla-ansible-centos8-source-bifrost:\n              files: ^docker\\/(base|bifrost|openstack-base)\\/.*\n          # Test rabbitmq \u0026 mariadb changes in multinode ceph jobs.\n          - kolla-ansible-centos8-source-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8-source-upgrade-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8-source-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-centos8-source-scenario-nfv:\n              files: ^docker\\/(base|barbican|heat|mistral|openstack-base|redis|tacker)\\/.*\n          - kolla-ansible-centos8-source-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-centos8-source-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-centos8-source-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-centos8-source-masakari:\n              files: ^docker/(base|masakari)/\n          - kolla-ansible-centos8-source-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8-source-prometheus-efk:\n              files: ^docker/(base|elasticsearch|grafana|kibana|prometheus)/\n          - kolla-ansible-centos8-source-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 2, column 3","accounts_in_message":[],"_revision_number":3},{"id":"ed7264f63ad683898ce08fa24af6d8ccd7b67591","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 13:18:27.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/victoria.  The error was:\n\n  expected a list for dictionary value @ data[\u0027check-arm64\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          #- kolla-build-centos8-binary\n          #- kolla-build-centos8-source\n          - kolla-build-centos8s-binary\n          - kolla-build-centos8s-stream\n          #- kolla-ansible-centos8-source\n          #- kolla-ansible-centos8-binary\n          #- kolla-ansible-centos8-source-upgrade\n      check-arm64:\n        jobs:\n          #- kolla-build-centos8-source-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8-source\n          - kolla-ansible-centos8-source\n          - kolla-ansible-centos8-source-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-centos8-source-quay\n          - kolla-publish-centos8-binary-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8-source-dockerhub\n          - kolla-publish-centos8-binary-dockerhub\n      release:\n        jobs:\n          - kolla-publish-centos8-source-dockerhub\n          - kolla-publish-centos8-binary-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos8-source\n          - kolla-ansible-centos8-source-bifrost:\n              files: ^docker\\/(base|bifrost|openstack-base)\\/.*\n          # Test rabbitmq \u0026 mariadb changes in multinode ceph jobs.\n          - kolla-ansible-centos8-source-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8-source-upgrade-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8-source-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-centos8-source-scenario-nfv:\n              files: ^docker\\/(base|barbican|heat|mistral|openstack-base|redis|tacker)\\/.*\n          - kolla-ansible-centos8-source-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-centos8-source-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-centos8-source-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-centos8-source-masakari:\n              files: ^docker/(base|masakari)/\n          - kolla-ansible-centos8-source-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8-source-prometheus-efk:\n              files: ^docker/(base|elasticsearch|grafana|kibana|prometheus)/\n          - kolla-ansible-centos8-source-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 2, column 3","accounts_in_message":[],"_revision_number":3},{"id":"f69ea1f41e70354a9c03ecb81f97ac227cae37f5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-24 13:21:32.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c18ba1156b28a2afb46b22acd6d591722095af77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 13:21:51.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/victoria.  The error was:\n\n  Job kolla-build-centos8s-stream not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          #- kolla-build-centos8-binary\n          #- kolla-build-centos8-source\n          - kolla-build-centos8s-binary\n          - kolla-build-centos8s-stream\n          #- kolla-ansible-centos8-source\n          #- kolla-ansible-centos8-binary\n          #- kolla-ansible-centos8-source-upgrade\n      check-arm64:\n        jobs:\n          #- kolla-build-centos8-source-aarch64\n          - kolla-build-centos8s-source-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8-source\n          - kolla-ansible-centos8-source\n          - kolla-ansible-centos8-source-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-centos8-source-quay\n          - kolla-publish-centos8-binary-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8-source-dockerhub\n          - kolla-publish-centos8-binary-dockerhub\n      release:\n        jobs:\n          - kolla-publish-centos8-source-dockerhub\n          - kolla-publish-centos8-binary-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos8-source\n          - kolla-ansible-centos8-source-bifrost:\n              files: ^docker\\/(base|bifrost|openstack-base)\\/.*\n          # Test rabbitmq \u0026 mariadb changes in multinode ceph jobs.\n          - kolla-ansible-centos8-source-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8-source-upgrade-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8-source-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-centos8-source-scenario-nfv:\n              files: ^docker\\/(base|barbican|heat|mistral|openstack-base|redis|tacker)\\/.*\n          - kolla-ansible-centos8-source-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-centos8-source-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-centos8-source-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-centos8-source-masakari:\n              files: ^docker/(base|masakari)/\n          - kolla-ansible-centos8-source-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8-source-prometheus-efk:\n              files: ^docker/(base|elasticsearch|grafana|kibana|prometheus)/\n          - kolla-ansible-centos8-source-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 2, column 3","accounts_in_message":[],"_revision_number":4},{"id":"df475e2eefccf21e545673c1407cf0b4175cb52e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 13:22:03.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch stable/victoria.  The error was:\n\n  Job kolla-build-centos8s-stream not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          #- kolla-build-centos8-binary\n          #- kolla-build-centos8-source\n          - kolla-build-centos8s-binary\n          - kolla-build-centos8s-stream\n          #- kolla-ansible-centos8-source\n          #- kolla-ansible-centos8-binary\n          #- kolla-ansible-centos8-source-upgrade\n      check-arm64:\n        jobs:\n          #- kolla-build-centos8-source-aarch64\n          - kolla-build-centos8s-source-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8-source\n          - kolla-ansible-centos8-source\n          - kolla-ansible-centos8-source-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-centos8-source-quay\n          - kolla-publish-centos8-binary-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8-source-dockerhub\n          - kolla-publish-centos8-binary-dockerhub\n      release:\n        jobs:\n          - kolla-publish-centos8-source-dockerhub\n          - kolla-publish-centos8-binary-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos8-source\n          - kolla-ansible-centos8-source-bifrost:\n              files: ^docker\\/(base|bifrost|openstack-base)\\/.*\n          # Test rabbitmq \u0026 mariadb changes in multinode ceph jobs.\n          - kolla-ansible-centos8-source-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8-source-upgrade-ceph-ansible:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8-source-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-centos8-source-scenario-nfv:\n              files: ^docker\\/(base|barbican|heat|mistral|openstack-base|redis|tacker)\\/.*\n          - kolla-ansible-centos8-source-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-centos8-source-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-centos8-source-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-centos8-source-masakari:\n              files: ^docker/(base|masakari)/\n          - kolla-ansible-centos8-source-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8-source-prometheus-efk:\n              files: ^docker/(base|elasticsearch|grafana|kibana|prometheus)/\n          - kolla-ansible-centos8-source-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@stable/victoria\", line 2, column 3","accounts_in_message":[],"_revision_number":4},{"id":"e28a423ec8a71acff8a1c3117c2d9e2ad992a82b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-24 13:24:24.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0b13a32d92655c2f230127a93930d7ad0eba9a1d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-24 13:27:02.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"3dbb3f5885c90b0572b2f9b0b583a152aef493ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 14:12:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce935aad448849218f2455ab208e1ac4 : SUCCESS in 40m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b82d2ca5be28c57bc6397bd52c1d11e32684c1dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 14:15:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7b66113773cb46c6ad650ce2b02e00d0 : FAILURE in 35m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fa9e2d6117f24afc9d3beb4f29c6538a : FAILURE in 42m 25s","accounts_in_message":[],"_revision_number":6},{"id":"10beb1962c5d46a2a29bcef51b10ca8c322a44e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-26 10:56:49.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"8efcde8c2600152a6aaee6f21ac99f3fd632cd28","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-26 10:57:48.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"8fa95a26349783554552a441080b8e95e879dbec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 11:41:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e71ac526a4f54f4782f16a85bc3f72cb : SUCCESS in 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0b091a83223bcaf793efaeeceb50005e082edced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 12:32:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/13f23fe81f344644a1b586129abe7124 : FAILURE in 1h 32m 44s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/903471a5538541a8b6e97f9a51e11e76 : SUCCESS in 26m 48s","accounts_in_message":[],"_revision_number":8},{"id":"2ca5e9320bdaa82fc0288069e83b5eecee5aa1f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-27 11:55:33.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"d47563a885bcf5e078a9dc08ce71b49d264e82ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 12:38:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f07e9eb1d62947fba66d6f40c86e42e3 : FAILURE in 5m 47s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2abbb6ff62d744c5835b79f1caf4a856 : SUCCESS in 37m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ba13ac1e3b1a46109d147a45f400bd10 : FAILURE in 3m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/111d26c5fb87479980c7ec19f968855f : FAILURE in 15m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/68aece79a30f4637b35a2de47009aff5 : FAILURE in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4ba5372f56ca232d5a0a360d6e060eb52e31d237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:24:19.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1440b67c91574337a732d5f25168951d : FAILURE in 5m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/79806b822dd5427ab561183d0eda835e : FAILURE in 5m 08s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8ac7edf3930e4793a8c43e8fb43552f3 : SUCCESS in 21m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bd3a3bd51812489aa4b14273245be379 : SUCCESS in 25m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d3e114c49fc443b89ed14bdb0761e1b6 : FAILURE in 42m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c0aaebb0c27a4d6d87b5a6651e35f628 : FAILURE in 46m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/71c2604e8d034ec9b59be3cfbd824c00 : FAILURE in 1h 07m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/729e0a19d3c94ac7b8aab43bd2367fca : FAILURE in 3m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0c9ddf45fe934024aca773141fdc7574 : FAILURE in 7m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cec9ed1ee62040cbb42fae2ecb76138a : FAILURE in 11m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b0bd85700e147fda355e755a29fabed : FAILURE in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe48ec19f5504d9a8bfa1d9ca58bcd83 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc6c73512179486eb61be29ba7cc9435 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5db38c813fd4ff49ff91bfc679f6d6d : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a08be440748d4de29c39549d19925770 : SUCCESS in 4m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b2f27053695248f3a081f92de38aa8b9 : FAILURE in 3m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/92a7f16a089e4239b7962c7bd4ac5a4a : FAILURE in 3m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/22f41dcd3bd745c382b6ac16129fbd30 : SUCCESS in 49m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a35b1622accb4970b74d20825c11322c : SUCCESS in 44m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/54520064253348a2b5dd231651234424 : SUCCESS in 1h 27m 19s","accounts_in_message":[],"_revision_number":9},{"id":"e2dcc7818e5785c983cfbfa5c6d36a358dd54383","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-27 14:44:14.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"ab2b77b7ecb336b47870d1459e29bb8b2409558b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:01:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6e4cd862b80d4cdeba4a911c135e5ba6 : FAILURE in 5m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cfd26df6053546508ffd31557b50d7bb : FAILURE in 5m 33s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/55b63865298e45dbb12ab6c3288ef9fe : SUCCESS in 24m 49s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/053420a2bea4411ab09073fcc7effa3a : SUCCESS in 27m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/288a374ba01742e2bb6bbb566dd243f6 : FAILURE in 53m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/441165f5dbc246c3b34ee082d4271936 : FAILURE in 50m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c3aa266d147e4b36a7a85f5c1075f544 : FAILURE in 1h 12m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6dc73ca13c9d46c09d751d195c881284 : FAILURE in 4m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ab1e1e06913248c59ab6232ea606e26b : FAILURE in 4m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7fa8b304a0244dd9bc091cd50ac6c0a3 : FAILURE in 13m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7684d437c4024dc7acb08fd53df847e2 : FAILURE in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fc0e644e0e1464787c7162cc7e61897 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb9a571102634f74b80b046b785dd958 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1827f95a87b04c058230decf7ca201b6 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34bd7c42003c4b1088c8fe6efa144c51 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/396accf451f34cc3a9e70814e973ec9b : FAILURE in 4m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9287ba9bd933427aac6784d75938b2f9 : FAILURE in 5m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb1ab595a61e4306aabaaec0eaeefd81 : SUCCESS in 55m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/35ca8193be224843886817391ac19bbb : SUCCESS in 49m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4266b6c4a08045199026a70c8cb0781f : SUCCESS in 1h 07m 30s","accounts_in_message":[],"_revision_number":10},{"id":"6205fc84434d2cb8b98a81cba68fc0878bb3305b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:25:25.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/59512faf3e7a4bac93faad907f163207 : FAILURE in 6m 50s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/39a312be43d14ae685023135555dab74 : SUCCESS in 1h 38m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/20787ffd644f4732a918ca99e53e3146 : FAILURE in 6m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2bd0232e821144edb78d89a75c70f0a0 : FAILURE in 15m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/550404ecfac4410a8876fa38da859f75 : FAILURE in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"54f969e3e16b6949ce155cd9150faad500ae2a8c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-28 15:14:27.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"9c2107eb62a1a89c67f6b3fe244b4042834fe03d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 16:14:58.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7d45a9c8770f4d60b1e7f00ba78de6cb : SUCCESS in 38m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d9522c32bca94eacb8adedc30dd4ee2b : SUCCESS in 23m 27s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/665f1f5fbb8c419f9c33ad2536742e58 : SUCCESS in 32m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e11202d475814509a1faaa0fc5021362 : SUCCESS in 27m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/56ab47af33c643b8966a15b56a329139 : FAILURE in 37m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/39e5770414cc431081004d2a8f2f672c : FAILURE in 41m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ff14d4cdb61e487497f105997e1fbb1e : FAILURE in 59m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/184d5a6d7df340728fb7043e06b930ab : FAILURE in 5m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/935b77b252bc466a91f09f40a0f07715 : FAILURE in 6m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f8a9b3621df04f1c8cf1e59afd3d5275 : FAILURE in 9m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba3334a4796469eadf52fd3a5a3fc88 : FAILURE in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/219b2617c3954d45af68b43b45051a9c : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/603d24c9a8f844269b0f7504414f7c4a : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae21a6b20aff4df9a5e787dacfc93ec3 : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df237159765e467d8139117f79dda42a : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b10b5b2b4c344aa4b8b8fbc01a86f75d : SUCCESS in 30m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a3e2f5130df470db5934561d42bafc1 : SUCCESS in 19m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9d433bc055154aba8c6c4e5c8bd719b7 : SUCCESS in 40m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f2e6bc9fe68420bbb0fd07e393f487f : SUCCESS in 47m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0898c4c766a04e61853bda15165ad975 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":11},{"id":"e8ef83855900cf21274110f8060e5913997d3d34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 16:32:53.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5498fb11a7304fe68e94c77c115b9ee9 : SUCCESS in 36m 13s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/246ec36a1b394aa4b7faf56c9c380bc5 : SUCCESS in 1h 16m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fee42fe4561149c8b725030cdf868332 : FAILURE in 9m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8f5796fbbd1f4e82b6808c40531cbe67 : FAILURE in 14m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5eb3e17ccc84afabd5c6eab1cea8cdf : SUCCESS in 1h 03m 42s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1104298fc4a0dcc5a45bd9975655314e17c3ba41","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-31 08:29:07.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"2bdb514d363e0209ad39e1433327bbf7949d77c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 09:48:39.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/59a819154e264590afb2e2359c9466bc : SUCCESS in 53m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/554cb0cc71a547cc90f74d983a91c95c : SUCCESS in 43m 30s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ad268b6f8bb947898f63de38dcf61855 : SUCCESS in 25m 11s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e04ab16a620d4336ba856c4b478830af : SUCCESS in 30m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8d47a08d249e4069b52a6480ef266dc8 : FAILURE in 47m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/882f5ea15a394a7892720d4aaa172907 : FAILURE in 47m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0f1fb36df57a4cf59aa99177bc0e683f : FAILURE in 1h 08m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5615826303a2468fa6dba35d8bcf127d : FAILURE in 6m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ccf0b29fd7e74061946a8e05c0b30f83 : FAILURE in 6m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9d3d5ddd737347cf81da33e1b2c92b73 : FAILURE in 11m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c957b4c8023412095b97c313fbd0000 : FAILURE in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c09a19526c643e68b6a9138898d8ea9 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c99ad10a80bb4e6d9982c5dcf9c0a291 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757d455550b447ab95ebbe4871054894 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44e90753204147708f297313758c3435 : SUCCESS in 4m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f1098748986b446a82d71ac2fb566c44 : SUCCESS in 22m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9dde3e7b8f31471cb5fc92e72e05024a : SUCCESS in 24m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bbcc7ea4435d4ea9b1a44df21eac64bf : SUCCESS in 55m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae170f68d85c43028b40b49aaa3dadf1 : SUCCESS in 50m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b50036e7cd564c1282f2ec5d975fc966 : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":11},{"id":"af4eec69e8be43910dd24ea00ac00e23972beaac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-31 09:58:25.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"fd9a974662f978a858f199a9cc21add328c0020c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-31 09:59:18.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"2b6c4875fe65efa5f24d1235bfb6b252ddb08a30","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-31 10:00:00.000000000","message":"Topic set to victoria_centos8s","accounts_in_message":[],"_revision_number":13},{"id":"28281ce8b0ebab5ad5621792f83fa581c534be58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 10:17:52.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/95b5807955224ff7ac66eccf036f4f94 : FAILURE in 6m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/390adc0451ce43abb88b6d083ecc1e13 : FAILURE in 6m 50s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a19d3a4ea4264d968b57357f3707b60b : FAILURE in 6m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8ba7f960069f4f7e82be81e7932b823d : FAILURE in 6m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/087e3469a6a343159d31871631097688 : FAILURE in 13m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ec684e14fa3542a8a7b65144c03e1247 : FAILURE in 12m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/769d4d26fdef49838e2b3570699f3476 : FAILURE in 11m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/84381e2ada4b4cd59864686258d40359 : FAILURE in 7m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/844dde7029134d7a8d5d96275e402d6e : FAILURE in 6m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c7d875b6535242f3aa032fb7741ecb95 : FAILURE in 11m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f96016f0eda0490db0e0dadbf80cfd73 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/211cb2710b7e4a4ea3d2feb43e06753b : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b144ea6d43849e78f20676b5a5f21f8 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb5487f60b843bb9448ee8fc0b6504b : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43da87aacef0464eb59e7b3aa457da8a : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e266d005853d4bd5bab8d20c29e059f2 : FAILURE in 4m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/92458aca978848938517f52135f74964 : FAILURE in 4m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c81ed5967e2e48c0a1b4f6e2eb3ef571 : FAILURE in 8m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/134e9080519d41509eb3ac974ae5ae64 : FAILURE in 8m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2432099a332e4ab1ae2cba15a9897ac8 : FAILURE in 8m 38s","accounts_in_message":[],"_revision_number":13},{"id":"ad9402b2c993b749a2409578cea02d6f064173ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 10:19:55.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1732c13d2a04f2e9cc573ed7dcaf014 : FAILURE in 7m 22s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/83e259e7dff342c8921e8a6f1beb49c2 : FAILURE in 6m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc51b5d9737347af8e2dcc6b2d7d431b : FAILURE in 4m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bfe3427a764642f0bef49b9a91565c06 : FAILURE in 10m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/246644e7f42043f0960308b2ca49adb5 : FAILURE in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"27b878f607d8a4e933d180d75f5ec0bd358afc75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-09 08:41:27.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"06f17f26df184f6c8c7f305870d0850632820ca0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:22:14.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/254d4f33d30248b49dc3e93ce1ab911e : SUCCESS in 1h 31m 11s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/024ae2bd771e42da93df7da835e0d111 : SUCCESS in 1h 38m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f9baf18056e402488b569f1aaec9506 : FAILURE in 9m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c54922f91644869b2e4f5aab4a4e75a : FAILURE in 17m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7151861dbf54caf9cbde15fa7b481a9 : SUCCESS in 1h 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2e14ace645b9b9957872771132a5d03e8f9ff097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:26:46.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ddc72cda38da4b7d9a3a3e648f9e78fd : SUCCESS in 28m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7ab0522406284442a87c5db68ff58c1d : SUCCESS in 36m 46s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/78df4670161740aabac0543a48ccba6c : SUCCESS in 28m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/db11a051f00b4744bf14795c31d053a1 : SUCCESS in 49m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/21f970a9659348599f2e27ef73c59294 : FAILURE in 50m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b3792e2296c142a0ae9deb9f2600693a : FAILURE in 1h 07m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c53c52261f2f4ad2bab4e1ff5a760bc8 : FAILURE in 1h 09m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5c33597af0784619b825fc8eeb0953ba : FAILURE in 6m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c920e2a3a64d481d8798680240fbb931 : FAILURE in 6m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4d96dcd2857e44a390cdecb1773ed638 : FAILURE in 11m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98628247a02c430d9209550466913b32 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8aa9819c34944129dda485e7c166b3f : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97fc7dd37eda458ba81c899f956e4633 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c51370b62545adbaed390431ab66f8 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa903cd284d1453192bbd688343f6af4 : SUCCESS in 4m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/836da746f8cc4171acfca665f1b62ad8 : SUCCESS in 48m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/99bf3ed686454a0f8bd9c1d3cbd8fea5 : SUCCESS in 47m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2fd7fecf69a041f4884d9f4168e17543 : SUCCESS in 1h 30m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e51054434fe04a9488791c148faa3bbd : SUCCESS in 1h 12m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c635fbc76e8b41be828100d70e6ecf8a : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":14},{"id":"be9b1d0996ce76b4b481e075cdb0d9dbd20cc51d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-17 06:28:41.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"1cb79db161dde273f6682ac5e89335462144211b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 08:11:34.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7d4ded2f37904b349e1910c866ae2e1a : SUCCESS in 27m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c09de8dbff774742ab1b3bda7a2fc56e : SUCCESS in 28m 09s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/857669e7130045618b1ac1bcc0a49b11 : SUCCESS in 27m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/347b639d3704441caf4cdf01b72c4255 : SUCCESS in 28m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1627670a21b54802a3963a6e9f7a845f : SUCCESS in 1h 05m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c715594cce1c41958d73a0f02ddd5bd1 : SUCCESS in 1h 23m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9097665e3ee444a3952101b45f571afd : SUCCESS in 1h 16m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/99ac9adc45a94aefa37e99d565aaaed2 : SUCCESS in 24m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/520c74681d334ba1ac5bfd271fa5853b : SUCCESS in 1h 17m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b8d1b1236523445a8c9016df2373b42c : SUCCESS in 58m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f5cff221427430cb8b413eee7e5633e : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17656738cc4e4711825edc244a74f0b8 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a95109ad810f4cb2bbd810617ab3a39b : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ca087992c84f8f957de1702b38e2cc : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47c7c9d23be740d1bc3e3397ec78a8bb : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2adff96eecb6447cb727dd58501e4f97 : SUCCESS in 22m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb8be3434c7b4561b62dd49c6b230d6c : SUCCESS in 22m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dca1ba6148bd41a4a672d7e183be5bcd : SUCCESS in 57m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf0011b3c7a34f2788a89939a29d1a65 : SUCCESS in 54m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/710b6546d6b743a68ef64b3b63814523 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":15},{"id":"0f09c9addeaaa94c6d822b865eacf67e266c30cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 08:52:16.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb8113dee2b44777891a30f11b2ab77d : SUCCESS in 1h 59m 33s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9737980edcfc467094f91b3232ac404c : SUCCESS in 2h 04m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aab7cabbc9fa41baa64f9abb13cbfbe4 : SUCCESS in 1h 49m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0bcd43d9c4f400cbd32def6d5a7a796 : SUCCESS in 2h 20m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee2927ae59e54188a0c459072d0938e1 : SUCCESS in 1h 50m 58s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8c67f127331661bacf8b4a6db433cd8efb76b840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-18 07:10:53.000000000","message":"Patch Set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"7091be77b6146a11a8e99fca94dc87a30ca7fa10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 08:45:50.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/88f691d48faa47d7a09c99b84e6b1500 : SUCCESS in 41m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ad7f2caf22704097a7df99048f414945 : SUCCESS in 42m 01s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9400fb0ada0e45d79a0c2a82fbde80b9 : SUCCESS in 40m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/661b19861dea445080ebfb4e4802d559 : SUCCESS in 24m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ab243ab817bc4138bf7b21231938e152 : SUCCESS in 58m 51s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/78f41ef6f1144de1938fd4536a765cf7 : SUCCESS in 57m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/94900de0b2604ccb9044c065105c1697 : SUCCESS in 1h 14m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/203338520ad04e2b94a25a59ee7e2cff : SUCCESS in 21m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c6f83d5c4c104ad1851661e0c62fe700 : SUCCESS in 19m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f2f200685da34e54a37fcf8ed7770edd : SUCCESS in 58m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e77c902ba9144f1bad09c785d88491f : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dde0f9b1563040bb83086a5ea868a9fc : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab81c2be84504bf3bf529dcee8f95026 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de4c0435bb3c4605a0e0e19f89e555c1 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f11ef6ccda6848cb95bb17403fb0343b : SUCCESS in 6m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae5b147761454d819b9b07970f667316 : SUCCESS in 25m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b38b41727893474493666bc8e8781f92 : SUCCESS in 26m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c3f4577506014a7b97cb78f1fa279d38 : SUCCESS in 1h 05m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a53fed07ea784b6aa24d94ca1d38b42d : SUCCESS in 54m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/07c053064cf14ec0a31e2d7af607fe8e : SUCCESS in 1h 22m 30s","accounts_in_message":[],"_revision_number":16},{"id":"88efdf20d1df219f85abcf1160e94ab4f51fde1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 12:57:48.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd0ee29ac9924064b7235a6b5a2fd684 : FAILURE in 2h 32m 53s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8702d075714c437e8c81e0e0cb012cc2 : POST_FAILURE in 5h 34m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d427522f8d80447090b2c4bc6c5c4fba : SUCCESS in 1h 34m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2aeb91689cd9472bbfd51823dcfedda8 : TIMED_OUT in 3h 02m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3aeaa3ffd0884010b774ea00a8fcbe92 : FAILURE in 2h 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"e7d8485b897be78c47f8d439486de56df61a8991","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-23 15:48:25.000000000","message":"Topic victoria_centos8s removed","accounts_in_message":[],"_revision_number":16},{"id":"3b90b1869f68a2286fa90ce2f9bf2580bcf519b2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-23 15:48:36.000000000","message":"Topic set to centos8_stream_victoria","accounts_in_message":[],"_revision_number":16},{"id":"ba8c68ba869c09ead1d103277d2bfc264761e2bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-24 08:38:19.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"7e12bd1075654ec6d96bcf68c88b430a6abdb31d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 09:59:26.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b43b52d2354c406f9a8a3bf2ab9c6611 : SUCCESS in 33m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8f6608523f754b5993ee05da1ba95e89 : SUCCESS in 34m 15s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f112c11d078947479964cc46a3dc3848 : SUCCESS in 26m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d3d0a746d7fe427e9311106f47f4df78 : SUCCESS in 26m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/20a5fa12e66e464bbdd8b735fe7f0812 : SUCCESS in 1h 01m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fdcfff216e2a40bf9a7ed6121d02b556 : SUCCESS in 1h 14m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/68b0aa0075524e1383d499e8d35fba40 : SUCCESS in 1h 06m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/31a2a68e36db4a50b3da7cd4fad4592c : SUCCESS in 22m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7f23dbf6493c4fef92e12dced3e73ccf : SUCCESS in 20m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c728e79bcb5a4622bc0a5c9383c1310b : SUCCESS in 59m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67bcd796db274c838ebe2e61dc1c0f65 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07523d8d218d48ecbaa05491c49b0cdc : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b66756022f224b7086be89c415047e9c : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a67cd11947c844f697ea4a4d7ce18d4a : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5facfe4aea084fd19c75fecfb0865293 : SUCCESS in 5m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3dd284995e8a4fa1933a3f379cfb7c8a : SUCCESS in 23m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/485d607b677d4ab8abd1186255a04580 : SUCCESS in 22m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0733263a3dad4bad9f839677e8c06861 : SUCCESS in 54m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f95c0cf65c6f411282fffff377948904 : SUCCESS in 54m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d785fadcf7de4f2ca92268642e063cce : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":17},{"id":"3382785cc4ed6b3f21471350d3d3ab4bcd332612","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 11:12:46.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef2c8e5bc28c483b8a45e137a8d57ac8 : SUCCESS in 2h 21m 31s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/aeda4543f5b4492f8c53820d61a3cf5c : SUCCESS in 2h 31m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ed14bdf6e5c4213a9d0e5ebbcfc10e7 : SUCCESS in 32m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/300944d903ae4eaaa0dc83eb4c273d19 : SUCCESS in 1h 30m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/90fe267817884fa3b7992a2015d6a953 : SUCCESS in 2h 05m 26s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"7d11e499b7b324b1641136351776db61fdfc9b34","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-01 10:55:22.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)\n\nCentOS Linux images use stream as a base: https://8a28d37c8a21ba47ad86-2bdb8be3dd1329f8a48d0e165eec17e9.ssl.cf1.rackcdn.com/792763/17/check/kolla-build-centos8-source/8f66085/kolla/build/base.log","accounts_in_message":[],"_revision_number":17},{"id":"b35a21d175b2b64e211c548395e27839911d4d19","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-01 11:03:46.000000000","message":"Patch Set 17:\n\nSupport matrix needs stream adding as a distro. Currently the supported images are the same, so we could keep existing columns. We\u0027ll need some docs describing how to build stream images. Here\u0027s what we added for Train:\n\ndiff --git a/doc/source/admin/image-building.rst b/doc/source/admin/image-building.rst\nindex 7adfcac21..6f6d3a4cd 100644\n--- a/doc/source/admin/image-building.rst\n+++ b/doc/source/admin/image-building.rst\n@@ -74,6 +74,25 @@ There are following distros available for building images:\n - rhel\n - ubuntu\n \n+See the :ref:`support matrix \u003csupport-matrix-base-images\u003e` for information on\n+supported base image distribution versions and supported images on each\n+distribution.\n+\n+The Train ``9.1.0`` release supports both CentOS 7 and 8 base container images.\n+The default is to use CentOS 7. CentOS 8 may be used as follows:\n+\n+.. code-block:: console\n+\n+   kolla-build --base centos --base-tag 8\n+\n+Kolla uses the tag to set defaults for the distribution Python version and\n+package manager. If you need to use a base image with a different tag, use the\n+following options:\n+\n+.. code-block:: console\n+\n+   kolla-build --base centos --base-image \u003cbase image\u003e --base-tag \u003cbase tag\u003e --distro-package-manager dnf --distro-python-version 3.6\n+\n .. warning::\n \n    Fedora images are deprecated since Newton and will be removed","accounts_in_message":[],"_revision_number":17},{"id":"7de532610d08f5384d68a1a2354e56d45f8c697b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-01 11:06:38.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"9cb50589fc003ed845ea5d38efd4ccd39455d85c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-01 11:09:38.000000000","message":"Patch Set 17:\n\nAdded kolla-ansible jobs here: https://review.opendev.org/c/openstack/kolla/+/798892","accounts_in_message":[],"_revision_number":17},{"id":"f10c6d28ab8c4d5eb5276853df3e53d70660e1e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-02 11:01:30.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"06987695aa323e85d1253c9f5cc87dd3aafb1d27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 13:19:33.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6e72e678d89449d89e3ca369b05a3fe : SUCCESS in 1h 49m 21s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/922f3210fa724848a8128eb1f9c9c5e9 : SUCCESS in 1h 57m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c53df5613240446d98589f4671de0f07 : FAILURE in 1h 55m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/08ba9039602445f68c316a97b5d73c9e : SUCCESS in 2h 16m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/61749ec0b3184d598d1c8aac41b69077 : SUCCESS in 1h 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6bdffb014053cc7cf3c063ea47022506ad3b1427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 13:27:41.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/216f195f646745daacb53d13991794f9 : SUCCESS in 28m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0a73138ef789419e82a790ba80fddc8a : SUCCESS in 29m 01s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5d84f06455ab4ac999edc289ade1feaa : SUCCESS in 53m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f039f01dbc7d46ed9cbe6a4df9bf4a12 : SUCCESS in 52m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9516f2c20e4a43dfbe64c4449571972a : SUCCESS in 1h 08m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b53a63ad64cf433ca16708805ce7f375 : SUCCESS in 1h 19m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0b3841658b8a426daee1f908ce197f83 : SUCCESS in 1h 24m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f0365521323e438b8476a2ae91e4988c : SUCCESS in 21m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f036d900ad04c62aae96aac885e9fe4 : FAILURE in 17m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc9dfe55d79b4bd2b3700d47fb8173f7 : SUCCESS in 57m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e3f056b688c42a19c07bc00bfea7a57 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cda98f973ea548a2ac0a4e7f85fb234f : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6dbf7612e394d73a97a9e25562bc296 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de980746edd74a5aa3478bc44834c37b : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cce015cb5de4d8e916c86c06f1d50c7 : SUCCESS in 4m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eefe3c59725f462bbb790fb784c4bad6 : SUCCESS in 26m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/92a1cf07e2fc49ce98a7674493d912ac : SUCCESS in 2h 24m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ae053a9d04e4e0b85bf30b4791493c1 : SUCCESS in 54m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/374e306e6d6b49ef95c1132f700d5514 : SUCCESS in 53m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8b3c73d14b5746c3a2e3ee0147a9f9dc : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":18},{"id":"5c7d5985134a2b2560f1616e45a71c2278f2161b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-03 09:19:35.000000000","message":"Patch Set 18:\n\nPlease address previous comments","accounts_in_message":[],"_revision_number":18},{"id":"9a74e35e3dc843c3e38edb6b916b99cc79759d78","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-03 11:36:36.000000000","message":"Patch Set 18:\n\n\u003e Patch Set 18:\n\u003e \n\u003e Please address previous comments\n\nOn my list, wanted to check build logs after addressing the first problem.","accounts_in_message":[],"_revision_number":18},{"id":"e7839536f6df7d37c591a44a97613c0cbe5bd385","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-03 11:36:40.000000000","message":"Patch Set 18: Workflow-1","accounts_in_message":[],"_revision_number":18},{"id":"2b497367dc1b8c90e9a4ffe524a05ff331210616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-04 04:33:45.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"03d3ea386b906a0f6e1214e922f2fa11444b4ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 05:43:34.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1f4d5b3c78574c3b80686a05bf084530 : SUCCESS in 27m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9747bb10fc9c488694a52c9b11dfab25 : SUCCESS in 25m 31s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f76e62b48e054c74a5c35dd9091b0c8f : SUCCESS in 26m 21s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a8e72162424342daacd1c76a1d5a8a1c : SUCCESS in 26m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/106a39a4683d489d9c83e266a3d3e6a0 : SUCCESS in 1h 04m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0755baff34b14bf4a634408e9c65fdc9 : SUCCESS in 1h 00m 46s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e5ec673b8d684d30a879a7845d6abbd0 : SUCCESS in 1h 08m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/650ef81866304765a4a4e9b803e8d3f5 : SUCCESS in 18m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6f979047fd064c2dbfd301579fb6dd4b : FAILURE in 19m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8ee2a8b10e6c40fab21550e071caaaf6 : SUCCESS in 1h 02m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1ef91a9ed9442f7bc2c6d7420a126da : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbaafd0db7124b8b978837da1a79ae13 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fde56ef06d34cc7b981e3891b46b719 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c34fccba5946a29ad9d7fc643526fb : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b755afd98e6f48f5a2776dc1ee01fd2a : SUCCESS in 4m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/994906e1dceb4dfb8678410cf35e3ee5 : SUCCESS in 21m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8bdec690a722470bbc6f7de333a70cca : SUCCESS in 21m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d2228dc1b9c4a089e1203ddb4a1f7e7 : SUCCESS in 47m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d914b952a6642b189bf643b0a44c486 : SUCCESS in 49m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0ef5bc091e6c450bb614ca7d71bf839e : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":19},{"id":"ccf117139a80766ce0790fbd090f1e9e5f7712c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 06:23:56.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/01e66f288d684193a7cee8e45e473e84 : SUCCESS in 58m 09s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8283e47c83484415aa5796c45bf234ca : SUCCESS in 59m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/30ec1172684a40c7b82f43d46b8f3b85 : SUCCESS in 54m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f56e4a415a647f08a7d5ee74a05e30a : SUCCESS in 1h 40m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5fe356e44c68457bbb0714fa3ec9047a : SUCCESS in 55m 40s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"eeed77ba3f829815e63ffe58a831c20a8782a1a8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-04 09:14:47.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"3add78ffb329a9365f9244a28e7bd635c291654b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-06 07:48:52.000000000","message":"Patch Set 19: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":19},{"id":"10702d3148929baf21000bd9f73ab1bb229f5590","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-11 07:11:47.000000000","message":"Patch Set 19:\n\n@Pierre: I think we would need to fix these in the newer branches as well because that\u0027s where these actually come from. I agree we used the non-canonical way.","accounts_in_message":[],"_revision_number":19},{"id":"376182035fab7489105e0fc821bc0e91fe7bf080","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-11 07:23:30.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\n(1 comment)\n\notherwise very nice; I am approving; please follow up on my and Pierre\u0027s comments; thanks!","accounts_in_message":[],"_revision_number":19},{"id":"6098a494fc4649ec0e257a627de3a347ccb6ad71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 07:24:14.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"4b6764dd68be7a414ade6784e91ba3f965b78c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:39:14.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a71f7a60c1734869a568e46cfbdf3904 : SUCCESS in 27m 40s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6bc268ac256e485d9a25f0e2b6b79769 : SUCCESS in 28m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6896837845ea478bb757a76e53c3b216 : SUCCESS in 1h 02m 54s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cb45726a66ee4ce48de4792e3235fa6d : SUCCESS in 1h 09m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02114cec7d114dad85a7a49c74aad5a7 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba5c90c810294a29bec1056713800d97 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ef20e56278c4197bfacb1765b77d488 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9292d08acb49098d5016a678166525 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccb17e988c3d49f4abd822ebe4053111 : SUCCESS in 5m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae3e74407a62489f88bf64e514dd9dda : SUCCESS in 26m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f62414434044370abd8ae47668e9805 : SUCCESS in 55m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a9eda0d4618d47be927f7ee828715fd0 : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":19},{"id":"06c14dc7abb97ee5448f9299d0f5ce4977a57433","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:39:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19},{"id":"e914e99fdf9feea909e452783f696b1eeb054cff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:42:37.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657196ea0609480f956df60c938ec0ec : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07b838e713194195aff5a0d7b9ea9478 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":19},{"id":"770078a2742236ea36205b94c55736ef3fe14425","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-11 14:30:00.000000000","message":"Patch Set 19:\n\n\u003e Patch Set 19:\n\u003e \n\u003e @Pierre: I think we would need to fix these in the newer branches as well because that\u0027s where these actually come from. I agree we used the non-canonical way.\n\nThe form is already correct in both master and stable/wallaby.","accounts_in_message":[],"_revision_number":19},{"id":"13462740d9dfd080b8723196f4d121dcc5141921","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-11 14:32:10.000000000","message":"Patch Set 19:\n\n\u003e Patch Set 19:\n\u003e \n\u003e \u003e Patch Set 19:\n\u003e \u003e \n\u003e \u003e @Pierre: I think we would need to fix these in the newer branches as well because that\u0027s where these actually come from. I agree we used the non-canonical way.\n\u003e \n\u003e The form is already correct in both master and stable/wallaby.\n\nExcept for the reno, were you talking about it?","accounts_in_message":[],"_revision_number":19},{"id":"5d1d7bb70062c2594c385373b34b502126361e53","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-11 14:34:52.000000000","message":"Patch Set 19:\n\n\u003e Patch Set 19:\n\u003e \n\u003e \u003e Patch Set 19:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 19:\n\u003e \u003e \u003e \n\u003e \u003e \u003e @Pierre: I think we would need to fix these in the newer branches as well because that\u0027s where these actually come from. I agree we used the non-canonical way.\n\u003e \u003e \n\u003e \u003e The form is already correct in both master and stable/wallaby.\n\u003e \n\u003e Except for the reno, were you talking about it?\n\nYeah, I have probably looked at it. Anyhow, I hope Michał follows up on this. We have not released this yet and it\u0027s not overly confusing.","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"2877c2c40e571fda90ce1352a633b8b95cf9003e","revisions":{"471ac558ca74ab408809fd8746bd0c1a74ee9708":{"kind":"REWORK","_number":1,"created":"2021-05-24 13:08:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/1"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-24 13:08:16.000000000","tz":0},"subject":"switch to CentOS 8 Stream","message":"switch to CentOS 8 Stream\n\nWe use official CentOS 8 Stream image as a base.\n\nSo far Stream does not have own SIG repos\nRDO Wallaby packages will be built against Stream\n\nKolla CI jobs are renamed to centos8s. Kolla-Ansible ones will follow\nin later patch.\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/471ac558ca74ab408809fd8746bd0c1a74ee9708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/471ac558ca74ab408809fd8746bd0c1a74ee9708"}]},"branch":"refs/heads/stable/victoria"},"b744ce92ac719cbbda4eef17bff6170c86ff0f7a":{"kind":"REWORK","_number":2,"created":"2021-05-24 13:10:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/2"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-24 13:09:06.000000000","tz":0},"subject":"WIP: switch to CentOS 8 Stream","message":"WIP: switch to CentOS 8 Stream\n\nWe use official CentOS 8 Stream image as a base.\n\nSo far Stream does not have own SIG repos\nRDO Wallaby packages will be built against Stream\n\nKolla CI jobs are renamed to centos8s. Kolla-Ansible ones will follow\nin later patch.\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b744ce92ac719cbbda4eef17bff6170c86ff0f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b744ce92ac719cbbda4eef17bff6170c86ff0f7a"}]},"branch":"refs/heads/stable/victoria"},"e15e79f0316905d020cdf01417df24c0a22405a9":{"kind":"REWORK","_number":3,"created":"2021-05-24 13:17:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/3"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-24 13:17:48.000000000","tz":0},"subject":"WIP: switch to CentOS 8 Stream","message":"WIP: switch to CentOS 8 Stream\n\nWe use official CentOS 8 Stream image as a base.\n\nSo far Stream does not have own SIG repos\nRDO Wallaby packages will be built against Stream\n\nKolla CI jobs are renamed to centos8s. Kolla-Ansible ones will follow\nin later patch.\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e15e79f0316905d020cdf01417df24c0a22405a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e15e79f0316905d020cdf01417df24c0a22405a9"}]},"branch":"refs/heads/stable/victoria"},"c1c7c157a638340a7187615c5950a98254411464":{"kind":"REWORK","_number":4,"created":"2021-05-24 13:21:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/4"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-24 13:21:07.000000000","tz":0},"subject":"WIP: switch to CentOS 8 Stream","message":"WIP: switch to CentOS 8 Stream\n\nWe use official CentOS 8 Stream image as a base.\n\nSo far Stream does not have own SIG repos\nRDO Wallaby packages will be built against Stream\n\nKolla CI jobs are renamed to centos8s. Kolla-Ansible ones will follow\nin later patch.\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1c7c157a638340a7187615c5950a98254411464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1c7c157a638340a7187615c5950a98254411464"}]},"branch":"refs/heads/stable/victoria"},"35b485d92a267733fbc3ea605915a7a0d1c3e110":{"kind":"REWORK","_number":5,"created":"2021-05-24 13:24:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/5"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-24 13:24:20.000000000","tz":0},"subject":"WIP: switch to CentOS 8 Stream","message":"WIP: switch to CentOS 8 Stream\n\nWe use official CentOS 8 Stream image as a base.\n\nSo far Stream does not have own SIG repos\nRDO Wallaby packages will be built against Stream\n\nKolla CI jobs are renamed to centos8s. Kolla-Ansible ones will follow\nin later patch.\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35b485d92a267733fbc3ea605915a7a0d1c3e110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35b485d92a267733fbc3ea605915a7a0d1c3e110"}]},"branch":"refs/heads/stable/victoria"},"ce33b405298c9cb2ecb350bbb6579f33455ef3be":{"kind":"REWORK","_number":6,"created":"2021-05-24 13:27:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/6"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-24 13:26:56.000000000","tz":0},"subject":"WIP: switch to CentOS 8 Stream","message":"WIP: switch to CentOS 8 Stream\n\nWe use official CentOS 8 Stream image as a base.\n\nSo far Stream does not have own SIG repos\nRDO Wallaby packages will be built against Stream\n\nKolla CI jobs are renamed to centos8s. Kolla-Ansible ones will follow\nin later patch.\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce33b405298c9cb2ecb350bbb6579f33455ef3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce33b405298c9cb2ecb350bbb6579f33455ef3be"}]},"branch":"refs/heads/stable/victoria"},"f8d068f33f11ccdc5721397a5dc5ef16e8d963f3":{"kind":"REWORK","_number":7,"created":"2021-05-26 10:56:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/7"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-26 10:56:45.000000000","tz":0},"subject":"WIP: switch to CentOS 8 Stream","message":"WIP: switch to CentOS 8 Stream\n\nWe use official CentOS 8 Stream image as a base.\n\nSo far Stream does not have own SIG repos\nRDO Wallaby packages will be built against Stream\n\nKolla CI jobs are renamed to centos8s. Kolla-Ansible ones will follow\nin later patch.\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8d068f33f11ccdc5721397a5dc5ef16e8d963f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8d068f33f11ccdc5721397a5dc5ef16e8d963f3"}]},"branch":"refs/heads/stable/victoria"},"e02f0fd479a98f1ddd6bff094130b4b82e633cf2":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-05-26 10:57:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/8"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-26 10:57:44.000000000","tz":0},"subject":"WIP: Add support for CentOS 8 Stream","message":"WIP: Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e02f0fd479a98f1ddd6bff094130b4b82e633cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e02f0fd479a98f1ddd6bff094130b4b82e633cf2"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"},"618dd7690fab9d00c325d166b74016213d74f0af":{"kind":"REWORK","_number":9,"created":"2021-05-27 11:55:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/9"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-27 11:55:29.000000000","tz":0},"subject":"WIP: Add support for CentOS 8 Stream","message":"WIP: Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/618dd7690fab9d00c325d166b74016213d74f0af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/618dd7690fab9d00c325d166b74016213d74f0af"}]},"branch":"refs/heads/stable/victoria"},"19a9cb49dafaac5426404d3c15a1dddd1da6891c":{"kind":"REWORK","_number":10,"created":"2021-05-27 14:44:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/10"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-27 14:44:07.000000000","tz":0},"subject":"WIP: Add support for CentOS 8 Stream","message":"WIP: Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a9cb49dafaac5426404d3c15a1dddd1da6891c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19a9cb49dafaac5426404d3c15a1dddd1da6891c"}]},"branch":"refs/heads/stable/victoria"},"5a2b3778fc5608fcf96dd72d2b1b9d0a29d15f13":{"kind":"REWORK","_number":11,"created":"2021-05-28 15:14:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/11"}}},"commit":{"parents":[{"commit":"1366332e59ffef4121dced126ef7d19a8b12a72f","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1366332e59ffef4121dced126ef7d19a8b12a72f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-28 15:14:23.000000000","tz":0},"subject":"WIP: Add support for CentOS 8 Stream","message":"WIP: Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a2b3778fc5608fcf96dd72d2b1b9d0a29d15f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a2b3778fc5608fcf96dd72d2b1b9d0a29d15f13"}]},"branch":"refs/heads/stable/victoria"},"13313d8c34b7308e7f0ed71e6b5e61ef4c582a06":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-05-31 09:58:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/12"}}},"commit":{"parents":[{"commit":"cf59a10eaa8e27d23b76a1faca77a3fae7c422dd","subject":"WIP: libvirt: pin edk2-ovmf for CentOS Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf59a10eaa8e27d23b76a1faca77a3fae7c422dd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-31 09:58:25.000000000","tz":0},"subject":"WIP: Add support for CentOS 8 Stream","message":"WIP: Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13313d8c34b7308e7f0ed71e6b5e61ef4c582a06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13313d8c34b7308e7f0ed71e6b5e61ef4c582a06"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"ddeb7e7107f55d16ec57f5cd20168447b7d9a6d6":{"kind":"REWORK","_number":13,"created":"2021-05-31 09:59:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/13"}}},"commit":{"parents":[{"commit":"cf59a10eaa8e27d23b76a1faca77a3fae7c422dd","subject":"WIP: libvirt: pin edk2-ovmf for CentOS Stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf59a10eaa8e27d23b76a1faca77a3fae7c422dd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-05-31 09:59:15.000000000","tz":0},"subject":"WIP: Add support for CentOS 8 Stream","message":"WIP: Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddeb7e7107f55d16ec57f5cd20168447b7d9a6d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddeb7e7107f55d16ec57f5cd20168447b7d9a6d6"}]},"branch":"refs/heads/stable/victoria"},"d32c69e66509e5602f32d8cfac84f3dc12919b08":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-06-09 08:41:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/14"}}},"commit":{"parents":[{"commit":"c916ec1eb409f1d111b47a0c187e5fbd81acc9e4","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c916ec1eb409f1d111b47a0c187e5fbd81acc9e4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-09 08:41:27.000000000","tz":0},"subject":"WIP: Add support for CentOS 8 Stream","message":"WIP: Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d32c69e66509e5602f32d8cfac84f3dc12919b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d32c69e66509e5602f32d8cfac84f3dc12919b08"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"139c36bc3f6caed331a86bbd2e195aaf492542a2":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-06-17 06:28:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/15"}}},"commit":{"parents":[{"commit":"d52bb8e7b94e532a3b8c7e9c5ce4560a8824a26d","subject":"Pin td-agent to 4.0.* to fix missing logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d52bb8e7b94e532a3b8c7e9c5ce4560a8824a26d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-17 06:28:41.000000000","tz":0},"subject":"WIP: Add support for CentOS 8 Stream","message":"WIP: Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/139c36bc3f6caed331a86bbd2e195aaf492542a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/139c36bc3f6caed331a86bbd2e195aaf492542a2"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"d655584a84cd8d80110b473563371ad5a3c7b2eb":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2021-06-18 07:10:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/16"}}},"commit":{"parents":[{"commit":"d52bb8e7b94e532a3b8c7e9c5ce4560a8824a26d","subject":"Pin td-agent to 4.0.* to fix missing logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d52bb8e7b94e532a3b8c7e9c5ce4560a8824a26d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-18 07:10:49.000000000","tz":0},"subject":"Add support for CentOS 8 Stream","message":"Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d655584a84cd8d80110b473563371ad5a3c7b2eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d655584a84cd8d80110b473563371ad5a3c7b2eb"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"},"19cba434ee4848579e36cfaba0ce73fd070d0dfb":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2021-06-24 08:38:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/17"}}},"commit":{"parents":[{"commit":"32eb32f1b66416a2ccd385bd4afee765bf1eb712","subject":"horizon: Correct location of monitoring_policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32eb32f1b66416a2ccd385bd4afee765bf1eb712"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-24 08:38:19.000000000","tz":0},"subject":"Add support for CentOS 8 Stream","message":"Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19cba434ee4848579e36cfaba0ce73fd070d0dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19cba434ee4848579e36cfaba0ce73fd070d0dfb"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"8d4f88b748806f1d8ab0610bc117a7fdb39630ed":{"kind":"REWORK","_number":18,"created":"2021-08-02 11:01:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/18"}}},"commit":{"parents":[{"commit":"32eb32f1b66416a2ccd385bd4afee765bf1eb712","subject":"horizon: Correct location of monitoring_policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32eb32f1b66416a2ccd385bd4afee765bf1eb712"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-08-02 11:01:26.000000000","tz":0},"subject":"Add support for CentOS 8 Stream","message":"Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d4f88b748806f1d8ab0610bc117a7fdb39630ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d4f88b748806f1d8ab0610bc117a7fdb39630ed"}]},"branch":"refs/heads/stable/victoria"},"2877c2c40e571fda90ce1352a633b8b95cf9003e":{"kind":"REWORK","_number":19,"created":"2021-08-04 04:33:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/792763/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/792763/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/792763/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/792763/19"}}},"commit":{"parents":[{"commit":"32eb32f1b66416a2ccd385bd4afee765bf1eb712","subject":"horizon: Correct location of monitoring_policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32eb32f1b66416a2ccd385bd4afee765bf1eb712"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-01-28 09:06:00.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-08-04 04:33:42.000000000","tz":120},"subject":"Add support for CentOS 8 Stream","message":"Add support for CentOS 8 Stream\n\nChange-Id: Ibb23c9d0caf115fcbba7bd322ecebe8d80b6821a\n(cherry picked from commit be544f4e66a6d0dbaa0c26e4565ccfe6a2548838)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2877c2c40e571fda90ce1352a633b8b95cf9003e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2877c2c40e571fda90ce1352a633b8b95cf9003e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
