)]}'
{"id":"openstack%2Fkolla~792838","triplet_id":"openstack%2Fkolla~master~I4faabdfa9c273fc61b536e6ce88b8d71ab2fc581","project":"openstack/kolla","branch":"master","topic":"bug/1928408","hashtags":[],"change_id":"I4faabdfa9c273fc61b536e6ce88b8d71ab2fc581","subject":"horizon: Correct location of monitoring_policy file","status":"MERGED","created":"2021-05-24 18:02:10.000000000","updated":"2021-06-17 07:53:00.000000000","submitted":"2021-06-17 07:51:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792838-bug/1928408","meta_rev_id":"5fe447be528f4ba9b7aee335dc727e8828bb155c","_number":792838,"virtual_id_number":792838,"owner":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 07:51:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2021-06-17 07:53:00.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-06-17 06:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-17 07:53:00.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2021-06-17 06:30:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-11 06:25:33.000000000","_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 18:03:51.000000000","updated_by":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-24 18:03:51.000000000","updated_by":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-05-25 08:45:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-06-17 06:30:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2af1f9f295eadadfc4f80258657d913c244213d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 18:02:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1256a4db85da1269430d3e7e5467bb3ea8062ca","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 18:02:42.000000000","message":"Assignee added: Scott Shambarger \u003cscott-openstack@shambarger.net\u003e","accounts_in_message":[],"_revision_number":1},{"id":"e860dae865024144cca48941cb64e50f3ab293a7","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 18:03:51.000000000","message":"Patch Set 1: Backport-Candidate+1\n\nHere\u0027s a patch to fix the monitoring_policy in horizon","accounts_in_message":[],"_revision_number":1},{"id":"15c2838fea8f0f378cc9ed5e17c1e69f2ba76bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 19:25:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b52be6e29eae4093b4409a600c01437c : SUCCESS in 45m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ba7c1b0c53844ac28bab96c56e66fcb1 : SUCCESS in 57m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d99e8565c5cd44848280dccdc3f8d661 : SUCCESS in 1h 19m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6ec01df68dab4016af5592bdda8541c9 : SUCCESS in 1h 18m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a2e078a0ed75481c80ec648a0c9291f7 : SUCCESS in 1h 17m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/965d513486d0477995cf2c9ea806f058 : SUCCESS in 40m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1868eea5f1f34ce98cabeb31199aff95 : SUCCESS in 18m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b541596ab30c490f8333b907abb5759d : SUCCESS in 1h 04m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ccabfc10bc540b4b0b35e560f60d95c : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7300591be20d476f84d8e7928425b2f4 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e4642d0264f406b9d43f87d800cd3cf : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3fff5525b4c4eebae1a05d6db32258d : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4a48e884c94a54b620f4bea53b18d5 : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/909337ca15b342198c42e8d093fef3ad : SUCCESS in 21m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/39b1ef93a3f94e3d83af0f1335dc36eb : SUCCESS in 29m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/12bbf8c7a728419eb3b8a25f78d8426c : SUCCESS in 55m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9db750eef509430eaeaf9a6c859e376a : SUCCESS in 52m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/200e92c838cb44bba668baa6fd1d5a8b : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":1},{"id":"93400f32389c7826b3c401aa6c3106548db571f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 20:00:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/40128f218a7c40d98133c2ff2097af70 : FAILURE in 1h 30m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/86734f48dbbc463489b6b73444f67ce2 : SUCCESS in 45m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e8233093a0449df998a6a26cf9e9040 : SUCCESS in 1h 56m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3b6d239c35f48279895ee972e8a00cd : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/175c43f21bc546328ac855add83e38b1 : SUCCESS in 4m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7491c4103c84397907368d53b307bd8 : SUCCESS in 39m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c79f14ac57f89f79307d2ae50da0a7c598208c24","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-25 08:45:10.000000000","message":"Patch Set 1:\n\nCould you add a release note?","accounts_in_message":[],"_revision_number":1},{"id":"7d3d6c53c53e986724e84294025dcac91c3a53e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-25 19:15:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"591da3f5f4d29b97fb12d82cae01c0c2469f789d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 20:32:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e359041dcad04e418fbf70b29fb210c2 : SUCCESS in 24m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/27218393118e48838a9b0473f5b88ca9 : SUCCESS in 42m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/748382617dd341e9a5f16768beef2084 : FAILURE in 47m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/232781811a3d46cfaf14adbfe54d2695 : FAILURE in 50m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c2e9af7517fb48f2ac29026d7a670e98 : FAILURE in 1h 13m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e2a83108fe4a41e492ef7ecf1b129277 : SUCCESS in 32m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0916ee69220f4c7a971d7ebb45cfdf20 : SUCCESS in 20m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/22b518db37bb457ebaad2889e0e7cb29 : SUCCESS in 59m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf52e1930bc047f1a53752f016f76087 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7216fd47a41246e387d866e7af0d477d : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e05bcf9539014a318d8d843bfd2970fc : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2cac0d83f2f469bb99820c2e80bb986 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e660f92f07364cc89e2ebb966d76c012 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/609954b6bd234ab5a49cb754dce786be : SUCCESS in 4m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5dfc654766284d909b5ba24ff329f7ac : SUCCESS in 17m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eab054787d244b74ab5cf002250f9fb0 : SUCCESS in 28m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/84f9cff5200d4e2d97a14ce4e3732f38 : SUCCESS in 57m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bbc3c202192046049466c163abbe0288 : SUCCESS in 39m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/93b2aef0a1034ff0931205626eb1b73a : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":2},{"id":"56fa05039424d2e88cef5b3a328270a04a91f80d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 21:25:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ff2e4dfc5d20407dab5142325fd34cf4 : SUCCESS in 1h 42m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fbbcc5f12c1b4b4fbbf321083d164f70 : SUCCESS in 45m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/de1e6f47956b404bafb140f91b249edf : SUCCESS in 2h 08m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e17a7b049d2048f7bc04dff92737feb9 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b18f50cd9804c689f7bd5e546494acc : SUCCESS in 7m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e426cada2cdc443a9730ca8894dd795e : SUCCESS in 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"502b9b92616703cd15d86a475c8ad05ba247d73f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-27 14:21:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2e1daf70f9c8e1f0f30068da4cd9eb82a0a284e7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-27 14:22:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7f9029f3842b4fe243853f58e860751a502e2d8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:38:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9e21ed113a714b1a83e4538fa3b2afb5 : SUCCESS in 23m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/78e5f918310b47e0a533d23bf8a01da2 : SUCCESS in 1h 00m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f7bfbda34b5242e6854ed5e9929f934a : SUCCESS in 1h 05m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/07f18356c267480c9daa6741d4e30599 : SUCCESS in 1h 09m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/371e31f85e3b44a68edf9e96f611cfb3 : SUCCESS in 1h 14m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f042a59feceb4fd2bd6a3f2afe13e52d : SUCCESS in 38m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1b3874a64c334c079b6b2b2757f771c7 : SUCCESS in 19m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cd94c5c621674f39a910421f81f8209d : SUCCESS in 1h 02m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c011149521954a4aba265cfc8d701916 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/746dbba59424404cbf19a9a766c6a932 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96800e32d6e8477f9e7d58b29b3a45d5 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/896ac5bdd9ef44efbd9f33f6b6027bdd : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/917ed5eec3e34a67b1a52a397959c842 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4359d2faf58d4e0c868980a6bc198c02 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f2e905bde254861a6eaaab2b16e911d : SUCCESS in 23m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7db255f2aaf74249b122c484370d0d61 : SUCCESS in 33m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c04390548547471a9b1b04ad78732e42 : SUCCESS in 58m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c48257b142bf46d39bb33e2141fdaedf : SUCCESS in 1h 00m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/317eb748916541b6b6cb67c7ad2e10a5 : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":2},{"id":"86db5fcbbe4228ef6084f1e525bbe3d78b186ab8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:31:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/00a91f019fb245fba7cbeff328f7fc85 : SUCCESS in 1h 32m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b360a69f76634ac5a9bcf853ad086f73 : SUCCESS in 1h 07m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8d96da4b5b047dbbbeaad9606a21d09 : SUCCESS in 1h 56m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f29e1b55a3ed422990eeb03f1d1fe213 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/94c0d1d5745944279661cf3c91189d84 : SUCCESS in 6m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b62e1f0d15a4f0bbc01f6261cd0ccf0 : SUCCESS in 1h 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"37666927e08783ef811e885270be383e73a0e1e6","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-05-28 15:32:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9507e5beed13f4316b97e8d84e43cd707307e77c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-06-03 00:52:14.000000000","message":"Removed Verified+1 by Zuul (22348)\n","accounts_in_message":[],"_revision_number":2},{"id":"01011144b528487099b83f266675c3a2cf682170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-06-03 00:52:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a5939621cd897872cb786a5b408ab239bc82a1a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 02:01:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4ae3c0e5732e460698968d371c325059 : SUCCESS in 23m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5d1183a9bf4c481289c04bab8c4136fc : SUCCESS in 37m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/78984483b8554a05a94de8408dd48f75 : SUCCESS in 58m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2850219568984fd280d9427870eb63db : SUCCESS in 57m 12s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/625bd7dd0c7a4e6a8dbb50831f06f3fb : SUCCESS in 1h 08m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0328ce2bea444191bdbf2781e237457f : SUCCESS in 36m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ff0352a08d7f4537847cd969203c290f : SUCCESS in 19m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e8f22ddc24024aaab3853261916801f1 : SUCCESS in 55m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd140ebbfbd454bb66c516630fb3415 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c39c0bdfcae46c4980b7c86a55be9ee : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/691a8f9b21bb421aafc7812436aaa329 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c264918f0c5c4c639a969efd6edf0f16 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b49b1f7fddc64170b1f8dba8ab794fa2 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e48cbbe64b2947ec887e17d5998144fa : SUCCESS in 5m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d177f629c0664199ade4d1eac876d85f : SUCCESS in 22m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3dfc2c23c7048a8bd5b607bb734e0d4 : SUCCESS in 46m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dcd04353f1e74342a1f6942c45c64117 : SUCCESS in 54m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc1c8b15ae4f4b57a0f78b4d3c8da72a : SUCCESS in 49m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5af2a0e3176f4e86b95c6ef50580e079 : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":3},{"id":"8f265bf4f7a5fd738ceb969314c528824abe1520","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 02:24:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ca81eaeee3e43b38a1a8c9783357ded : SUCCESS in 49m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/50f644352c8c467cb907802a930f3d68 : SUCCESS in 38m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6931188307e1460a8928f93e848c36bb : SUCCESS in 1h 26m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93e0a03044cf4228b23019c3c85567b3 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d284a3ab1b4e44ffaca8cb8bb0e51f70 : SUCCESS in 7m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e4b2972fd674390a49f45f5ff20d9c7 : SUCCESS in 55m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"81fe40c00785e13c03ff1a7767a6594fd45c330d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-06-11 04:59:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e2558bb1f7ef2b0705757ef6dedf0e7a7aca910e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 06:22:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2d5c59543a344e44b4244f435314741d : SUCCESS in 33m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/804b3f78ff1649f186462002d2351cc3 : SUCCESS in 35m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a054920d56084e5d84befe306d44a680 : SUCCESS in 1h 15m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b800c46d378d49a0b464a22ed4f29486 : SUCCESS in 1h 02m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d2da505afc3c4ed5811bb696ce4c55d8 : SUCCESS in 1h 06m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/408b40b669bf41cd83e04c63359faf31 : SUCCESS in 28m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0358270a6d324e3598547d6e69064f7d : SUCCESS in 15m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0830a40544a64123b7a0d2787e6c9f68 : SUCCESS in 56m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0270079a154468b9f9c21aa023f137c : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f14b65b9aca4c8db8d1094f90533a43 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f40dcfa0b27c4e8bb7515d466ed2f739 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3f27b692e804d0c944c05f7c301c032 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcbd7e4409dd47988a4cff3c192329d5 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f05451d816fb46c69b512003e5a7aa3f : SUCCESS in 4m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/30fdcf3c200b4f27adaa3b910d057d93 : SUCCESS in 20m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f8aef679d37445e7a00b7311bddcefb7 : SUCCESS in 35m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b28e41b4a6df44d585921cf51b6f1b5a : SUCCESS in 48m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a8dc9d8d7054e78868b610ea09e08b4 : SUCCESS in 1h 04m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/89be3aa4bb05439db096664817600106 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":4},{"id":"ee564b7bc252ab21993c5c122481225b3e5011aa","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-06-11 06:25:33.000000000","message":"Patch Set 4: Backport-Candidate+1\n\nAnything else required for review?","accounts_in_message":[],"_revision_number":4},{"id":"5ac6de16da8446eff65d00cd16446e0622160719","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 06:58:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b53615cbb6341cfa6f8d4fa73a6305d : SUCCESS in 1h 10m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7ef7103da784150905d20e15f990f77 : SUCCESS in 1h 14m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/98f6af6fd95a4bbb93d660c3637f85e5 : SUCCESS in 1h 50m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/641d5942191f4f13b35a4bfe02c7b5b3 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c906ebf30c84e73823d31817a21859f : SUCCESS in 6m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ea5ca936a464735a3ca5e1248da8e95 : SUCCESS in 1h 00m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"20cd154f1378bfc6487a52409a13ef9058e66336","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-11 19:56:31.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Backport-Candidate+1\n\u003e \n\u003e Anything else required for review?\n\nMore reviewers or more time! Help is always appreciated :)","accounts_in_message":[],"_revision_number":4},{"id":"1cd7a5d6ae97c62091c8af8a05b9643a687354dd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-17 06:30:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"20bee9b4cfd2f3dbd3c20f07065cbe6b92330f22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 06:30:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0fd110563e1420400e4948174a0802c06d41abfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:51:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/257a582997fe43fdb03c061037345484 : SUCCESS in 57m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c599dda19ed4459aa299bf72d4cb11be : SUCCESS in 1h 07m 48s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0abf547694394a74991583968b986729 : SUCCESS in 1h 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c8c12e8c7346058788e327a6e10b79 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8e20237942643d4b3cec41eddf297a7 : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2a1afee3a0241868c65af3ab17271e8 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceacf80339684dd0825a321f3afc6d52 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2966b02e1a14cbeafc9765a38880057 : SUCCESS in 6m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d2ce2ad7b8d48f693d8609548519881 : SUCCESS in 43m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2446011fd09e4950aa2aaf321b0d9a04 : SUCCESS in 1h 02m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/065bb7d6f54e408a903c87ed87a9fce6 : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":4},{"id":"d5a462ce3d52af3132e2fb4de417f509ac368b27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:51:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"dbbe9bc5a39ab1c3354fc1baa242362604f1d0b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:53:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8860f0ca12504913984391e14b664c14 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71bf686828924f9c8ba2a6e2b9d394e6 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f68dfb88fdbad51322727575457b3d78a70466b1","revisions":{"f97d92e93f1035670d767056d8a466774ab99e84":{"kind":"REWORK","_number":1,"created":"2021-05-24 18:02:10.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/38/792838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/792838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/792838/1"}}},"commit":{"parents":[{"commit":"a80de4a7926ab9188191505a86c24fe6db3fbb75","subject":"Merge \"Install Monasca datasource plugin for Grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a80de4a7926ab9188191505a86c24fe6db3fbb75"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 17:51:23.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 17:51:23.000000000","tz":-420},"subject":"horizon: Correct location of monitoring_policy file","message":"horizon: Correct location of monitoring_policy file\n\nPatch to correctly copy monitoring_policy.json into\n/etc/openstack-dashboard.  Policy was misplaced, and not being enforced.\n\nNOTE: current policy enforcement will, by default, hide the Monitoring\nmenu from admin.\n\nCloses-Bug: #1928408\nChange-Id: I4faabdfa9c273fc61b536e6ce88b8d71ab2fc581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f97d92e93f1035670d767056d8a466774ab99e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f97d92e93f1035670d767056d8a466774ab99e84"}]},"branch":"refs/heads/master"},"9ff488c40882c2a32d4db82485e17b5e45157782":{"kind":"REWORK","_number":2,"created":"2021-05-25 19:15:08.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/38/792838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/792838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/792838/2"}}},"commit":{"parents":[{"commit":"a80de4a7926ab9188191505a86c24fe6db3fbb75","subject":"Merge \"Install Monasca datasource plugin for Grafana\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a80de4a7926ab9188191505a86c24fe6db3fbb75"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 17:51:23.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-25 19:13:21.000000000","tz":-420},"subject":"horizon: Correct location of monitoring_policy file","message":"horizon: Correct location of monitoring_policy file\n\nPatch to correctly copy monitoring_policy.json into\n/etc/openstack-dashboard.  Policy was misplaced, and not being enforced.\n\nNote that by current default policy, admin doesn\u0027t not have Monitoring\naccess.\n\nCloses-Bug: #1928408\nChange-Id: I4faabdfa9c273fc61b536e6ce88b8d71ab2fc581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ff488c40882c2a32d4db82485e17b5e45157782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ff488c40882c2a32d4db82485e17b5e45157782"}]},"branch":"refs/heads/master"},"a5ee403af21a63cb306e274a4c3cb744d1d13efb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-03 00:52:34.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/38/792838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/792838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/792838/3"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 17:51:23.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-06-03 00:52:34.000000000","tz":0},"subject":"horizon: Correct location of monitoring_policy file","message":"horizon: Correct location of monitoring_policy file\n\nPatch to correctly copy monitoring_policy.json into\n/etc/openstack-dashboard.  Policy was misplaced, and not being enforced.\n\nNote that by current default policy, admin doesn\u0027t not have Monitoring\naccess.\n\nCloses-Bug: #1928408\nChange-Id: I4faabdfa9c273fc61b536e6ce88b8d71ab2fc581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5ee403af21a63cb306e274a4c3cb744d1d13efb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5ee403af21a63cb306e274a4c3cb744d1d13efb"}]},"branch":"refs/heads/master","description":"Rebase"},"f68dfb88fdbad51322727575457b3d78a70466b1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-11 04:59:42.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/38/792838/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/792838/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/792838/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/792838/4"}}},"commit":{"parents":[{"commit":"b9b60f62621e84e4861b526078b62d5636518be1","subject":"Merge \"Enable Debian updates repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9b60f62621e84e4861b526078b62d5636518be1"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 17:51:23.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-06-11 04:59:42.000000000","tz":0},"subject":"horizon: Correct location of monitoring_policy file","message":"horizon: Correct location of monitoring_policy file\n\nPatch to correctly copy monitoring_policy.json into\n/etc/openstack-dashboard.  Policy was misplaced, and not being enforced.\n\nNote that by current default policy, admin doesn\u0027t not have Monitoring\naccess.\n\nCloses-Bug: #1928408\nChange-Id: I4faabdfa9c273fc61b536e6ce88b8d71ab2fc581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f68dfb88fdbad51322727575457b3d78a70466b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f68dfb88fdbad51322727575457b3d78a70466b1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
