)]}'
{"id":"openstack%2Fkolla~793027","triplet_id":"openstack%2Fkolla~master~I4340ac228dc9624e59ed613265e47c3e95d2dc50","project":"openstack/kolla","branch":"master","topic":"feat/sort-openstack-base-packages","hashtags":[],"change_id":"I4340ac228dc9624e59ed613265e47c3e95d2dc50","subject":"Sort the package lists in openstack-base alphabetically","status":"MERGED","created":"2021-05-25 18:32:46.000000000","updated":"2021-06-05 16:58:14.000000000","submitted":"2021-06-05 16:57:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793027-feat/sort-openstack-base-packages","meta_rev_id":"bb29b84147e61da96b8bdcc865b88e5eb53e5325","_number":793027,"virtual_id_number":793027,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-05 16:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-03 09:58:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-05 14:19:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-06-05 14:19:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-25 19:36:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-03 09:58:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-05 14:19:36.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"b2b6d9802a9d535d5743c76efa4bcedb9d9cdc85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2021-05-25 18:32:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c6c7a721e6df79da22032a909be20de87dfc070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 19:36:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a4bfd4f3665f4b418c705913c26e03ee : FAILURE in 43m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/aa357bcdf2bc4df587f4a3eabc5ba59b : FAILURE in 34m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f3e9017e40814854a5e4f5d3923fa458 : FAILURE in 27m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c2661f4cb7664edd8b6b6ea4a8bc7219 : FAILURE in 24m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/409619c72cb748b18975803a3aca4a77 : FAILURE in 28m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3052d049587f471b81d1190216e1aab8 : SUCCESS in 35m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6bbfd49185414644837a82a57159f038 : SUCCESS in 19m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/74b3c5c1a82a4830b1732fc93d7430b6 : SUCCESS in 57m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8212ea62f8843e3930a34ececeff0c5 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09932fc91bca48a28e56e2b3c0ccc968 : SUCCESS in 32m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8ac35157cea4b35bd0d8d8461d02efb : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef26bb641d264529bcf07cbba580a901 : SUCCESS in 31m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed775b026f664c5a97eaa9117c027836 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5aeeaef09d684a8282e9edfd63bc82a1 : SUCCESS in 20m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4f5b11406904abab0084a9ff26d471e : SUCCESS in 35m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/61793d3f48d942548a682418a40f47c5 : SUCCESS in 53m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/36a334d567184c6f8ef953675399d773 : SUCCESS in 51m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d86b35459d12442e812530f9c0f1c03e : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":1},{"id":"483f25c01b83b0515b5f685cb18b8248594ebf79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 20:10:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/74b0ef657919451796d6444be163db75 : SUCCESS in 1h 35m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a03a4c3b504e4f718680f22d18ed8575 : SUCCESS in 42m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b219f336dae04863bb471158cca419ba : SUCCESS in 1h 22m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4f1276755484bd8a873b56a680bf4af : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a0cb12120bc4a6a80be909b822d1698 : SUCCESS in 7m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/959e769c65504c89ad448b78a14fc90a : SUCCESS in 1h 02m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30e3e3889bb7ae02ac0c336bec2c726a0a2fdf31","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-03 09:58:14.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"951ca2f49378e18af8a537b26104ca39bc9fcf38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 11:12:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8263f99e3a254de1bac20ea3abaf2eef : SUCCESS in 28m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0320ef5b505e43d7a9f827e0d0c918a3 : SUCCESS in 37m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e8d58fbf21774428a4652c81a26c9543 : SUCCESS in 54m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fa552ee033ff48fbb23024dc5b1c9e1f : SUCCESS in 59m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b2eebc0e10214453b9c735f82da325a8 : FAILURE in 37m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a1a80c0db074c5992451118925d6430 : SUCCESS in 41m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8b951a1a98444fa5af0bd9f823ca2169 : SUCCESS in 22m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/01aab6c0fc824a2f9e9edd8e6d5c7377 : SUCCESS in 57m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bbbfd358f0d4b0fb0b974231d72450c : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c33f8a74a16e4fb089d442a9fc258610 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/990dcda18fc34729b5da9f159c44fde8 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22326ff4d59d48af8fd9f9b4f5ca9b9d : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28690a0d007e43afa594556f39c9c416 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9bf2d45a897b4c5f94ff470fa5037570 : SUCCESS in 18m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/126237c667344bb081e1ade20fd8d4f8 : SUCCESS in 34m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6933409ed1b4465faecdd3830bbaa085 : SUCCESS in 1h 12m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2525925ebbc44624bfbbb1560c84403a : SUCCESS in 55m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4eb995e24d50448ea701a15790581794 : SUCCESS in 1h 00m 07s","accounts_in_message":[],"_revision_number":1},{"id":"683f40f787f313af980f732069c1fea48266f7f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 11:46:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9e2a9846d474310ac5075cdfc9e71cb : SUCCESS in 1h 04m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ccd2f6302fcf4747ac0d7beaa98bc058 : SUCCESS in 1h 05m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/65adb184996046e18fc6a85939dc2a1e : SUCCESS in 1h 35m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9e55432071e42e38d74c52a8024ff24 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f32c5859b2044775ad432e66b4cb37b3 : SUCCESS in 5m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/73f9f8bab25644c2a0fac5dbc977719e : SUCCESS in 59m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bec2b5ed8e9ac3f8e4e180714c8d1c33abc4a83","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-05 14:19:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bee9a6f91c50c84eb3969a1edb3e1499e17f92d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 15:37:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0d2ce65bd2034e69bd29058908aef987 : SUCCESS in 31m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3541f0dcf01749fe8d7cf5bd8d71b7f9 : SUCCESS in 44m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cb78745e967e44ec9d10737490b4214d : SUCCESS in 1h 06m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/30af356544fb4f3c8218ff5999fa3785 : SUCCESS in 1h 02m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bdea23612ede4eb7b6d39b83da42afe5 : SUCCESS in 1h 14m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f544e9d732bd4f049edf96b219b15de7 : SUCCESS in 28m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3b0aada5177c4d3f99117e0d1222a502 : SUCCESS in 21m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2460d6d68bda4a0a933218a528f639e6 : SUCCESS in 58m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63d5142121a24a528edbfebc8a2a90e1 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea95bb0afc3142f9b896380f6ace5a57 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d6d1f1bb6b941de84c6b64041318219 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73c73a78b5ed47ba8b22967b01ae5e1e : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aed05b33e0ab4a799b088283dd7f032d : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/78c06ad18fd44b319945733b3951988a : SUCCESS in 15m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a9a7f6279e14bbe8eea9476663f4001 : SUCCESS in 34m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d228e950acb4ea3a3a93102683c6a2d : SUCCESS in 36m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e2a92092ef7450dbc8e6ab01f9ef696 : SUCCESS in 36m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/656406ae5d9f4daa818dfeb1945c86fd : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e85d0ebb514723b917511393eddec62bc66d9f0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 15:37:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a81ffa92a1900289e9c504231fefabc335b6b094","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 15:53:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c8b4d6690704ab09c74f4a536188e03 : SUCCESS in 49m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7718e3a985374a3abd7b56e13d72f13b : SUCCESS in 48m 33s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/30644123064d4a319314b049c7a96566 : SUCCESS in 1h 23m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4dd08cdcf4345f2baff31c932a03b39 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8dbc30c2680048cc84c01792030820be : SUCCESS in 6m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/766fb7f7e77248718595c52e8ef49d32 : SUCCESS in 1h 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"752fdd1118060c0dd708f396ce1241dd8eb7bacf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 16:57:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/31ba56dbd97d480dba4eb32b9ffc16f8 : SUCCESS in 1h 18m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a67e52c1bf8949588d3b244be1521fd5 : SUCCESS in 1h 07m 58s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e66ee8e039b140a9b5833aaa78af4a3d : SUCCESS in 1h 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ddae1bbe8de4b53935231cbc7158904 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f825ac35e75f4a279a8bdc13bcf4d297 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e1f8c87363f499e876facc181e2b254 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc86e08d1c504cfdac7d506e58467bae : SUCCESS in 6m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec9ef5a2ba5a47fe8166e8597f42ad0e : SUCCESS in 31m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3c9d582e30945f1a667e67ec2778e59 : SUCCESS in 54m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3df9c3de16c74925be8556c8ef533dee : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":1},{"id":"62e254525c016808cc5e466ae6116fcbe337c1e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 16:57:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bb29b84147e61da96b8bdcc865b88e5eb53e5325","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 16:58:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e701f86953be46a996ea50cebf94e793 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c82683a9ef8e807f0561f07a9110432c99e848d4","revisions":{"c82683a9ef8e807f0561f07a9110432c99e848d4":{"kind":"REWORK","_number":1,"created":"2021-05-25 18:32:46.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/27/793027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/793027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/793027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/793027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/793027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/793027/1"}}},"commit":{"parents":[{"commit":"1167be2d2f1ba91cdb6ece78d785b6721dc8e6c9","subject":"Merge \"add openstack-python3-xena-jobs-arm64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1167be2d2f1ba91cdb6ece78d785b6721dc8e6c9"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-05-25 18:32:10.000000000","tz":120},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2021-05-25 18:32:10.000000000","tz":120},"subject":"Sort the package lists in openstack-base alphabetically","message":"Sort the package lists in openstack-base alphabetically\n\nSigned-off-by: Christian Berendt \u003cberendt@betacloud-solutions.de\u003e\nChange-Id: I4340ac228dc9624e59ed613265e47c3e95d2dc50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c82683a9ef8e807f0561f07a9110432c99e848d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c82683a9ef8e807f0561f07a9110432c99e848d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
