)]}'
{"id":"openstack%2Fkolla~794153","triplet_id":"openstack%2Fkolla~master~I0b3eb6744a34a601266179fb358510a5c500e6ba","project":"openstack/kolla","branch":"master","topic":"pip_local","hashtags":[],"change_id":"I0b3eb6744a34a601266179fb358510a5c500e6ba","subject":"Fix _get_images_dir when pip-installed with --user","status":"MERGED","created":"2021-06-02 07:54:46.000000000","updated":"2021-06-08 08:04:37.000000000","submitted":"2021-06-08 08:02:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"794153-pip_local","meta_rev_id":"c06c3b2ba348f4d410d917d2021ddc27454d2474","_number":794153,"virtual_id_number":794153,"owner":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":33546,"name":"adam aquesbi","email":"adam.aquesbi@objectif-libre.com","username":"aaquesbi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-08 08:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"recommended":{"_account_id":33546,"name":"adam aquesbi","email":"adam.aquesbi@objectif-libre.com","username":"aaquesbi"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-06-07 14:38:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33546,"name":"adam aquesbi","email":"adam.aquesbi@objectif-libre.com","username":"aaquesbi"},{"value":2,"date":"2021-06-07 09:58:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-08 06:23:11.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":33546,"name":"adam aquesbi","email":"adam.aquesbi@objectif-libre.com","username":"aaquesbi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-08 06:23:11.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-06-08 08:04:37.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":33546,"name":"adam aquesbi","email":"adam.aquesbi@objectif-libre.com","username":"aaquesbi"},{"value":1,"date":"2021-06-07 09:58:50.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":33546,"name":"adam aquesbi","email":"adam.aquesbi@objectif-libre.com","username":"aaquesbi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":33546,"name":"adam aquesbi","email":"adam.aquesbi@objectif-libre.com","username":"aaquesbi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-02 08:07:29.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-06-02 09:20:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 09:37:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-06-07 14:38:55.000000000","updated_by":{"_account_id":33546,"name":"adam aquesbi","email":"adam.aquesbi@objectif-libre.com","username":"aaquesbi"},"reviewer":{"_account_id":33546,"name":"adam aquesbi","email":"adam.aquesbi@objectif-libre.com","username":"aaquesbi"},"state":"REVIEWER"},{"updated":"2021-06-08 06:23:11.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"71c9b8a78b25dcb1ccf2362e3d9744999bb99899","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"date":"2021-06-02 07:54:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c89fe7d1e51fc017d032bd1614afcebdac5dd86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"date":"2021-06-02 08:00:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c583b4d93bf7aca3cc6806e152d541b0e8be0666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"date":"2021-06-02 08:01:08.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"18a0a6ba11a757d69fd5c789652648255419f6c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"date":"2021-06-02 08:03:17.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"dfba92d0328fc607c2b83f64415c18dc4ca8296f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"date":"2021-06-02 08:06:02.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f5a967b844f19d75b0d36afc87a67c9902a17dd3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-06-02 08:07:29.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"b973d5ebc1629288a70d28d13458da15d01d151f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 09:20:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/397e6c4162eb4747a8d938a2e5f76a79 : SUCCESS in 24m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fda364a812ea4681a3c296321ce90ca4 : SUCCESS in 33m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/39651c3403204c0dba688a2f9143bcdc : SUCCESS in 1h 07m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/115982749a584231a50668eb36f53ea4 : SUCCESS in 51m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2fc4e48ddd494e0d9924bb2af36f0d93 : SUCCESS in 1h 13m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/57bf429cb04242b6b4a098f35f7f2c4b : SUCCESS in 31m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bff4fab75b38423c994f2a663f87d061 : SUCCESS in 19m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7cb20614d00c4e8284da05a74816e90d : SUCCESS in 1h 06m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ac29b879d94a8584933b0de0b2fc43 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8f98d7f9b39478a872caaa709f6f47e : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9c24479a08142caafc349099163a53f : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/894e121a859d4d108b87c2257e015266 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888ff0add13e40b3b2da4f6e4091d955 : SUCCESS in 6m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49053f1702d94575a2687f15ad68df36 : SUCCESS in 20m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e358481aeb0c4382a22703ad3c7d9227 : SUCCESS in 41m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36f34220c3da447d92cf0175749c90a4 : SUCCESS in 1h 01m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/00a7d2d95a57470b960d298dd9f68355 : SUCCESS in 56m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/938e9111ca6b47c7a27f92e5f75a6aa6 : FAILURE in 15m 50s","accounts_in_message":[],"_revision_number":5},{"id":"0f63e13d49c7a2f28d52342f7a3fbbe1b5638d47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 10:14:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0782e3deb6a14cef858097cb8501ee8a : FAILURE in 1h 44m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/530131b03b26483ab131bbe740aa05e1 : FAILURE in 1h 40m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fcf22d517a23426ba6bbebaca16f7706 : SUCCESS in 2h 02m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bedd8b1f24cd4033b7f099f34b641b7e : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5232d8472614ffe8e6a18fe66266306 : SUCCESS in 7m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b2994ee168d44d5b6a61542d70ba731 : SUCCESS in 1h 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ecf3c0dde7932169021a9d4f0aaa1d21f0694a0f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-07 09:37:48.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nPlease add a release note as well: https://docs.openstack.org/kolla/latest/contributor/release-notes.html","accounts_in_message":[],"_revision_number":5},{"id":"feb657b8a80ce402c3f2703145aeb339a4a2b4fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"date":"2021-06-07 09:50:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fdd2800635af1ac5cac22f7e66cb8a12cd541b96","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-07 09:53:50.000000000","message":"Patch Set 6: Code-Review-1\n\nplease add the fallback value I mentioned in the inline comment","accounts_in_message":[],"_revision_number":6},{"id":"5c6267afb4fe8546be9286a47aae723ceea0cddc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-07 09:54:08.000000000","message":"Patch Set 6:\n\nand the reno is fine, thanks!","accounts_in_message":[],"_revision_number":6},{"id":"9510db7d842c8f22207877dce9d76b76f982094d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"date":"2021-06-07 09:56:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"93451702f88a96b5836f8a525067fa7406ab0d0b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-07 09:58:39.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"07739b9de3fafe2f28c0e8c4c4f5ad5f64d21638","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-07 09:58:50.000000000","message":"Patch Set 7: Backport-Candidate+2","accounts_in_message":[],"_revision_number":7},{"id":"708dc88dad37fe19b463972de2f23ef2702e3dd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 11:06:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/44dfd398bb4d411989646b5b0c4d87b2 : SUCCESS in 30m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/558ad9bf9ac046d8858369fef9b8db20 : SUCCESS in 51m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5643c64306704adab6c6ca17c1cb8ef0 : SUCCESS in 58m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/49152ef674ca444b96ad09200c848f77 : SUCCESS in 46m 54s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/47c409ea51a14be19220451e99ac60c0 : SUCCESS in 56m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4c499f349d174ce38cd5d1a657d625aa : SUCCESS in 42m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/382fd1835faf455dad3e608a044dfe66 : FAILURE in 30m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b1948922231a4a2ab54003529da83da7 : SUCCESS in 1h 00m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e1efc8c26944ea7b6cebf44aff7888e : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8338564619d54ec390781541e3766233 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f777262e0cdb45c9b404c3de03fbe6c1 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dcad9d254ed4312bd86f98293c85cb0 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef9594d1c194fa990b4fa37e715203c : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43490a208d464f029267ba89c6ab4f98 : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/72aaf83ad2f24e3784f04a0b9962e9bf : SUCCESS in 26m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff54132a647545f198c7058fbdb0e2cc : SUCCESS in 49m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2edbb620b91645b89c6e9e96e349f67d : SUCCESS in 57m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b921db482c7e40f39b553d45ce45a7ac : SUCCESS in 53m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c56cd9a717fa48a9a474659c1e1116e9 : SUCCESS in 1h 08m 11s","accounts_in_message":[],"_revision_number":7},{"id":"6a5e040f3777ad5ff5b733e881da7c715fee6b5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 12:28:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/07b935db20984175b0413668b2456db8 : SUCCESS in 2h 30m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a4bfe4c79a034dce8053bac801fdb1ad : SUCCESS in 1h 28m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/869557b214a14c82b0d676f78b14a424 : SUCCESS in 2h 00m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d9334932f1a474c94a7e3f9ed0f6371 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5014b75bfb349cea615b76a36c9dbc2 : SUCCESS in 8m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5280bda6bd74f6ab19d76b53573d1b7 : SUCCESS in 1h 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5dea8fb5182d02fc3786cae62b4fef475bc4b8ea","author":{"_account_id":33546,"name":"adam aquesbi","email":"adam.aquesbi@objectif-libre.com","username":"aaquesbi"},"date":"2021-06-07 14:38:55.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"4d0a1a1f443b5acc57ea7b2d5dea7e6c9ac37ed3","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-06-08 06:23:11.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"79366c108d37a6e12febab3e84b7de9b35e8b160","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 06:23:29.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1e6ab11f56957140fefb542abdadf65f9d54806c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 08:02:46.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/61ef2a54f766444c9aad217a99c70c23 : SUCCESS in 1h 07m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cb29d6a16c464ceca77e40c73fe7194f : SUCCESS in 1h 25m 23s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ffb6a517cef440f48dffe898f761c1b8 : SUCCESS in 1h 35m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72cea38e09f1408bafa944d5f3d7cf44 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34aacabda8da4e80bfe02058a383f346 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a126c670a0c946a6b0e4b51b060f795b : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54e17eb141540dba78dfb9d9cc7f5df : SUCCESS in 20m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/678a8c9a39024438ba9a1858563e3e6a : SUCCESS in 5m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/01dc76a071014f4d83775eae76642d66 : SUCCESS in 33m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b8f166507e54e04b9d5eca960e15707 : SUCCESS in 59m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bcf506d4ccc84dd3b7184e7c91c33c9f : SUCCESS in 1h 08m 21s","accounts_in_message":[],"_revision_number":7},{"id":"dec993a29955e16911af1d938e867a8cd18fdb08","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 08:02:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"0206ca6b564021d294e4c4d7f696c7604c961638","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 08:04:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa4394232e3408da80298a601d8db34 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dc937d7d41d4933b9231c47b8027904 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"db78657c231663f3e8ebbc18de43c06ddaa94c97","revisions":{"343ef810e4259c04ec18eede16ba2da0d81a347c":{"kind":"REWORK","_number":1,"created":"2021-06-02 07:54:46.000000000","uploader":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"ref":"refs/changes/53/794153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/794153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/794153/1"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"jpic","email":"jpic@yourlabs.org","date":"2021-06-02 07:42:35.000000000","tz":120},"committer":{"name":"jpic","email":"jpic@yourlabs.org","date":"2021-06-02 07:44:10.000000000","tz":120},"subject":"Fix _get_images_dir when pip-installed with --user","message":"Fix _get_images_dir when pip-installed with --user\n\nChange-Id: I0b3eb6744a34a601266179fb358510a5c500e6ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/343ef810e4259c04ec18eede16ba2da0d81a347c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/343ef810e4259c04ec18eede16ba2da0d81a347c"}]},"branch":"refs/heads/master"},"7d22f43f435e15c787f324b5b58e3d878449a5d2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-02 08:00:35.000000000","uploader":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"ref":"refs/changes/53/794153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/794153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/794153/2"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"jpic","email":"jpic@yourlabs.org","date":"2021-06-02 07:42:35.000000000","tz":120},"committer":{"name":"jpic","email":"jpic@yourlabs.org","date":"2021-06-02 07:51:34.000000000","tz":120},"subject":"Fix _get_images_dir when pip-installed with --user","message":"Fix _get_images_dir when pip-installed with --user\n\nCloses: #1930544\n\nChange-Id: I0b3eb6744a34a601266179fb358510a5c500e6ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d22f43f435e15c787f324b5b58e3d878449a5d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d22f43f435e15c787f324b5b58e3d878449a5d2"}]},"branch":"refs/heads/master"},"4c371aa9e4456967b82a205e8b0cb2b915e31501":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-06-02 08:01:08.000000000","uploader":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"ref":"refs/changes/53/794153/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/794153/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/794153/3"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"jpic","email":"jpic@yourlabs.org","date":"2021-06-02 07:42:35.000000000","tz":120},"committer":{"name":"jpic","email":"jpic@yourlabs.org","date":"2021-06-02 07:52:08.000000000","tz":120},"subject":"Fix _get_images_dir when pip-installed with --user","message":"Fix _get_images_dir when pip-installed with --user\n\nCloses-Bug: #1930544\n\nChange-Id: I0b3eb6744a34a601266179fb358510a5c500e6ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c371aa9e4456967b82a205e8b0cb2b915e31501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c371aa9e4456967b82a205e8b0cb2b915e31501"}]},"branch":"refs/heads/master"},"34c035d2d57ec9df5dba2720feddd3890726daf9":{"kind":"NO_CHANGE","_number":4,"created":"2021-06-02 08:03:17.000000000","uploader":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"ref":"refs/changes/53/794153/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/794153/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/794153/4"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"James Pic","email":"jpic@yourlabs.org","date":"2021-06-02 07:42:35.000000000","tz":120},"committer":{"name":"jpic","email":"jpic@yourlabs.org","date":"2021-06-02 07:54:18.000000000","tz":120},"subject":"Fix _get_images_dir when pip-installed with --user","message":"Fix _get_images_dir when pip-installed with --user\n\nCloses-Bug: #1930544\n\nChange-Id: I0b3eb6744a34a601266179fb358510a5c500e6ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34c035d2d57ec9df5dba2720feddd3890726daf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34c035d2d57ec9df5dba2720feddd3890726daf9"}]},"branch":"refs/heads/master"},"a264027fcbcb2ca0ddece4a9c1b34a535dd899f3":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-06-02 08:06:02.000000000","uploader":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"ref":"refs/changes/53/794153/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/794153/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/794153/5"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"James Pic","email":"jpic@yourlabs.org","date":"2021-06-02 07:42:35.000000000","tz":120},"committer":{"name":"jpic","email":"jpic@yourlabs.org","date":"2021-06-02 07:55:48.000000000","tz":120},"subject":"Fix _get_images_dir when pip-installed with --user","message":"Fix _get_images_dir when pip-installed with --user\n\nThere are two cases where pip installs kolla share into ~/.local/share/kolla:\n\n- with `pip install --user kolla`\n- with `pip install kolla` as non-root\n\nThis patch helps kolla find the share dir in these cases.\n\nCloses-Bug: #1930544\n\nChange-Id: I0b3eb6744a34a601266179fb358510a5c500e6ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a264027fcbcb2ca0ddece4a9c1b34a535dd899f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a264027fcbcb2ca0ddece4a9c1b34a535dd899f3"}]},"branch":"refs/heads/master"},"732d46b318f936c7a64939d06a5cf0398a5f88c3":{"kind":"REWORK","_number":6,"created":"2021-06-07 09:50:29.000000000","uploader":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"ref":"refs/changes/53/794153/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/794153/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/794153/6"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"James Pic","email":"jpic@yourlabs.org","date":"2021-06-02 07:42:35.000000000","tz":120},"committer":{"name":"jpic","email":"jpic@yourlabs.org","date":"2021-06-07 09:39:56.000000000","tz":120},"subject":"Fix _get_images_dir when pip-installed with --user","message":"Fix _get_images_dir when pip-installed with --user\n\nThere are two cases where pip installs kolla share into ~/.local/share/kolla:\n\n- with `pip install --user kolla`\n- with `pip install kolla` as non-root\n\nThis patch helps kolla find the share dir in these cases.\n\nCloses-Bug: #1930544\n\nChange-Id: I0b3eb6744a34a601266179fb358510a5c500e6ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/732d46b318f936c7a64939d06a5cf0398a5f88c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/732d46b318f936c7a64939d06a5cf0398a5f88c3"}]},"branch":"refs/heads/master"},"db78657c231663f3e8ebbc18de43c06ddaa94c97":{"kind":"REWORK","_number":7,"created":"2021-06-07 09:56:03.000000000","uploader":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"ref":"refs/changes/53/794153/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/794153/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/794153/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/794153/7"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"James Pic","email":"jpic@yourlabs.org","date":"2021-06-02 07:42:35.000000000","tz":120},"committer":{"name":"jpic","email":"jpic@yourlabs.org","date":"2021-06-07 09:46:58.000000000","tz":120},"subject":"Fix _get_images_dir when pip-installed with --user","message":"Fix _get_images_dir when pip-installed with --user\n\nThere are two cases where pip installs kolla share into ~/.local/share/kolla:\n\n- with `pip install --user kolla`\n- with `pip install kolla` as non-root\n\nThis patch helps kolla find the share dir in these cases.\n\nCloses-Bug: #1930544\n\nChange-Id: I0b3eb6744a34a601266179fb358510a5c500e6ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db78657c231663f3e8ebbc18de43c06ddaa94c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db78657c231663f3e8ebbc18de43c06ddaa94c97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
