)]}'
{"id":"openstack%2Fkolla~794238","triplet_id":"openstack%2Fkolla~master~I3560c3a2bee4e391cac841b9bcd8b8d2199f920d","project":"openstack/kolla","branch":"master","topic":"remove-rally","hashtags":[],"change_id":"I3560c3a2bee4e391cac841b9bcd8b8d2199f920d","subject":"Remove rally","status":"MERGED","created":"2021-06-02 14:41:00.000000000","updated":"2021-06-21 14:20:04.000000000","submitted":"2021-06-21 14:17:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"794238-remove-rally","meta_rev_id":"8dac0a773333b08fe9df0ab51aa674bb13fc1ec4","_number":794238,"virtual_id_number":794238,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-21 14:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-21 09:12:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-20 13:55:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-06-21 09:12:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-02 16:05:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-11 19:50:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-20 09:53:24.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"7a5d55b89ebf699a6e67e1749265b05867912fbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-06-02 14:41:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82f0cd07094dad5c2d9e63dc065f2dc04fa96207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 16:05:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e0a10f63499b4c43bd26c83cec245aaa : SUCCESS in 3m 30s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b872e9c26d7f45d8b5ccc099255265a2 : SUCCESS in 22m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/89acd136e3ec471887165ee9f1944bc0 : SUCCESS in 42m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a34ee02d21484c8ea12f86b7d0a0f541 : SUCCESS in 1h 18m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/729e757503bb47768dc1fde3e971da99 : SUCCESS in 1h 14m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1ce243d1f983406ea919615bab483292 : SUCCESS in 1h 13m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bf4bd29d470a4466887bd56fd6160aeb : SUCCESS in 32m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ff3735f30a724af4a073af50beacbe23 : SUCCESS in 22m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c0a850e498594579b3cd2648b2fda523 : SUCCESS in 56m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7c6b666723c4c7cbbe0bb6286b351d6 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30ed9ba320ae4e56b8576b9a1963e830 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1231972e902649a591f3ff93c569a20c : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cb390b5f8c446a5b03b63e770ea6a01 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d56196c4064e43af98c9e3e473855fae : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9e2389430424ae5963dec7dc9df2d36 : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8479fefa83854048adf3623656affbfc : SUCCESS in 25m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/429ef0c538e64447bd90234916e6422c : SUCCESS in 44m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3389d2ca44e34e448554e7987184f654 : SUCCESS in 51m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a3ae9120a7bb47f6a334e7c6cc23e998 : SUCCESS in 47m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cb5d3135a1ec4452988d758fda7f0b80 : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5ed7a5a30fe2ddd9c4a0e3d166708fa72a270ab4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 16:19:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f1bfddbb435d4bbe92c15b7f5159fb46 : SUCCESS in 52m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5630117dc25410f94b28671e873f4b8 : SUCCESS in 47m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ff9eca0d0f3d42d5bd01b44fbe301939 : SUCCESS in 1h 30m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75d337b2d1434f84b64db21757740589 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/afdae8f6db7245598fc9acd6fd5810cb : SUCCESS in 6m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3963fa0e85724a7d90331919e7a90533 : SUCCESS in 51m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f000f87f1d367d2a7b15a62a318ef1d9c53daee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-11 19:50:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33f67be078a6aaa87eaabf523a2f3d5d9547685e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-06-17 05:45:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"396e0a328891d3f67d111072e62e41bccd026944","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-06-17 05:46:01.000000000","message":"Patch Set 2:\n\nfix conflict","accounts_in_message":[],"_revision_number":2},{"id":"3f503da5c2b48c1ab87c5f33c2f238783938bcbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:23:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/312c9591d5fe466ba2832ec5d19d317e : SUCCESS in 2m 37s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2f5b96abe6a640059038bb30610326e3 : SUCCESS in 21m 01s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/14cc229f3f4840f78a07728e72e5ee38 : SUCCESS in 46m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9f01e3f8ee304f68943491d1d485fc27 : SUCCESS in 1h 04m 07s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0b51a27a4ee0492fa9f6775901eed30a : SUCCESS in 1h 07m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/40d400344c8a422fbb1ce097ccb8508e : SUCCESS in 1h 15m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5d2e2bd733044da1bbc1d41aa21830e0 : SUCCESS in 29m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7dfa90c5e72a43b0905ecd98bdfbdef1 : SUCCESS in 18m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/54d0e19f17cd4eaaa8a55df4e30425e2 : SUCCESS in 58m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4788d6030c6495fa5895aa3ef6e16eb : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/722b0dcb05e742899b9388afc2f64425 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c7ca75a5cb44e2a8924db47fc0b3d23 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bd1e13cae2240f79af6ee374ffc7416 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb19d05fb980430787dc9e65a9796753 : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21a6b033485846f0b012abaddf272809 : SUCCESS in 3m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9be180de2ab467995e6a54e36d74153 : SUCCESS in 19m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0eb24970ed549fb967715fea9b96d76 : SUCCESS in 32m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c7be6b7239a498986f37dd370a3eddc : SUCCESS in 1h 13m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24b1537483e749b39b5c6b0319aa0c4f : SUCCESS in 49m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/89f77e7a217c4f9e846fc45e3e241405 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e1dabd632520548269ccf66e90f842cebbca0d6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:26:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0201d294651e4ec28b434f4547bfa710 : SUCCESS in 1h 02m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58fb47da1e954f69b180d2ca25f14561 : SUCCESS in 50m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b87564897754888ac81c7c57d2d8b44 : SUCCESS in 1h 36m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bfbffb076904abab89b1cc42d25424c : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8eaeab45ac9b48988e81bb85695a1bcc : SUCCESS in 6m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8fd14d4ead14a288312fafc02c42925 : SUCCESS in 1h 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8168534d269304c2a6f9e4b5f8364caf1a5e3ef7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-20 09:53:24.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4a173185fe17b4f1e55b2caf158bc844337fead4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-06-20 13:20:26.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d1665fb370aacbd5fc7a370cab0bdd22d66b8fe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-06-20 13:22:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da13977e143f422e21b2e6421b274d3007f7c2d7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-20 13:55:02.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0810d3693197251568ae9c065124873960dd5de6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 14:32:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f61797d451894041ae7bc214556a91c2 : SUCCESS in 3m 11s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/93d4b6288ecf461fa33d994e9a2a4c3d : SUCCESS in 34m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7d2ea1eb986c4e59bc1f606cabefcb66 : SUCCESS in 34m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/89e627a3dbb644ff86e95d0e437edb60 : SUCCESS in 1h 02m 05s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0d8fcc51f9b64c13a644e81c336d7bcb : SUCCESS in 1h 05m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d914b7192481440dbc5c108a03673694 : SUCCESS in 1h 08m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9c796c07dbd041709ce006a6766c7f8f : SUCCESS in 30m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9998113b9f8e4401896d6be176895887 : SUCCESS in 20m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/023226e3b7a448d9908b10719aa9dab5 : SUCCESS in 1h 02m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfbcd7a736aa403daa99776dbbf555d9 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54cb156b56774923b8671a5741b79647 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ff000a8678641b5940ffbd0f5d2f86a : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3036ef9b497f446db863192be03b8dc0 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa408a2ff7b425d9d46b6a884cad0de : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c7791cec00e427e9e5afcd28adc7387 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5c40f472d7e74fdb947572abddbe0dc6 : SUCCESS in 19m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/62cc3ce21ac145e69ed016a6ddca6aa4 : FAILURE in 25m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/26de319e41ba4393a3056778e8dfdd0a : SUCCESS in 42m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aa41d59900ea46c38ef78a911163f5db : SUCCESS in 42m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c7847a0ac4384fc09bf957bc376cd4ac : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":3},{"id":"7a29a5a35a8e4e5a7a20bf94236a357ed02bede9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-20 14:40:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"11c9a1a78cbdfee0a739da3c3f5964337fda8d94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 14:53:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b9061b00581e46afb77af619c58e0b1d : SUCCESS in 1h 28m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9697acfc9daf436b8e557442f6e602de : SUCCESS in 36m 56s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/63160b0861cb435f824a5dea6fb1eae8 : SUCCESS in 1h 20m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bce92ad543fc46b5a1a5cc55ca00f599 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b837ca3e691840be86a6729f456c4d11 : SUCCESS in 4m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a0d0da666894f91a70ae20c6c04ed1a : SUCCESS in 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a382bc64101d03e9f349759758d8262bd685d479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 16:02:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a2ffbf294b1b49c19a3933c2428f51a8 : SUCCESS in 2m 37s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a3c083bac89b49f1bca5f7d57d07dea2 : SUCCESS in 24m 23s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/61e707f7552a4ec3ac1982d911371af6 : SUCCESS in 55m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3170df7a853c49d98511e687e5de2add : SUCCESS in 1h 08m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/95135334c17c43f7a213fb89936b0779 : SUCCESS in 1h 03m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bf90dc8452cc40519f422d23d61d92d5 : SUCCESS in 1h 17m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e224cb47c5fb484bb9adaeb192a6fb27 : SUCCESS in 45m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1d7803d0f73b407c8913d18c97e4e238 : SUCCESS in 21m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fd4c381e0ea1452ea8e972fcb89cdeed : SUCCESS in 1h 08m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a7f4c5ce5a45e6989f6d5964a55a74 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44709129762a4db0af300ef616d75458 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/937e1aa7669b4325a40868cddbdfc19f : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bce4d9389104f8bb87897a5ea9f4a03 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e9791247674035bf1eeee89711a760 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1624b8e47cc54c31a88536f4aafaf585 : SUCCESS in 4m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf8b25d3f31f4c4e8dd123c1b6ea51a1 : SUCCESS in 21m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/14686704c69247b19f13a69fc6958435 : SUCCESS in 32m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a49624a3dbde4c9ab2dcfc5637be4b23 : SUCCESS in 50m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/31f5a5b536144dacac2a5f82412146a0 : SUCCESS in 54m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5285836f9b104c7692aa71b36074c649 : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":3},{"id":"4731a456ef399df083fc7eee9cef3e670f56fe6b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-21 09:12:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nBye bye rally","accounts_in_message":[],"_revision_number":3},{"id":"db9310a415272f313c9a7c48975581d919a7af5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 09:13:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ad04b70352ce4e9d45915a9cee1320d361d881c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 10:33:53.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/19275b19a9554afa88a99bf5b1dddcaa : SUCCESS in 4m 18s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/98be0f60636c4286b93151bb4e3ac977 : SUCCESS in 44m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a9adcb78368b43289abfd57fb555824e : SUCCESS in 1h 12m 07s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7edfdcb9dc43448f8993ffd6f467c440 : SUCCESS in 1h 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9780de4a893b47a1b75d8b5471a7c26d : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/622c57e7a75f444d9c6103afff68a42e : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70350bf6de1e4936b05503d62adeb823 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05f0e4498e4473fa294a20ef0feb41e : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bef753c0e154931a1dfe35247d34131 : SUCCESS in 6m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9c1c5dc74994c658af8c20f5fd02163 : FAILURE in 33m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b2d8fac6162245dd8235877d97090c99 : SUCCESS in 51m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9f3e45f6db574e3a91ba5fe98adf3291 : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":3},{"id":"93d6ad23481665fdb91777c5278a3a7eb9cbd03c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-21 10:34:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"42f3464265b56c5e65be2ef375824ebbfa5f2223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:39:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e360047c80e404e9f1a67713a4b7a01 : SUCCESS in 3m 47s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/cd4786b33bb647a48d26326e75157ce4 : SUCCESS in 25m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/29086a410a2046b194ed8935bcf41cc4 : SUCCESS in 38m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/44d75ed9863f475da41d42c3151b84d9 : SUCCESS in 1h 00m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cb0519a2dfbc47219719b6c64d65819f : SUCCESS in 1h 33m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/602973a6006c4a0aa7fe743092617394 : SUCCESS in 1h 53m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/29137e627f8a47128ae8782e19bc7435 : SUCCESS in 1h 05m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1717a47ba22243c4a98cb3ed6c987249 : SUCCESS in 1h 07m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/770d8d74e90c4951b62c63df99398537 : SUCCESS in 1h 28m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c62a304f85d4b599580efeeee57ab75 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4124ba1949243c782172b6ef12e051a : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d951bb90f1c54e2c9b01636b4a66dfb2 : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42d5e49e1bd6405d9ecf3c9cc04c269c : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9aa5e6c21d5447b8260fb45da087a7d : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d27872956794cd6a4f1ea09a8124e79 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c6f7314aa5848ebb9ca45cfa415be00 : SUCCESS in 58m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/93642f343c2d4fe7bf1f2e1f8f8eb2b7 : SUCCESS in 41m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b61bf5956004940aff1b170b5b3cf2c : SUCCESS in 1h 17m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7060b0e292344118bd9f2becb6806cf : SUCCESS in 59m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c1fe1c435533451291d6bd68db260d13 : SUCCESS in 1h 31m 26s","accounts_in_message":[],"_revision_number":3},{"id":"cea1b327f7036b52ef48994faade615f54c97cd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:39:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"54edc250892c1eb2c4331c249f0cbf1865875457","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 13:02:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7259139968b24f338dbbb1d61fb6a344 : FAILURE in 1h 48m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d29b1d6ed054e3fb74b58496a83c867 : SUCCESS in 1h 47m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3e9160bc2624de08531947fa7b594a1 : SUCCESS in 2h 26m 21s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f51ff985774d4e53988a16be2a0a3e90 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6dd2e605cd524017b68655969fe5b2fd : SUCCESS in 7m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f0cd3ecf581468c87e19c2b8206103a : SUCCESS in 1h 50m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5d89eea4e433225b7095c38761e2ca77b7d9c64d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 14:17:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cdeaf99cf4e6448dbcf01ef97b3d5de4 : SUCCESS in 3m 15s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b85a53dbeb6c4ee19a6060385d841136 : SUCCESS in 30m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/010c39433de2483ba15f0fcc72ef01a9 : SUCCESS in 1h 04m 24s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a703fe15b7324d098e8008f6269937ec : SUCCESS in 1h 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a48614671cf4cbb8478b1b6ecf5bc65 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae76f364c84e4bfd9e56da97c48e9df5 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0caf0e14b60a45529dcfc3d585fa0d02 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20cdb0830c6c468694ef2c2269b9cbcb : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e73237a625ff483db1bdf8850f89d692 : SUCCESS in 4m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe94ae4fee084361aec4ac45c5ca9b38 : SUCCESS in 27m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc7a14840d0c4239b2b497cda8be1247 : SUCCESS in 57m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f6921983e6394632892dbf0d666aeb8f : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a0f163330017252a91b058ff8f406e59f2ac3382","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 14:17:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8dac0a773333b08fe9df0ab51aa674bb13fc1ec4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 14:20:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd686642fb664912976ad0ab16f556ad : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b73af44a90a14e508e2c0f6ed1f3b31b : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e6cea7b93df7b60619c1348f18356579bf4e0946","revisions":{"7cd7cb8bdb3bc4e72bf109fdc06b6f9632e6c4c8":{"kind":"REWORK","_number":1,"created":"2021-06-02 14:41:00.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/38/794238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/794238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/794238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/794238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/794238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/794238/1"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-06-02 14:32:29.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-06-02 14:39:25.000000000","tz":480},"subject":"Remove rally","message":"Remove rally\n\nRemove rally project as planned\n\nChange-Id: I3560c3a2bee4e391cac841b9bcd8b8d2199f920d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cd7cb8bdb3bc4e72bf109fdc06b6f9632e6c4c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cd7cb8bdb3bc4e72bf109fdc06b6f9632e6c4c8"}]},"branch":"refs/heads/master"},"1a8b195450c1652a638328c317bbfc0777680320":{"kind":"REWORK","_number":2,"created":"2021-06-17 05:45:15.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/38/794238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/794238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/794238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/794238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/794238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/794238/2"}}},"commit":{"parents":[{"commit":"ebb871338a7c3315677e510f5a3b1ad4c7040c52","subject":"Merge \"Pin td-agent to 4.0.* to fix missing logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebb871338a7c3315677e510f5a3b1ad4c7040c52"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-06-02 14:32:29.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-06-16 15:06:58.000000000","tz":480},"subject":"Remove rally","message":"Remove rally\n\nRemove rally project as planned\n\nChange-Id: I3560c3a2bee4e391cac841b9bcd8b8d2199f920d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a8b195450c1652a638328c317bbfc0777680320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a8b195450c1652a638328c317bbfc0777680320"}]},"branch":"refs/heads/master"},"e6cea7b93df7b60619c1348f18356579bf4e0946":{"kind":"REWORK","_number":3,"created":"2021-06-20 13:22:27.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/38/794238/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/794238/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/794238/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/794238/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/794238/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/794238/3"}}},"commit":{"parents":[{"commit":"ebb871338a7c3315677e510f5a3b1ad4c7040c52","subject":"Merge \"Pin td-agent to 4.0.* to fix missing logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebb871338a7c3315677e510f5a3b1ad4c7040c52"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-06-02 14:32:29.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-06-20 13:22:10.000000000","tz":480},"subject":"Remove rally","message":"Remove rally\n\nRemove rally project as planned\n\nChange-Id: I3560c3a2bee4e391cac841b9bcd8b8d2199f920d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6cea7b93df7b60619c1348f18356579bf4e0946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6cea7b93df7b60619c1348f18356579bf4e0946"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
