)]}'
{"id":"openstack%2Fkolla~794909","triplet_id":"openstack%2Fkolla~master~I0cae18c1d5169cd444a14ba3b56de24dede5919c","project":"openstack/kolla","branch":"master","topic":"heat-dashboard-policy","hashtags":[],"change_id":"I0cae18c1d5169cd444a14ba3b56de24dede5919c","subject":"Horizon: Install required policy files for heat-dashboard","status":"MERGED","created":"2021-06-05 03:17:37.000000000","updated":"2021-06-17 08:46:00.000000000","submitted":"2021-06-17 08:42:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794909-heat-dashboard-policy","meta_rev_id":"fe5cd085d11c7de83976d4a60e3a3ff189927192","_number":794909,"virtual_id_number":794909,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 08:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-17 06:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-07 09:29:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-17 06:32:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-05 04:38:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 09:29:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-17 06:32:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"63d0aa2fa4d00135ad0dc60b03f870e91040e1fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-05 03:17:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d4e2f876d8f91db741b1addb9c4208bdcb7da93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-05 03:21:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f4e6632317c53f313015529e8fb6587e42b37c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-05 03:22:31.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9a4ca8486f49139b9523e1f640d59c89c982f201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 04:38:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8cb225021635486a9a58456c556eca3c : SUCCESS in 20m 56s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/afe240cf2aeb443bb0a5c70c598571b6 : SUCCESS in 37m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/415833acdbf541e18d063de94fa1c829 : SUCCESS in 1h 08m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/26052385119e445b8829c77e99982b02 : SUCCESS in 1h 06m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4bac5bf693a74d03b187ec18222b7c5e : SUCCESS in 1h 03m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df3200981e7f445eb2bbba9e54ac97ac : SUCCESS in 36m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/399d4857e0364dafb27e451a2da8b368 : SUCCESS in 16m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c46a36ed322c44edaf88fcb28b8a89a3 : SUCCESS in 1h 00m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/524b96258d174c3483312ee567175e50 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/477d8b75b959470b8bf016af094a0051 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51c4d8a35d70462fa97fd9b0b4fd299e : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/361c28c09be342b9b075ac86a3f57a5d : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f049a787224428e966b523dac10e56d : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8a2777a9cfa9491986686b9e7d580b68 : SUCCESS in 16m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/524f34aa591246a5ad3e60e4f700a861 : SUCCESS in 27m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9936838e3794c1690be4544c8d61b86 : SUCCESS in 59m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/895f65d2483c4a429e4df584550e50f7 : FAILURE in 41m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d91b7987552441b88521eae0d18a1400 : SUCCESS in 1h 09m 18s","accounts_in_message":[],"_revision_number":3},{"id":"dd7c004deb9b197b43ad7a73935b6f2b881df003","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 05:15:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5d7a78bdcd5421aa8c88bd0c196c80d : SUCCESS in 1h 41m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8aadaddfe1946788b8611110721450a : SUCCESS in 52m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ed4c053ded640b881707778f930bd6d : SUCCESS in 1h 39m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eda1a401a9f84a27a9491dc9a653506a : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01d2edf79c224b71ad976d5ec5172a69 : SUCCESS in 6m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd3e34dc09d1406b99d0120a7afe3dbc : SUCCESS in 59m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f179364fca89b619ba8303f5a5e0598ca3e676aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 09:29:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"43b2a57daa56c5a5975f49a0b81b6e2de3e082af","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-17 06:32:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"127a1815f015cd475eaa98b082c5399ec3165cd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 06:32:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"da7a9eee5fc1e6fb4ab67e61aa5944a367e3cf28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 08:42:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/002263508c284621bba69f17240bc6ac : SUCCESS in 57m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/75a35eb72d74495abdfbfbb62ee33ca4 : SUCCESS in 1h 58m 54s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c35bc9189c7d4d0988750093cc17077b : SUCCESS in 2h 04m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20fd890e400c40f4b0fd86db98256213 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9dd32c90fb4a41f1956d47ec0180e5a3 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eff83b70b95d4a2f9f49f5228e10ee0f : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e83f6a35a39542c59f9fbba331a4b330 : SUCCESS in 7m 19s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc18c481ced74c29a83a7bfe7ad2c9b0 : SUCCESS in 34m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1dddff0275824536bbf516e56da1d273 : SUCCESS in 52m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2193c5c9486941edb39d736e09d3a061 : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":3},{"id":"f9c86f30b99e8fe942302e11c1a4ff3688c2d93d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 08:42:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fe5cd085d11c7de83976d4a60e3a3ff189927192","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 08:46:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f16d1f26384f259145d8c513189592 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f9715a66e24638cdcf11223c23a7958a1e2c3a8f","revisions":{"8d9f5cb2626f6939cdf197cd8f5cf3fdfb1cb2b5":{"kind":"REWORK","_number":1,"created":"2021-06-05 03:17:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/794909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/794909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/794909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/794909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/794909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/794909/1"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 03:15:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 03:17:34.000000000","tz":540},"subject":"Horizon: Install required policy files for heat dashboards","message":"Horizon: Install required policy files for heat dashboards\n\nSince the commit 8e7914fce24d2c9d94a83795983aaa0fb05f020c was merged,\nheat-dashboard no longer provides policy.json but it use two yaml\nfiles to manage policy rules. This change updates managed files for\nheat-dashboard accordingly.\n\nChange-Id: I0cae18c1d5169cd444a14ba3b56de24dede5919c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d9f5cb2626f6939cdf197cd8f5cf3fdfb1cb2b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d9f5cb2626f6939cdf197cd8f5cf3fdfb1cb2b5"}]},"branch":"refs/heads/master"},"ef35081ef2b8b95e2eb7d98a93d2093290c3530c":{"kind":"REWORK","_number":2,"created":"2021-06-05 03:21:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/794909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/794909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/794909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/794909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/794909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/794909/2"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 03:15:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 03:21:50.000000000","tz":540},"subject":"Horizon: Install required policy files for heat dashboards","message":"Horizon: Install required policy files for heat dashboards\n\nSince the commit 8e7914fce24d2c9d94a83795983aaa0fb05f020c was merged,\nheat-dashboard no longer provides policy.json but it use two yaml\nfiles to manage policy rules. This change updates managed files for\nheat-dashboard accordingly.\n\nChange-Id: I0cae18c1d5169cd444a14ba3b56de24dede5919c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef35081ef2b8b95e2eb7d98a93d2093290c3530c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef35081ef2b8b95e2eb7d98a93d2093290c3530c"}]},"branch":"refs/heads/master"},"f9715a66e24638cdcf11223c23a7958a1e2c3a8f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-06-05 03:22:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/794909/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/794909/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/794909/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/794909/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/794909/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/794909/3"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 03:15:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 03:22:05.000000000","tz":540},"subject":"Horizon: Install required policy files for heat-dashboard","message":"Horizon: Install required policy files for heat-dashboard\n\nSince the commit 8e7914fce24d2c9d94a83795983aaa0fb05f020c was merged,\nheat-dashboard no longer use policy.json but it use two yaml files to\nmanage policy rules. This change updates managed files for\nheat-dashboard accordingly.\n\nChange-Id: I0cae18c1d5169cd444a14ba3b56de24dede5919c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9715a66e24638cdcf11223c23a7958a1e2c3a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9715a66e24638cdcf11223c23a7958a1e2c3a8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
