)]}'
{"id":"openstack%2Fkolla~794910","triplet_id":"openstack%2Fkolla~master~Id92145ba74237da2fd8430b9d84413465297d3a7","project":"openstack/kolla","branch":"master","topic":"manila-dashboard-policy","hashtags":[],"change_id":"Id92145ba74237da2fd8430b9d84413465297d3a7","subject":"Horizon: Install policy files for manila-ui","status":"MERGED","created":"2021-06-05 03:21:34.000000000","updated":"2021-06-17 07:54:35.000000000","submitted":"2021-06-17 07:51:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794910-manila-dashboard-policy","meta_rev_id":"296a7595b3e69bef090b8df324e064dba0b5a8fd","_number":794910,"virtual_id_number":794910,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 07:51:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-17 06:32:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-07 09:39:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-17 06:32:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-05 04:44:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 09:29:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-17 06:32:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ca1af328796d8af7aff8a7c908435509541946b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-05 03:21:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e012912031ebd152b7d9ee15a1dc4799543a6d9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 04:44:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/cd7f54c16a5447b6a7d3b4a2845ddf6a : SUCCESS in 30m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d73047adc7d34c1eb30320c6ffaa4b6a : SUCCESS in 31m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c844b89a04fb4601b264f26f5a047bbd : SUCCESS in 1h 11m 34s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e6252c364c19497ba8a6f98dfbca8f6b : SUCCESS in 1h 08m 19s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/823ae698d9fc43d0a5f930c0b22da632 : SUCCESS in 1h 15m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2d87f0c427f744fcb2e3af44d830c495 : SUCCESS in 34m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cdc62cefc2e84e2ea9291048e627255a : SUCCESS in 19m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/447ad629238a4e9aa0aa997f94bddee5 : SUCCESS in 1h 01m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c3530b1a7b4446798e57377cd0e81b7 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f618e8d4a8c48458d31644d52edcddc : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4b53cdb422343f391fbe777e4c3cf22 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f8650dbb2de4380b3b748958ea579c5 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e052209436540b7ba00df64feff955e : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e2afb82c6b645a6ae3634ff02b4abf2 : SUCCESS in 23m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0441a986566e450eb5c7acde904e0cd1 : SUCCESS in 33m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3707f10924ab46aca6c6fda4fb0b1be7 : SUCCESS in 48m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c0ac16f16524b519382299e8b019cab : FAILURE in 41m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5a268fc03e3040c3861b4aa4faefc11e : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3d8b7dccefd1aa0f15f3f09e8a9cf124d7887489","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-05 05:30:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8822b3355aa4c26a802abeb33c13567 : SUCCESS in 1h 37m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/41717d242b6f45c79ccc1818dc40b16c : SUCCESS in 1h 01m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e5f15af0eb8a424db48ca6d3eadb814c : SUCCESS in 1h 48m 24s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a49ad0f5dc4f4663bbe9ef60ee1b29f5 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e39fb6d61b94121bc50bb78abf4e5f5 : SUCCESS in 8m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/283b9b38b061491881cbb855750b1d1b : SUCCESS in 1h 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa709d106624ca6c8ffdb8a5e576d9b389163f1e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 09:29:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d27e6a03a8cdfc4a15173a341b44c4fa3ae17775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-07 09:33:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"542441d99edf857e37fcbf6322625be3677cb94a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-07 09:33:48.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"461f0be7f482f5a4aec82f2e7426590fde3af094","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 09:39:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0a3b3a1d60af57ea8e2445a365cbc4a922e50b04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 11:07:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8b51f76afb374a39ae6f09d27acf3ee2 : SUCCESS in 29m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a8f6306a3d654ecd94c5ecea10fe4c01 : SUCCESS in 50m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9821980330e34941a52d662831560a57 : SUCCESS in 1h 08m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7bbf35e2db584a548da7a6f348be5aa7 : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5586b8035ce14ff88a119b29eda201ac : SUCCESS in 1h 20m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ceac2f33e38647f1960b3171cd728aba : SUCCESS in 38m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c2aa18bb590b4136b9307cab9adc7b4f : SUCCESS in 25m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c3977abd9f3943d08501846c1f58710d : SUCCESS in 1h 01m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f067387f2e9a4d62b42ec8555d19aedc : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c7470ef57ea485cb39683aa83b29de7 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2f63e584e394b44af0b478cfb2cdc7a : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89350643e1c64ac280b562b15883fffc : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2fa55eaac24253b2700ea83b67fbed : SUCCESS in 6m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/35a5e41d79954c8f9bef719b4dccbcf2 : SUCCESS in 18m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/55dc23226b2a47febc3edd9149631482 : SUCCESS in 42m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/15f81a60d54f4222ab72d7dfe23a4d7e : SUCCESS in 1h 01m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b80a276a151b4870a1c2d265eae3bd46 : SUCCESS in 49m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3b5ade8bb23844828e078a0f2cdfd2f4 : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b80118f9219fa6770c7287c992fc30f85fdf0bf5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 12:06:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c44814a2f0f54021ae7e6c87cce2f9e9 : SUCCESS in 1h 38m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca0c8f8e9dec4987a72b59b956212758 : SUCCESS in 1h 36m 59s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fa110d62ce2745f2a6d2232ceb41ebc8 : SUCCESS in 2h 10m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b8d46c6e07a4428b8e2c09b526e3816 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/402a09e5407b49329169dac4037acebd : SUCCESS in 8m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8d4b11edc474e30ba581b1c0f87796a : SUCCESS in 1h 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36286b80179144f19f6f224119c490dd157c03d9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-17 06:32:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a13e987f4dd47356a012c8063d661d0fb809f53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 06:32:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"21ae5128813610280e75c545e2bbf6b48a3207ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:51:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c112b19cc1f748f9b02bcddb12cd2db0 : SUCCESS in 40m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c5389418c44a4fc583ac26a0a7e01697 : SUCCESS in 1h 03m 23s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/65f3f4a5de4c4eeb83e9a178319fc8ed : SUCCESS in 1h 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e04e87a46046629871b7f1611b9398 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a318990d5ab4481908f915bb5a6b8d0 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0a46ca87404409ab2b4fffee56c02b4 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49fc0702e2d345f794e8f6e4d4810ae2 : SUCCESS in 8m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a7b5a0b2acf460285dfaa9d722d8248 : SUCCESS in 46m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/18de232d805543f1814effe5a8f82c35 : SUCCESS in 58m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/60fa051bde664ca2b4d63b414cb0d113 : SUCCESS in 1h 14m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e293bd433eec45a9c24ab7996de27e6616f4f491","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:51:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"296a7595b3e69bef090b8df324e064dba0b5a8fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:54:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c78a27745842dd9a7d2725cf994986 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"77fc442a91bf479772b45dd8a3c9de1691e450b9","revisions":{"dcc74a26e54bffc4f1fd99d1b04aa04144647dce":{"kind":"REWORK","_number":1,"created":"2021-06-05 03:21:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/794910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/794910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/794910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/794910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/794910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/794910/1"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 03:20:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 03:20:14.000000000","tz":540},"subject":"Horizon: Install policy files for manila-ui","message":"Horizon: Install policy files for manila-ui\n\nSince the commit 89a90ff9773b93062760df5e3deefb9750112633 was merged,\nmanila-ui provides two policy files. This change ensures that these\nfiles are installed when manila-ui is enabled.\n\nChange-Id: Id92145ba74237da2fd8430b9d84413465297d3a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcc74a26e54bffc4f1fd99d1b04aa04144647dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcc74a26e54bffc4f1fd99d1b04aa04144647dce"}]},"branch":"refs/heads/master"},"77fc442a91bf479772b45dd8a3c9de1691e450b9":{"kind":"REWORK","_number":2,"created":"2021-06-07 09:33:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/794910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/794910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/794910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/794910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/794910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/794910/2"}}},"commit":{"parents":[{"commit":"5338b559ea12af998d89da3f339788a18e775dcd","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5338b559ea12af998d89da3f339788a18e775dcd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-05 03:20:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-07 09:33:06.000000000","tz":540},"subject":"Horizon: Install policy files for manila-ui","message":"Horizon: Install policy files for manila-ui\n\nSince the commit 89a90ff9773b93062760df5e3deefb9750112633 was merged,\nmanila-ui provides two policy files. This change ensures that these\nfiles are installed when manila-ui is enabled.\n\nChange-Id: Id92145ba74237da2fd8430b9d84413465297d3a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77fc442a91bf479772b45dd8a3c9de1691e450b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77fc442a91bf479772b45dd8a3c9de1691e450b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
