)]}'
{"id":"openstack%2Fkolla~795038","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I3c6fdcb9b7dd7443a7755599f7e4ee59f67e0a91","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1930586-stable/wallaby","hashtags":[],"change_id":"I3c6fdcb9b7dd7443a7755599f7e4ee59f67e0a91","subject":"Add missing default policy files for ubuntu-binary-horizon","status":"MERGED","created":"2021-06-07 12:39:04.000000000","updated":"2021-06-11 03:55:15.000000000","submitted":"2021-06-07 16:06:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795038-bug/1930586-stable/wallaby","cherry_pick_of_change":794589,"cherry_pick_of_patch_set":1,"meta_rev_id":"caf3cb55d18859c67ed7621830d2657c18cfed69","_number":795038,"virtual_id_number":795038,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-07 16:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-07 13:23:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-07 13:18:49.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-07 12:43:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-07 13:23:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-1,"date":"2021-06-07 12:43:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-06-07 12:44:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 12:39:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-07 12:45:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-06-07 12:45:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-07 14:04:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce4245a8b7b56e39dd39f571333d6d7928d0741e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-07 12:39:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eb1ec543f6b5ac576df063aa48d1b8bc876758d2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-07 12:43:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ff220a1bc3e0cb6a93384c2354d9eccb3653427","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-07 12:43:41.000000000","message":"Patch Set 1: Backport-Candidate-2\n\nnot below wallaby","accounts_in_message":[],"_revision_number":1},{"id":"7aa7eaf75141712a936ccbebb859358877c3245e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-07 12:44:37.000000000","message":"Patch Set 1: Review-Priority+1\n\nrelease blocker","accounts_in_message":[],"_revision_number":1},{"id":"57ae7c05688234442bd84d5b9109b0784bf2d450","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 13:18:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11c2aa6849704ddd7de358bc0014eedc2b835482","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-07 13:23:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"761b2b3a4e46db703975ae91f8a7bf5cb0de0dc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 14:04:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9763fc7384c14e79b20bf09c163d3356 : SUCCESS in 22m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3b5e2fe296b244249eb135b33427691f : SUCCESS in 41m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/473e42166ce44b97a238bcf4722989f4 : SUCCESS in 1h 10m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8b4d2b7a991b47bda87692e53c9a2c2d : SUCCESS in 1h 08m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7b7b8eb6e26c4fe894c90f78faea2abf : SUCCESS in 1h 20m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/37634b5bfafd4d6daad44f014bcba97c : SUCCESS in 30m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bbeffe9678ad48afaf446a7ba7d508c9 : SUCCESS in 21m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2f531453c4fb4cdab2107892da2eca13 : SUCCESS in 1h 00m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04570f70033b4e2c8ce22987f9ec3a32 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bf6429597364ae4ad50c6a4b9796c3b : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c594d183ae834834ac5a1c945e93cdb5 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/201f1552208749c09f5153b34f041e06 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffdbeeeda7814abb96f031b110c8849d : SUCCESS in 6m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d22a83f1ec9d4ecab1df40aa9813de39 : SUCCESS in 24m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c56b7d914d054da2a10aa4c8983fd914 : SUCCESS in 33m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/77a43f8a80cb455d92c08700ee11b05e : SUCCESS in 56m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eca124e18cd4427c8f4edccf1c486cc5 : SUCCESS in 1h 02m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6dd9df1a4cc84da5ac792941ad9518ca : SUCCESS in 1h 11m 12s","accounts_in_message":[],"_revision_number":1},{"id":"fab59428de339d1cdc5376285a3e5d55dcb18ee8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 14:04:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"427026ae87561eb310eaf811de2f548eb6ecc574","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 16:06:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f7ea5122d07f4930a5d16630fbf185f8 : SUCCESS in 1h 21m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/78860382fae84cacbe25b67d16818ca4 : SUCCESS in 1h 25m 22s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/626e0ee3a01247708e3c005dd85fab8e : SUCCESS in 1h 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d9d08c102848e3968f3bb04b3f2d45 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4db9ad48f874e7886ec5c6519865380 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7c17c3f8d7d433e81e40c2a1291f49d : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8417dcb4f1430aa6ce324dbd18fbf5 : SUCCESS in 6m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c3016cf404d45bbacfde2051ecff00b : SUCCESS in 36m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/35cf4e37e13541ce83a6007caa4b1c9b : SUCCESS in 55m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cc82919d25b1483e8ea7d66b6ee8c397 : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":1},{"id":"85181bc483814ac844973a3ac912fb5bb8737ec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 16:06:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"496cce4cba400647f1e6af099a80fee663e705b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 16:08:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d43e348352043cc8fda7144822970d0 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1},{"id":"caf3cb55d18859c67ed7621830d2657c18cfed69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 03:55:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1594a0dd06544f299edbc74e13c951b3 : SUCCESS in 55m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e2a05724bab04518af9759db9e88909d : SUCCESS in 52m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/67a72c6f710146c69aa569d4411bc765 : SUCCESS in 1h 36m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc65aeec21d342548b81d1a2f530780f : SUCCESS in 38m 57s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51802de9614f7c8cf3d8b40b19654b0065fc1a0a","revisions":{"51802de9614f7c8cf3d8b40b19654b0065fc1a0a":{"kind":"REWORK","_number":1,"created":"2021-06-07 12:39:04.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/38/795038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/795038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/795038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/795038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/795038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/795038/1"}}},"commit":{"parents":[{"commit":"cfa9ee48674d549fed38bdae5e1a422f3b778f5c","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfa9ee48674d549fed38bdae5e1a422f3b778f5c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-03 13:37:45.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-07 12:39:04.000000000","tz":0},"subject":"Add missing default policy files for ubuntu-binary-horizon","message":"Add missing default policy files for ubuntu-binary-horizon\n\nThe ubuntu-binary-horizon image is missing the default policies files\nwhich should be in /etc/openstack-dashboard/default_policies. By copying\neverything from /usr/share/openstack-dashboard/openstack_dashboard/conf\ninto /etc/openstack-dashboard, we get the default policy files, as well\nas commented out policy files (e.g. cinder_policy.yaml) and the\nnova_policy.d directory containing api-extensions.yaml.\n\nChange-Id: I3c6fdcb9b7dd7443a7755599f7e4ee59f67e0a91\nCloses-Bug: #1930586\n(cherry picked from commit 740448a658f0060b86d56e78915d8b751ade3616)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51802de9614f7c8cf3d8b40b19654b0065fc1a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51802de9614f7c8cf3d8b40b19654b0065fc1a0a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
