)]}'
{"id":"openstack%2Fkolla~795054","triplet_id":"openstack%2Fkolla~master~I74bfa6326d8a72f9bba409e6a725fdb55b48c6b1","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2021-12-09 15:31:58.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-12-09 15:31:58.000000000","reason":"Change was submitted"},"30523":{"account":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"last_update":"2021-12-09 15:31:58.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2021-12-09 13:50:24.000000000","reason":"Michal Arbet replied on the change"}},"hashtags":[],"change_id":"I74bfa6326d8a72f9bba409e6a725fdb55b48c6b1","subject":"horizon: add various dashboard plugins for ubuntu/binary","status":"MERGED","created":"2021-06-07 08:09:49.000000000","updated":"2021-12-09 15:35:23.000000000","submitted":"2021-12-09 15:31:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795054","meta_rev_id":"ef5d84e9059b08335e0da727f07b7c3eba4b902d","_number":795054,"virtual_id_number":795054,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-09 15:31:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2021-11-24 14:06:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2021-11-25 08:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-12-09 13:50:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-12-09 13:50:24.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 09:38:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-08 17:33:58.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2021-11-24 14:06:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-11-25 08:54:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-11-25 10:25:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-12-01 14:23:29.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-12-07 10:52:28.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"}],"messages":[{"id":"10ea5e79ee78d3e15539597aba0e38bd58828427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 08:09:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ba29ff2b1ab5a58bff99583587231fa1df24309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 09:38:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/eef712c7d924452ba8fd1421781657ac : SUCCESS in 47m 31s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3083bb59f308471ab1a53a1a67ad8389 : SUCCESS in 38m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/07ef922c4173400fb11dfec601aa73ab : SUCCESS in 1h 17m 54s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bf47da6e16274b8a9165018a5b4c7c3f : SUCCESS in 1h 20m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/18ea731a33a04ea49e2c8c117d11f4fa : SUCCESS in 1h 19m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/745227e602f3435f86c27050ab582865 : SUCCESS in 31m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bebc9a7693ad45c389587f2314d74662 : SUCCESS in 45m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c54dbf62d99d49a4bb7321f4dd897a67 : SUCCESS in 1h 09m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed7790a1b76c42768ae7c6f9dec6027d : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4feac8956f6f4f2593949bd6e5bb622a : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/156a02d3d9c74ef298b3a86f57543810 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1163273b32fd460a90ca71a7ef63a0e3 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b847fa2208b446eb8f90e110868b12a9 : SUCCESS in 6m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/282f83834d974bf6a82c733501f07fcb : SUCCESS in 18m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c1610d3477034e23b9991640ca97548b : SUCCESS in 43m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa3e8b4d646c45cc8658197dd5400585 : SUCCESS in 53m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b98de4fc6464416ebc9186d5662dc846 : SUCCESS in 56m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/adbcafa82aad4e8c84674b956f306606 : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ee8feeeb0c341a4793b32e458a08c4e351a48142","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 10:36:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb36ad4d29eb404dae1e5eab7d57682b : SUCCESS in 1h 07m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4a266d1382cf44549a45199beb89e956 : SUCCESS in 1h 05m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef3c4e8a9d1344be81c6c1a50f094c11 : SUCCESS in 2h 26m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5f1eea4a6eb4729b0e8a47bd01f6a33 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62681738e08c462fb603b090e91dcce3 : SUCCESS in 6m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/255300d4ebc642428161e2389c2525e6 : SUCCESS in 1h 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a9f821dbaea40f53c0d59963fa5190e42f44acd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-11-23 11:35:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"161ba6bac42702486c735d8f2ca9696f3c7cf0a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 13:05:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d761a6732a3b4520bab0422b9f3bb72c : SUCCESS in 23m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2862f1d145e14a3f939704cc9a25cda1 : SUCCESS in 34m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cafd27a5799e4e4fbdf7a0aff5f56480 : SUCCESS in 1h 13m 52s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6f3ccb7b9a4b4c8594ac4ec1115f80e7 : FAILURE in 1h 10m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b3d0cb4b65154acf9ccc2991eefe8c61 : SUCCESS in 1h 26m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d3d5c9bf2c764e64ade5de2d5c2ce463 : SUCCESS in 23m 39s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/729ade74196447d3abb16730783921e0 : SUCCESS in 17m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d4c6abf6431e463dbf2396a5afa48627 : SUCCESS in 52m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6a876e6031ce461199d39191070b0fa8 : SUCCESS in 59m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4c87072468b47c38706e342bfc3f565 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e32db31a428414b998c9589b4d48318 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82001bc6c0e74c83af35993a7adc8e60 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e2cf136d10845afadace05685f8e909 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f115eda8f349f188bac6b2f3db987b : SUCCESS in 7m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f075aa6bacf04896af0ac8ab559e54fb : SUCCESS in 29m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/89d2f194d7904e478f1cd6e2b0a6cd3f : SUCCESS in 26m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/643c61c7ce714c5db1de80e1b333e06b : SUCCESS in 48m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c3f4ddd28e76441c9256012f3237bb0b : SUCCESS in 47m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6a5ff0ac9fb4458cb1627a5c0578852b : SUCCESS in 1h 04m 16s","accounts_in_message":[],"_revision_number":2},{"id":"51feae70064bdb66b767ec0e771632845f75433b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 17:54:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c79f8f136a794671a2864f679a230e8d : SUCCESS in 1h 04m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/02de8230f7e74bb0bff644b04806f501 : TIMED_OUT in 6h 16m 10s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc643f851ced47fb94b9934bad4d551f : SUCCESS in 1h 43m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15d0f641084349babf8b3dae453fcc2f : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c35cf96102974abd82408495c4054e1f : SUCCESS in 6m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a25d2fec98f748e69f0c3cffaca079e2 : SUCCESS in 1h 01m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a5f7351754061a68fa3f52a629c81ceabf0409b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-24 14:06:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5608d7c6ca6770b1136f8332557fb93ba20ea86f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-11-25 08:54:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d3ace5988e793eda90d36c5ec1946e5d10f66bcf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-25 10:25:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d22f2f3c609252310c31667ee67effadd31072f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-12-01 14:23:29.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e37016d98f19f771fe709501b14d68af1bb5349","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-12-07 10:52:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f868a3456cc73ba64aede2a80d13a058fcc844ca","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-12-09 13:50:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c096054c12484c188a7d6f2796e58004c92c29da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 13:50:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"310cf2c1775a8a705095fbbe8a3ada9123447a11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 15:31:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2188174510544a489f3d3bf1b2f69007 : SUCCESS in 43m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3da9fd77c78840179ef2ac1e549e1d7a : SUCCESS in 1h 17m 38s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/273885a445a444fe8228ff320fb3c9ea : SUCCESS in 1h 29m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df1a71c9d3d940bab5bb1daa7f2de7c1 : SUCCESS in 25m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/799b77869a5f4667878345589af98a8a : SUCCESS in 1h 00m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9fb2dc1ac42c4f299194944a79ee2fd7 : SUCCESS in 1h 07m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/140834e14d39426da7eac66a71c6d122 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36ac2b1dba9148f998afc0231159318a : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa964cf65c4e4e02be243e71addebb83 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce04e1a3cda94220b025845b42fa44eb : SUCCESS in 6m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/83d1ccc7d8ca403fb86575d18c76f85d : SUCCESS in 23m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad38cbdb75cf490c9319fe24de88c4c8 : SUCCESS in 53m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac52754ba6ff497582f248caada7b52c : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f3ff3ea36a15d1baf01e5628bec33bb3e90c0c2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 15:31:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ef5d84e9059b08335e0da727f07b7c3eba4b902d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 15:35:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56ff71b584448c7b3978a6f81e94e95 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3479eb78141a6e8629b6bfb800319ade9e2f9a33","revisions":{"b51f0eb2bcf36cd166c8a0f7d176945c630c1a0b":{"kind":"REWORK","_number":1,"created":"2021-06-07 08:09:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/795054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/795054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/795054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/795054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/795054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/795054/1"}}},"commit":{"parents":[{"commit":"1a2c33512b6c4bdcdbf7d7bb08e791fb53e4f063","subject":"libvirt: pin edk2-ovmf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a2c33512b6c4bdcdbf7d7bb08e791fb53e4f063"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 08:08:55.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 08:08:55.000000000","tz":60},"subject":"horizon: add various dashboard plugins for ubuntu/binary","message":"horizon: add various dashboard plugins for ubuntu/binary\n\nThese were added for Debian/binary, but not Ubuntu/binary.\n\nChange-Id: I74bfa6326d8a72f9bba409e6a725fdb55b48c6b1\nCloses-Bug: #1931075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b51f0eb2bcf36cd166c8a0f7d176945c630c1a0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b51f0eb2bcf36cd166c8a0f7d176945c630c1a0b"}]},"branch":"refs/heads/master"},"3479eb78141a6e8629b6bfb800319ade9e2f9a33":{"kind":"REWORK","_number":2,"created":"2021-11-23 11:35:52.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/54/795054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/795054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/795054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/795054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/795054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/795054/2"}}},"commit":{"parents":[{"commit":"bc4c654e9be983c89bf8986863f39769ef36e7c2","subject":"Merge \"Use MariaDB 10.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc4c654e9be983c89bf8986863f39769ef36e7c2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 08:08:55.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-11-23 11:35:33.000000000","tz":60},"subject":"horizon: add various dashboard plugins for ubuntu/binary","message":"horizon: add various dashboard plugins for ubuntu/binary\n\nThese were added for Debian/binary, but not Ubuntu/binary.\n\nChange-Id: I74bfa6326d8a72f9bba409e6a725fdb55b48c6b1\nCloses-Bug: #1931075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3479eb78141a6e8629b6bfb800319ade9e2f9a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3479eb78141a6e8629b6bfb800319ade9e2f9a33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
