)]}'
{"id":"openstack%2Fkolla~795086","triplet_id":"openstack%2Fkolla~master~Iefcdd3100b7e3c5320bc5f1286a18251bdeab885","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Iefcdd3100b7e3c5320bc5f1286a18251bdeab885","subject":"Pin td-agent to 4.0.* to fix missing logs","status":"MERGED","created":"2021-06-07 11:08:55.000000000","updated":"2021-12-28 09:46:22.000000000","submitted":"2021-06-15 14:28:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795086","meta_rev_id":"c344510180a6200a856554430dbd9581283b462c","_number":795086,"virtual_id_number":795086,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 14:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-15 08:36:23.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-15 08:36:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-06-15 08:36:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-12-28 09:46:22.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-28 09:46:22.000000000","post_submit":true,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 11:31:33.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2021-06-07 12:33:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-08 08:40:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-14 12:59:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-06-15 08:36:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"5e4d61a207063dba2a452e1073d72ac9ef905438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-06-07 11:08:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"413b792a106fdd4a771edd5e4a5a59f161a54557","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-06-07 11:10:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"58ef3ac7092800e77b7a003b7993aeb46ce99a49","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2021-06-07 11:31:33.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\nPinning this version helped fix the issues. There doesn\u0027t appear to be a tdagent packaged version of 1.23.x fluentd yet.","accounts_in_message":[],"_revision_number":2},{"id":"f93c0c4d618f3c5f0d659a609db5c865f7118842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 12:33:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/43e4a6714098498ca26236939f0d2db8 : SUCCESS in 22m 03s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/209a529db0cb4191afafb10d8ab6c9e4 : SUCCESS in 39m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e10cc2f232894adaa7ed1d8359f136d3 : SUCCESS in 1h 15m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a66aff9776e847609bdbcda92170d605 : SUCCESS in 1h 07m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/59edf191b06e4715bf8eaeae26e2909d : SUCCESS in 1h 04m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/40f7659f63664c10a1076fd241ca350a : SUCCESS in 31m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2272ad603b8c4dfb8a7b3e6d5fb34492 : SUCCESS in 15m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/560b9c4e6be24cc19edda74f00fac995 : SUCCESS in 1h 04m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7ea141d2351403dad9055e18d37f66c : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38d6dda1896b4850ae827d42ac4c6aaf : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85d1539117544d16a90b1030a750d639 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8267b32fe3034398a1499e2388cf7732 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8795eda4072e4be7801da56a23bbbe1c : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50ad19a7bae249548d1493dad83690b1 : SUCCESS in 4m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6b7b43016bd94aba82b1468e311775d4 : SUCCESS in 22m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/36f02f3dbd774ecbb2e65e08ed97322d : SUCCESS in 34m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a2a1e786eb744828faa657caa941985 : SUCCESS in 52m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4bf7742262a04ab99cb962549550a48d : SUCCESS in 54m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/98470ae34c7140c4b10028eff3772f82 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":2},{"id":"aa8f353f72febee31c7fc80cfe07409dd8276084","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 13:11:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/43dae49d978846e8a373a8dd05af49c9 : SUCCESS in 1h 53m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5fa3cb714ad4cfc8176463d0902ca98 : SUCCESS in 47m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c6ac615460e48b6ae3af40b3378b32e : SUCCESS in 1h 53m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9973ed8612914aa49194ee185c6bf506 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb640703bf6c4ce88d4adc4f105a98ca : SUCCESS in 6m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/af26cbf058fb447d8824bd05d655406c : SUCCESS in 1h 31m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42012493a873678f4ef20124fe6dc4b75aacbe1c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 08:40:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"398fb9409e06d95111fa67447d41cc4e8412a2fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 08:40:46.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"13947b7880656acb5a3acb45b86a1a19f68a9847","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-14 12:59:45.000000000","message":"Patch Set 2:\n\n1) which branches does it have to go to?\n2) could I get a TODO/FIXME in the two affected places?","accounts_in_message":[],"_revision_number":2},{"id":"8f02bf71f643af174a51ccc9568be8d6f0b8b0ef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-14 13:04:22.000000000","message":"Patch Set 2:\n\nad 1) Mark told me \"all\"","accounts_in_message":[],"_revision_number":2},{"id":"f24b7c2a7b599a84ea41511a283ffa845b2c6524","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-14 13:26:12.000000000","message":"Patch Set 2:\n\nad 1) continued - I checked and it\u0027s only victoria+ where we use td-agent 4","accounts_in_message":[],"_revision_number":2},{"id":"94fcb47754acfd5c04d2a8ffef60316b5189545a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-15 08:33:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4c8e3665bfe59f4723e791b195c9709ab37f0821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-15 08:35:58.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0dae0fa78b17a0bc7a40990562395a445d28d695","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-15 08:36:23.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"392d251731148c5d8475bc0a48b36957511a3860","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-15 08:36:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nfixed the comments and extended reno","accounts_in_message":[],"_revision_number":4},{"id":"4002610e1b0d4f5325bcd6a98e53d8b983a67c7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 11:09:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/517678281ef143aea5bc40813443e978 : FAILURE in 1h 23m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9cb841e5eaaf4a08acb102fe9808e867 : FAILURE in 56m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c8da90b955e4d5fa049266e7202c88d : SUCCESS in 2h 11m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00995970cb6647d8ac58c5047f5836c0 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c78383d514c144adb8123162c66baf07 : SUCCESS in 6m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/209e68f6abc7482a9232226f5ef8e89e : FAILURE in 1h 06m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c427f878d0bb7899d07959cfc6dbdb26395ce879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 11:54:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2a4d2dc346c545fea14906f799efa622 : TIMED_OUT in 3h 13m 05s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0ba351abf9414ed89ee707b670bdc499 : FAILURE in 32m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3e2f1675c1b34f9c83a9224ad9697eb7 : SUCCESS in 1h 06m 52s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9c8b2cd00afe492481ed2a1979d6a081 : SUCCESS in 1h 04m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c30b6e989ff047e496b788c3ab97b817 : SUCCESS in 1h 33m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/38afcdcc0bdc4cc0a1137f9feb127b92 : SUCCESS in 34m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d6a53b6bab1a467f9cde02f24c89c232 : SUCCESS in 19m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/11f55d2bde1b401abcfb5b259eae4d3b : SUCCESS in 1h 01m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1166f6885de49de93c96ae251c32210 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/435d9b9c8c2248f3835fc65af8c6164f : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06766503e5004ab1b3db6ae8ebba6a1f : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fced64269e7f4bddb536686316335d40 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e414d3b4407c49849053baa8d88ad45f : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca6e64849f764347b3146f615bd4ceaf : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a2f32d00d02449f80a1812e2bb15abb : SUCCESS in 2h 27m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bcf8cb89f0040c8945afcc663a81926 : SUCCESS in 36m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d3e3dccd85f47529b576c7ec180b585 : SUCCESS in 51m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/306c5789cb814b79985e449a1aa39930 : SUCCESS in 49m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ae6fdca22d5b4231be389646d103aebd : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":4},{"id":"5b0299548d3824e500904c0957fd3e2d5cd658d5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-15 11:57:30.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"45db4a53eb7f06ae7d63fa8a532f91ab8b18c30c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 13:13:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d9418eb96ae6427c98feaf41aef459c4 : SUCCESS in 32m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bb8bd90527e349649bd13f8244848fe8 : SUCCESS in 49m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e2316c04e61443829e6d5d3a1a40578a : SUCCESS in 1h 04m 05s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/841dab14855d47458cc7d9bf50238925 : SUCCESS in 1h 02m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e844e715a3dd4c0a8124349463a8e909 : SUCCESS in 1h 13m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/de26d5ff4bbd439fae0695c570269156 : SUCCESS in 41m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/25610f4f81cd49bca943520bfd1c9fbb : SUCCESS in 16m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/de4a2f53ef514fb693212176be4650cc : SUCCESS in 54m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6e85e86c5a04dd5988d86721c91db3c : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb7cc6aef35d46809f58fdc1a97b0097 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e80348df550d4e5ca11c5f01be2c801d : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20d3ac45aa1642208294317ad6dbf90e : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198a422afdfc42d7b8c42e74415cefd9 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/763e24608ea74402ad95c331b37103fa : SUCCESS in 5m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2a74758b2044aecbc739cc540089c2b : SUCCESS in 23m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c601b63ad7da43ac8758fd3d1dbf8378 : SUCCESS in 36m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a9569c81ccf4b3d9fe556280bb198b3 : SUCCESS in 57m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/406252cdcbb14f199babbc5820141054 : SUCCESS in 51m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6375c7d7c092491a8a68ae8903250ac9 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":4},{"id":"c89bfb5a161f65c8a4ba11bc3f39855a23134494","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 13:13:37.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6ec734e48a35a26adc20dae86251eab00df1bf81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 14:28:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1dd38ca03f074b72a5fd02e406ec5159 : SUCCESS in 45m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0b3c9c2e5ec646e497b6725188e758dd : SUCCESS in 1h 10m 51s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ed2ab574e65b4510b3e7014ec801044f : SUCCESS in 1h 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9723d85cd4a4ae3bb9b3141da3770fc : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cda6bb399fc04c219b6203cd068a7c29 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1146113ad142474dba56f265ae991ba4 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ee1abb06b434ea2844edfd31c571c35 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/539dd1bfd650408086a38198dca11312 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/728595cd8d4847e582a5d36b0296b447 : SUCCESS in 36m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8368f8397fdd460f85fa48acafa55161 : SUCCESS in 49m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ff0010ef02a446483e05ad94eea80d2 : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":4},{"id":"8440b74c377461f6749d6412c4b830b69916e3c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 14:28:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a03b742fb12fb71cd08479ab2a1bec3606e77c3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 14:30:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/128b01e4ba56467a959bac844830549e : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a20449d87b4c4599b2fcc006b301ddd0 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":4},{"id":"a9e0e6a563f80f55dc43ac732c1dc2a67b26e803","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:51:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/54519a507bec4dbc9733e02e51f51e03 : SUCCESS in 1h 07m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e934b0606a8b47c9874e4e060b6fe2eb : SUCCESS in 1h 30m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/848ae05300df44fe9a937f5588b699b9 : SUCCESS in 2h 01m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae4c565bff2243bda617de4249a534a6 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52e8cada2b544d329cfae68966ffed6b : SUCCESS in 5m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a96d33c4d00647438c04d4ede7c3d20d : SUCCESS in 1h 06m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6ccaafef0345eedff6c8d65d58f6d3184ed07c0","tag":"autogenerated:gerrit:revert","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-12-28 09:46:22.000000000","message":"Created a revert of this change as I910d01fe28e142f40d5deb68ea9c9f1f41afbf0e","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6d0882ef799bb531d007c3e6839ef318fd512d5c","revisions":{"4eb2f843db03440cef710adf4aef9c7e75ef50d0":{"kind":"REWORK","_number":1,"created":"2021-06-07 11:08:55.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/86/795086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/795086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/795086/1"}}},"commit":{"parents":[{"commit":"53434625f64196529502bbeef618620b6267b0e0","subject":"Merge \"Sort the package lists in openstack-base alphabetically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53434625f64196529502bbeef618620b6267b0e0"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2021-06-07 10:14:55.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2021-06-07 11:07:57.000000000","tz":60},"subject":"Pin td-agent to 4.0.* to fix missing logs","message":"Pin td-agent to 4.0.* to fix missing logs\n\nThe release inclusion of Fluentd 1.12.x in Centos/Ubuntu packages\ncauses a significant number of logs to fail to appear in the\nElasticsearch or Monasca output plugins. The logs appear to\nsit in a local buffer, and are only flushed when Fluentd is\nrestarted. The same issue persists in the most recent release\n(1.13.0), so for now we pin the td-agent package to 4.0.* which\nshould restrict Fluentd to the 1.11.x release where we don\u0027t\nsee this issue.\n\nChange-Id: Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nCloses-Bug: #1930867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4eb2f843db03440cef710adf4aef9c7e75ef50d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4eb2f843db03440cef710adf4aef9c7e75ef50d0"}]},"branch":"refs/heads/master"},"cc5a66571ae40fdd05a82d57eaadec0d252d40da":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-07 11:10:40.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/86/795086/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/795086/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/795086/2"}}},"commit":{"parents":[{"commit":"53434625f64196529502bbeef618620b6267b0e0","subject":"Merge \"Sort the package lists in openstack-base alphabetically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53434625f64196529502bbeef618620b6267b0e0"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2021-06-07 10:14:55.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2021-06-07 11:10:23.000000000","tz":60},"subject":"Pin td-agent to 4.0.* to fix missing logs","message":"Pin td-agent to 4.0.* to fix missing logs\n\nThe recent inclusion of Fluentd 1.12.x in Centos/Ubuntu packages\ncauses a significant number of logs to fail to appear in the\nElasticsearch or Monasca output plugins. The logs appear to\nsit in a local buffer, and are only flushed when Fluentd is\nrestarted. The same issue persists in the most recent release\n(1.13.0), so for now we pin the td-agent package to 4.0.* which\nshould restrict Fluentd to the 1.11.x release where we don\u0027t\nsee this issue.\n\nChange-Id: Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nCloses-Bug: #1930867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc5a66571ae40fdd05a82d57eaadec0d252d40da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc5a66571ae40fdd05a82d57eaadec0d252d40da"}]},"branch":"refs/heads/master"},"5795f7d395ebe27de2231121804b5c739f317a7a":{"kind":"REWORK","_number":3,"created":"2021-06-15 08:33:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/795086/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/795086/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/795086/3"}}},"commit":{"parents":[{"commit":"53434625f64196529502bbeef618620b6267b0e0","subject":"Merge \"Sort the package lists in openstack-base alphabetically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53434625f64196529502bbeef618620b6267b0e0"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2021-06-07 10:14:55.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-15 08:32:22.000000000","tz":60},"subject":"Pin td-agent to 4.0.* to fix missing logs","message":"Pin td-agent to 4.0.* to fix missing logs\n\nThe recent inclusion of Fluentd 1.12.x in Centos/Ubuntu packages\ncauses a significant number of logs to fail to appear in the\nElasticsearch or Monasca output plugins. The logs appear to\nsit in a local buffer, and are only flushed when Fluentd is\nrestarted. The same issue persists in the most recent release\n(1.13.0), so for now we pin the td-agent package to 4.0.* which\nshould restrict Fluentd to the 1.11.x release where we don\u0027t\nsee this issue.\n\nBackport to Victoria.\n\nChange-Id: Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nCloses-Bug: #1930867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5795f7d395ebe27de2231121804b5c739f317a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5795f7d395ebe27de2231121804b5c739f317a7a"}]},"branch":"refs/heads/master"},"6d0882ef799bb531d007c3e6839ef318fd512d5c":{"kind":"REWORK","_number":4,"created":"2021-06-15 08:35:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/86/795086/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/795086/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/795086/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/795086/4"}}},"commit":{"parents":[{"commit":"53434625f64196529502bbeef618620b6267b0e0","subject":"Merge \"Sort the package lists in openstack-base alphabetically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53434625f64196529502bbeef618620b6267b0e0"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2021-06-07 10:14:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-15 08:35:51.000000000","tz":0},"subject":"Pin td-agent to 4.0.* to fix missing logs","message":"Pin td-agent to 4.0.* to fix missing logs\n\nThe recent inclusion of Fluentd 1.12.x in Centos/Ubuntu packages\ncauses a significant number of logs to fail to appear in the\nElasticsearch or Monasca output plugins. The logs appear to\nsit in a local buffer, and are only flushed when Fluentd is\nrestarted. The same issue persists in the most recent release\n(1.13.0), so for now we pin the td-agent package to 4.0.* which\nshould restrict Fluentd to the 1.11.x release where we don\u0027t\nsee this issue.\n\nBackport to Victoria.\n\nChange-Id: Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nCloses-Bug: #1930867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d0882ef799bb531d007c3e6839ef318fd512d5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d0882ef799bb531d007c3e6839ef318fd512d5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
