)]}'
{"id":"openstack%2Fkolla~795089","triplet_id":"openstack%2Fkolla~master~I7430b72f5d8abe59a7151df73e6d59898d942636","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I7430b72f5d8abe59a7151df73e6d59898d942636","subject":"Change network_mode default to host","status":"MERGED","created":"2021-06-07 11:58:01.000000000","updated":"2021-06-10 13:43:12.000000000","submitted":"2021-06-10 13:41:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795089","meta_rev_id":"a0f6e21982960f2af686cbb40f6af4c4fa4db3a5","_number":795089,"virtual_id_number":795089,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-10 13:41:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-10 13:43:12.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-10 13:43:12.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-06-10 11:06:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-06-10 13:43:12.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-06-10 13:43:12.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 14:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 15:24:05.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-06-09 15:27:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"aa8dee4c3eac316d6952e2e789bc3f2b99e2cea1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-07 11:58:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e51d6c3007a37bc3ec78a577ff4232eeb740552","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-07 12:18:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f79acefde5f7442b03456a6352bf09058f7bfce2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-07 12:39:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7388d79f576d9be6d328036fdf1c8683ebdd748b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 14:03:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fb3e5db7e6534591b5aa36e7ae3bef6e : SUCCESS in 27m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7915923716c945429e6e3e4d6446f541 : SUCCESS in 47m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b5e0335fada041aeaf0a7a97f0541dc7 : SUCCESS in 1h 10m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d55f511539fa49ab943f4db350d6b5d0 : SUCCESS in 1h 05m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/01a1274d2ca84187b6d58910d024dabf : SUCCESS in 1h 17m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8716586f1c3a470c8679f7c444e35abe : SUCCESS in 42m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c5d2f20a57484e25b98c5c6af9e4ec45 : SUCCESS in 22m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f56a30115a5c4777828f85ca762e5c9f : SUCCESS in 1h 03m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bff8cc65c914036bc9a9f1f392aec5d : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b244790eecad42a48be32a314ab39d58 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0eb0e827d22c4abd90b6c4388379c303 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb34671beb704017b77d9283f75cd6ff : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72385176be947f5a4b708213d5a3520 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fa2dc241f4747ef9153a5e3face7a70 : SUCCESS in 4m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e63b55a87abc4a5e86bc5a677e509fad : SUCCESS in 29m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a5cb115b8c4f463090ed154c162f8c2b : SUCCESS in 45m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/901faa3b66904281b6a97e3f4edd6aeb : SUCCESS in 56m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b716a4bab23c4eb7881dd95e584ca526 : SUCCESS in 1h 02m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/07fd784bbc9e4b05b949fbd63dadd271 : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":3},{"id":"5b39d44b74929381c2a55a916134d3f9662e0d7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 15:05:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c7fcc882e1e74273ac822e700598145f : SUCCESS in 1h 26m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/35bf47fa13a5492aa9acabc66a243a7a : SUCCESS in 1h 27m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a13807f9f3974a539237991352d84dbe : SUCCESS in 1h 58m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad4ca0bde7a74530bf255cec0acb45c5 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3128973d76464287a0e2c84b55ab9385 : SUCCESS in 6m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5fab2eb16264622ac6e33dee4377be9 : SUCCESS in 1h 51m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f9502334f2900d245856c83b3943aa9f4885362","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-09 15:24:05.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"593adfc1bc645056502f01eb27bd4c3fcf39f55b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 15:27:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ba5dc6e40d3fe105673f421395c65873318f6c70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 15:27:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6bf1f00b63ff99c7fa3ad8b9758f446d7254a203","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:50:05.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0cc0bdfc060a43d29fb7178406afae6d : FAILURE in 46m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ecdc5b533c9e4c0c85b40dcac3bbb772 : SUCCESS in 1h 13m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f871e5d0b138434ea69b1c0434be4d25 : SUCCESS in 1h 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31751f40d7ba4d03835d0309a94cf0ca : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49c7403dfe214b648429be68b92bba6b : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0440cb57cbe54ee3ae27a7a37e267e13 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e655c6c0bad8446aac272ba176e827c5 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f739635dcf5c4e628b2f250037945e2f : SUCCESS in 4m 24s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef1ca72a9b5741dabd9739f356d204a1 : FAILURE in 36m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/783bcef23abc41be9f9150ed4f6767d7 : SUCCESS in 51m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dc0c9b4a168d4c2eb82679b42114c609 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":3},{"id":"5ad55153aa96200d2f2954c94a45be4b30a22fe3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-09 17:04:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9595c6431d2cf4e5e356a7bb7d08dbccca725ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:25:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8ed665ae00c847d2b1f04df4ec767f96 : SUCCESS in 23m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7a08754ece404537b2e2f25cfc6384ab : FAILURE in 46m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3293290bee324ebb842a81dc8c7858fd : SUCCESS in 1h 03m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6e351d8ba5734df3bd17410855e72e8e : SUCCESS in 1h 14m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1c9e8632fc3046cea5b9ad453672cfbf : SUCCESS in 1h 17m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/47a874e0f8134307bc4920e4a7d6accb : SUCCESS in 28m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/694c4def8888483680e82ee7648bc276 : SUCCESS in 21m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b61e3a81c10b44e28816fa67e454d082 : SUCCESS in 1h 00m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f5932700a34e0fb378a3ed27fe4364 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab4b648ebfe84959a8897a7833880f4e : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dbde1e4bfa44af4811342004ac30576 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3a2b2677bf8483895075bed3f7a7b31 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9674828c967b43e5ba3d43d37df9e9d6 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c35691f2c6964991a6c9d39e3af6d56b : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/476b79ac6d784f5081515597554cfa46 : FAILURE in 22m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/491dd789224146b1aa780e5b8bbd8ba1 : FAILURE in 42m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f8a776f21864128971046249719ce7e : SUCCESS in 50m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e12f2eed863444fb0883ae5e2f13c58 : SUCCESS in 53m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9226267fb711447aaf452e01552efe84 : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":3},{"id":"44c9df9c6f4478cb7cf02247f3f2994211583741","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-09 18:27:28.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"af60890a12551bb407529d803631ec1a78873bfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:49:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/cdaadb51c3cc4f12bf0c62d895183b9d : SUCCESS in 1h 42m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a8c41fb9beb47c4ac5c0f3e5d77807f : SUCCESS in 55m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c14757ebee2a4480bcbc9a7a99d9fd35 : SUCCESS in 1h 31m 55s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aad794a9db784d148a991d41a038d7fa : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1dbab9fcd8e54c7c841220766583ba1d : SUCCESS in 6m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0707ba54b15e46cb8ce03020cd063bae : SUCCESS in 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3367772e9b2f96c29426343d1993c8189ffdd017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 19:53:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4a70444b176640f48c16bd5cf99e74c8 : SUCCESS in 24m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a0e21caf83f84391b9237e10f5bafb99 : FAILURE in 45m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/50da50a638284b0b96f1dc6061a2af3a : SUCCESS in 1h 02m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1ad55d3d0f01412eb7e2501d623a2d72 : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/623f1bb96fef44359833ed2c37556681 : SUCCESS in 1h 10m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4fe97ebf253a45cfa7d8d1d6810a15fe : SUCCESS in 29m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/11769a6ce0974fac9ac3af5132f19dc8 : SUCCESS in 19m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a402dd0ac4f846d380f8aea09af9efc3 : SUCCESS in 58m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c97f6f4f26ae40ffbefb21d8c50ae211 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92d4f470b4dd4fbb8eabe5c028cf87c6 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dff73cd93658427e82025c4c15c2284f : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3dbf4800aee4c859672d030f41cc969 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08967454023d476a84127b6df2beec19 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/351b07f308e54227bbc3489623c43dd9 : SUCCESS in 4m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c6edd582ca484f05bb78920092be47ec : SUCCESS in 20m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f94b8c6fc5aa4b5f92684839bf44c468 : FAILURE in 45m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b936d2c0f11242458ea70ea08070a29f : SUCCESS in 49m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f1be26c663b4c6386eae40d85b45bbd : SUCCESS in 53m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bace41a7efcd4b0c82db7f065de72cee : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":3},{"id":"8bbe716787511b6ef9452b0ca77c78ede2127339","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 07:08:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7bc8ff938aec2a78e6cb44602e94fd6fb2e2e008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 08:23:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2fce47e6445545a18c567b5304e5bc6c : SUCCESS in 21m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/eb26179af6674212b02634a3463d95e8 : SUCCESS in 39m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4f8cebba4b6f4c468003c10a76b6c88f : SUCCESS in 1h 06m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e991fe9bb6bc4304b4837b9de5629081 : SUCCESS in 1h 04m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a782227cb09e44d9b8441c861fa8538b : SUCCESS in 1h 12m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9684f14222b446faa4b456855bc450b0 : FAILURE in 36m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4314d2dcc7414929af4c44e906aa6b5d : SUCCESS in 34m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5d724ca24d974b40bec5c69b67e82575 : SUCCESS in 57m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ada7b2ca5844b40acf1e3b52e9f1ada : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72c58a148cb64055aaefd9444b141929 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39d46bfb74014e4e8c29a107973925b5 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/197c70d60f144ce09af7d5dc9e91f96c : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7cce83211d4d23a096d29f03ea03e5 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4800630fde654cc5ab70620fcdd58dab : SUCCESS in 5m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c621572ef4041b79e1f99c0ac493fe3 : SUCCESS in 19m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/255d3d388cd24b08b25b83318a5b10a4 : SUCCESS in 27m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1bba963eb88e4386a1f5d4aee433bbc4 : SUCCESS in 1h 08m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/45694bd21f5e485681996b4c81578992 : SUCCESS in 50m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/486cebec4e1a411a89d24dd8474089d9 : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":3},{"id":"215d7ed42ed52ed7a709ad0a741736e8aa634685","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 08:23:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1baf137605b45bd788df3a8da6b9f54e46002a7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 08:59:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f7b00f34f4984588980252a6a5e96750 : SUCCESS in 1h 47m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/869d91947fd3439fa469b39724f2651d : FAILURE in 45m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/18461abca9d64dceb93602afe3d07dbb : SUCCESS in 1h 34m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/597da35cf15548c490d451ee738d81a4 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1b5b75e19a0489d8e247404e042cca2 : SUCCESS in 6m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6499a69acddf4548a905b2f151ad9bdc : SUCCESS in 1h 09m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6edfdbe83c6707eb5eeb6dc5a9b781cac701ae77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 09:52:31.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c363ef17272043718f34a85b0af70173 : SUCCESS in 38m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/aa401c7866564a949bee8c05a3823577 : SUCCESS in 1h 15m 55s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cda7896ccb0d40acb754ed776de6992f : SUCCESS in 1h 27m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283f900f4b954a3086057221123102f8 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1c5585f69ef4c908b938616ed07c6bd : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c76808fd27db43b29101f7eaf73bf697 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b75071716c914029904cca821db54df3 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d3bea7ba7a14861abe7c08a9b08a516 : SUCCESS in 4m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/70372e830095433e97a184b45106e6ba : FAILURE in 31m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0051046bc441410583eb9cf24520c959 : SUCCESS in 56m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9c0f8abf2dfc43eba9057749095aa8f8 : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":3},{"id":"59ddb59e87f17ecc7fe5a5d5b8721abd949407db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 11:06:33.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"becd3304c5869d7520edd307c4bfa18ed682488c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 11:06:43.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fcb25a16729c19547f9eead7f5a5364541388070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 12:25:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/834e60a50461443c8a32f503134377c3 : SUCCESS in 25m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/feff5581c6b54822a536aeaa24678b47 : SUCCESS in 40m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fe97eb011bff446da644b964c9c0e6d1 : SUCCESS in 1h 08m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/66aff3dcb7004fce80dc293cadf1b9f1 : SUCCESS in 1h 06m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a4d00c2f86e24ab4966955732342b16f : SUCCESS in 1h 17m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/718d5ec4f6cd42dbb655193ebc827b71 : SUCCESS in 34m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/74b85dc31cf640978a7dd5bbc54d1228 : SUCCESS in 18m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f04f21d2ec774a1a8f79674da1dd2c7e : SUCCESS in 57m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fcdd653fc31404c88c59a469a8a3562 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd9f6d0e60034c4bb6da3af0e5bfe670 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fa697f89a684637807536a71138e503 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af05cb0efd94449584f3397a5eab188c : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b578baca87474b2fb11901c09cc4b347 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/159a2dc4a58a491788604ad638544a55 : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/368fed684c084342a4b540eee73e071e : SUCCESS in 29m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/beb202bcb60c4a1aa394874c56af2106 : SUCCESS in 37m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d177e1fcf17b42518ff17e7012e1572c : SUCCESS in 56m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b1fb548b39f141c69ab5cf39666e7052 : SUCCESS in 52m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/27c15becab044931b84114ce46ea6190 : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":4},{"id":"232ce3e827e189ed883798a7af479c64a05c1789","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 12:25:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"31413ec259061d8651c1e6620dd2179aaf744d34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 13:37:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bb8a6b0dbad94a0e9dfb39addb46280d : SUCCESS in 2h 29m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b4e5e80a04a4aac8ad9fe1bbdee8947 : SUCCESS in 48m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eeecd861d41e42bfa25b345be8793d43 : SUCCESS in 1h 33m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/742ef6200e8548af9a4b8045e2172c6a : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/500bc5480a9845048ad4aa0861b8aabc : SUCCESS in 5m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/708f5c1cee9f48d5a916cb78b76dc622 : SUCCESS in 1h 44m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fe4f3626223c456d1c73b293704568c160f34f8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 13:41:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0f48ceb3207e420598b95ffbaca68074 : SUCCESS in 38m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/505c224d338340b6aa4631d8983edb7a : SUCCESS in 1h 07m 51s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/84a62c9f974a406fb4d1ee69d351499f : SUCCESS in 1h 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07aced3e487f43b087ae187da32a2123 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8bce9f1f4ca4e14880d5595ff014e79 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a63a7dde52b4de1840495d257b218fa : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b9d95cd05148038df33a2bc1afbf98 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4bca1fb5266493699603b73b466a32e : SUCCESS in 4m 01s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ef6ed95f05d47fb97f7667d77e0dcce : SUCCESS in 34m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/798e53dfb2994a8ea90c58ad52bfd62a : SUCCESS in 54m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aa984372cf234b73b58e2ee7caa9ae31 : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":4},{"id":"6660b326d148c80815d4e3195950aa7f680e5934","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 13:41:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"30b7adac9834e5cd495b020175eaffa98e6441e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 13:43:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4862d8b2b7c54079905a97dd83138dc4 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/638837aeac064d5283e75947c15f37e6 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d7252ba5829f005915ce8e7efc9ec252c06d8bc0","revisions":{"96d1b18d314e8b8f956fe284a53bebc6b35902a5":{"kind":"REWORK","_number":1,"created":"2021-06-07 11:58:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/795089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/795089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/795089/1"}}},"commit":{"parents":[{"commit":"53434625f64196529502bbeef618620b6267b0e0","subject":"Merge \"Sort the package lists in openstack-base alphabetically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53434625f64196529502bbeef618620b6267b0e0"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-07 11:56:48.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-07 11:57:57.000000000","tz":120},"subject":"Change network_mode default to host","message":"Change network_mode default to host\n\nKolla-Ansible deploys Docker without bridge by default [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla-ansible/+/751795\n\nChange-Id: I7430b72f5d8abe59a7151df73e6d59898d942636\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d1b18d314e8b8f956fe284a53bebc6b35902a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d1b18d314e8b8f956fe284a53bebc6b35902a5"}]},"branch":"refs/heads/master"},"29edb05eb9fced9a10999616aa03df79469a9379":{"kind":"REWORK","_number":2,"created":"2021-06-07 12:18:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/795089/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/795089/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/795089/2"}}},"commit":{"parents":[{"commit":"53434625f64196529502bbeef618620b6267b0e0","subject":"Merge \"Sort the package lists in openstack-base alphabetically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53434625f64196529502bbeef618620b6267b0e0"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-07 11:56:48.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-07 12:18:10.000000000","tz":120},"subject":"Change network_mode default to host","message":"Change network_mode default to host\n\nKolla-Ansible deploys Docker without bridge by default [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla-ansible/+/751795\n\nChange-Id: I7430b72f5d8abe59a7151df73e6d59898d942636\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29edb05eb9fced9a10999616aa03df79469a9379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29edb05eb9fced9a10999616aa03df79469a9379"}]},"branch":"refs/heads/master"},"e2c56a1854d9574a59f1d0837be7917a1431066d":{"kind":"REWORK","_number":3,"created":"2021-06-07 12:39:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/795089/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/795089/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/795089/3"}}},"commit":{"parents":[{"commit":"53434625f64196529502bbeef618620b6267b0e0","subject":"Merge \"Sort the package lists in openstack-base alphabetically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53434625f64196529502bbeef618620b6267b0e0"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-07 11:56:48.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-07 12:39:31.000000000","tz":120},"subject":"Change network_mode default to host","message":"Change network_mode default to host\n\nKolla-Ansible deploys Docker without bridge by default [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla-ansible/+/751795\n\nChange-Id: I7430b72f5d8abe59a7151df73e6d59898d942636\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2c56a1854d9574a59f1d0837be7917a1431066d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2c56a1854d9574a59f1d0837be7917a1431066d"}]},"branch":"refs/heads/master"},"d7252ba5829f005915ce8e7efc9ec252c06d8bc0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-10 11:06:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/89/795089/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/795089/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/795089/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/795089/4"}}},"commit":{"parents":[{"commit":"a6884085e1a5336f17c2ed64f67d8f885c454207","subject":"[CI] Add grafana to allowed-to-fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6884085e1a5336f17c2ed64f67d8f885c454207"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-07 11:56:48.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 11:06:33.000000000","tz":0},"subject":"Change network_mode default to host","message":"Change network_mode default to host\n\nKolla-Ansible deploys Docker without bridge by default [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla-ansible/+/751795\n\nChange-Id: I7430b72f5d8abe59a7151df73e6d59898d942636\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7252ba5829f005915ce8e7efc9ec252c06d8bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7252ba5829f005915ce8e7efc9ec252c06d8bc0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
