)]}'
{"id":"openstack%2Fkolla~795631","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I7f79aeab53540df9bf84add6df3add300c9e9fa4","project":"openstack/kolla","branch":"stable/wallaby","topic":"allow-to-fail-grafana-stable/wallaby","hashtags":[],"change_id":"I7f79aeab53540df9bf84add6df3add300c9e9fa4","subject":"[CI] Add grafana to allowed-to-fail","status":"MERGED","created":"2021-06-10 12:58:51.000000000","updated":"2021-06-10 15:36:33.000000000","submitted":"2021-06-10 15:35:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795631-allow-to-fail-grafana-stable/wallaby","cherry_pick_of_change":795757,"cherry_pick_of_patch_set":1,"meta_rev_id":"2d73d2758836d2830ff897676f1af4957a27b757","_number":795631,"virtual_id_number":795631,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-10 15:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2021-06-10 14:01:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-06-10 14:01:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 14:17:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea6f165e308a086dd4fa4f862684b39ae99eb287","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 12:58:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d1e09ebd83a34a27df4204e0751704b7eb52a0f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 14:01:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4ea29cb9ef6d64d8ab71382934ac1fdb2b4b493c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 14:17:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7013ccf7e1ce4fd19ecc06e0d01da1f5 : SUCCESS in 23m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c891f32797f249318eeffc75e035d8ab : SUCCESS in 42m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d4e1cec051a8488786aea52d9c5e6e97 : SUCCESS in 1h 03m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/27c92a8a93fa47da8de8f0b58f370a92 : SUCCESS in 1h 08m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b00f35ed39014bea9155c749cfbbf06f : SUCCESS in 1h 14m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1e0693c474414dc69ffc93b45dd22921 : SUCCESS in 34m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5af6b2361fae4d3891678b3f0bcdfa90 : SUCCESS in 19m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d4deff9fecc24946b837d53e4edc3d0d : SUCCESS in 57m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c065ea257b5046cbbe1ad1b5888f4afe : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab5652c72f5e4806a52e674a7a52a360 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/258f1d3dd05c4a8e92d24e1c43d4f45c : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7352357b2b6540bd940f7c5904904c2d : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58aedc5140eb42dd8807fcef70ddfd45 : SUCCESS in 6m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d54c4145bcfe4870b3d6525693a18068 : SUCCESS in 22m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/99229572c64642798e265b8dba693bdf : SUCCESS in 39m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/50907dfbb218433f9b6046a3a6457b5c : SUCCESS in 56m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f5de798caf56472abb620ef8832cfdda : SUCCESS in 53m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ecf2e22032347379b80ab76f183e39e : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":1},{"id":"09b8789ab5f93a954eb16178d3bbc99d09653e0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 14:17:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2dd4dc114051467afe0e3dd7fe15c4a5a75355cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 14:44:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b038e11c61e1486aaad14a750ab26fff : SUCCESS in 55m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/124d8aa4ffbd47b581af5069fc8ea1eb : SUCCESS in 43m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6e03f0497ab74477b0275bc5e2b29bcc : SUCCESS in 1h 41m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2fbdf73cb5e540af89ebf507c773e60e : SUCCESS in 1h 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94fe8d207e3968eff9a93a5ba68f519a99e9d43f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:34:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ace0c76a9fc3474aa5a6326c9636d5e7 : SUCCESS in 30m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/024f9abee9c7447da1c83a96051765c4 : SUCCESS in 1h 16m 12s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e998262cca0b45d78f9ca284eaa8b1b1 : SUCCESS in 1h 09m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8708991ab624064a77152f8e071c4a7 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f378f138ff7b4b7a823febab1ffc3441 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6d6f6391ac14b5a849716f82203c01b : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/143543082edf41c9abc1961740b2b2ed : SUCCESS in 7m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/66478a85f7f347dcab7fd9f036502ce0 : SUCCESS in 40m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f9daad6874e4af5b1aabd023fed758a : SUCCESS in 58m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5a9521e1caae4c10b3c32c9000de1c84 : SUCCESS in 1h 05m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7ddb543b6becdaeae775354aa7bef8c119d9fd1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:35:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2d73d2758836d2830ff897676f1af4957a27b757","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:36:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/434a79b7121c4ba3afca734d6f9b99ce : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e38030b0cb01f41db4e41ab2d3da5198b0c2fb5","revisions":{"5e38030b0cb01f41db4e41ab2d3da5198b0c2fb5":{"kind":"REWORK","_number":1,"created":"2021-06-10 12:58:51.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/31/795631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/795631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/795631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/795631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/795631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/795631/1"}}},"commit":{"parents":[{"commit":"51802de9614f7c8cf3d8b40b19654b0065fc1a0a","subject":"Add missing default policy files for ubuntu-binary-horizon","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51802de9614f7c8cf3d8b40b19654b0065fc1a0a"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 10:07:57.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 12:58:51.000000000","tz":0},"subject":"[CI] Add grafana to allowed-to-fail","message":"[CI] Add grafana to allowed-to-fail\n\nThe grafana repo is broken for Debuntu.\n\nChange-Id: I7f79aeab53540df9bf84add6df3add300c9e9fa4\n(cherry picked from commit a6884085e1a5336f17c2ed64f67d8f885c454207)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e38030b0cb01f41db4e41ab2d3da5198b0c2fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e38030b0cb01f41db4e41ab2d3da5198b0c2fb5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
