)]}'
{"id":"openstack%2Fkolla~795741","triplet_id":"openstack%2Fkolla~master~I8460a4159c8da8ae0b856867e068d30fbffb4d34","project":"openstack/kolla","branch":"master","topic":"bug/1931544","hashtags":[],"change_id":"I8460a4159c8da8ae0b856867e068d30fbffb4d34","subject":"Enable Debian updates repo","status":"MERGED","created":"2021-06-10 08:25:39.000000000","updated":"2021-06-10 19:11:06.000000000","submitted":"2021-06-10 19:09:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795741-bug/1931544","meta_rev_id":"427e79f024b8cefcf407682b855f7ed44b516901","_number":795741,"virtual_id_number":795741,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-10 19:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-10 19:11:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-10 17:44:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-06-10 17:44:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-06-10 19:11:06.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 08:36:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-06-10 09:48:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-10 11:07:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"41d07b481e6979d000b5a8dcaa307ecd4fa3a974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 08:25:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96e82dc21b238364963e589524cdbadd33aebaa4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 08:34:15.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"3c0290ed3d38a249503d45398766b4edbcff578b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-10 08:36:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce4e3a94ccc1070021655a64af13417af3bc6298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 09:48:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fd3644893d4c491e988223a5e55bbcec : SUCCESS in 23m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fe9d5692b53d4e8baf570fd9929fdd13 : SUCCESS in 39m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b579194c194840a796b986caed353baa : SUCCESS in 1h 04m 02s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4ebe3c2ae443434c8e73a797bc0629d6 : SUCCESS in 1h 10m 44s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/832b10f762184f78b42d0786735e39b7 : SUCCESS in 1h 21m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/862f2c4b8e2a4d37a1c532786c98af09 : FAILURE in 36m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/77b9ec7127a64762ac099b010795fc0a : FAILURE in 18m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d5943681a620448ebb861e3f3e989724 : SUCCESS in 59m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0916d472b584d30a69399d72efa15d2 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bf589171b5e433dbab4f8b431dc2679 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df5a2a0bc7c34f5a9a792bcc03e15787 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e549f08c8b624d25833a67915a3c61d8 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cab2e6263354167aba38f30ff902d21 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/516066e88d3b4e4e9f494d82f5ff23b9 : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b1dfbc71a854f888e00a6bc79d982c3 : FAILURE in 21m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dcbd6742b3204787a6b03cf7809efea5 : FAILURE in 36m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4055b0f8f841457283ccf2e68ab28ea3 : SUCCESS in 57m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/818261801a9a4cde9aecb3af94225d17 : SUCCESS in 54m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e209b6c71c447d68939015d5089a8d7 : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":1},{"id":"57ca55e855f8914cf2b172bce445674f055060e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 10:50:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/cdd436b7cbae4f28bbe515d16d8ef611 : SUCCESS in 1h 56m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/829238db48344671a29d88ba0a34f2f1 : FAILURE in 1h 48m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/48a2cdfdb0a54f1f923749bb42d5a38f : SUCCESS in 2h 19m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02e25f628006497891bdff7dee4385db : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/042c6ff0261346fcb1ba639032d03156 : SUCCESS in 7m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7d24d47962241759d4b264eb38bb324 : FAILURE in 1h 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"510a7a437eb45af3ab84b45fcbaf81ac1d182d12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 10:56:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c8f019001a1d42ff3003e3b97dbbc589a71f82f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 12:17:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/aea73597071544f68b87dc670d3dac8a : SUCCESS in 38m 21s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/67b2cdf604b742f7b7819a05b2f2a996 : SUCCESS in 45m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/09d9c5dbb8a9467a94646ee3706da675 : SUCCESS in 1h 03m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f8df23e64bfe4add9dd8fd3c9b963329 : SUCCESS in 1h 08m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/33a7df80417440e8b2008bf94d7af2a0 : SUCCESS in 1h 09m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a6a792350e2044a8b1e834a6fe57aab6 : SUCCESS in 33m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4bf1b962d5a14daa8e72f13cea403ecc : SUCCESS in 17m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e52496f9aa9454481d645ef5f3a559a : SUCCESS in 1h 02m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce08929dc9b34d9c99fe2fc7d2669702 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b1eba215ff3464ebed5c69032483689 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c4ba910bb2c464092ab4e67c45d855a : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc6408d7f57f4124bdf69a2bb231a520 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/645154d75b264f2c8dcf1543b1aa3bc8 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/388a636a18b54737b398a8ba5152697f : SUCCESS in 4m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9c926c6026ba47ea8be86dc60107a2f3 : SUCCESS in 20m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d594a151c78e45528c5ffeba7ac8b639 : SUCCESS in 33m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5667b5a730d244219b9a732db68d5998 : SUCCESS in 47m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/81179ebda6c14dc6b083cdf0d930706f : SUCCESS in 58m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/29005758d47b4feba167c2e7472b0fe9 : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":2},{"id":"b5c1d09324e2dfdbadb6641ca4b2ba5fec8d9cc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 13:21:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f50e98549bb4facaa3cc321cd47d7a6 : SUCCESS in 2h 22m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aff8fb6065664e43bc66be41126206fe : SUCCESS in 52m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8497ecc404c42ce8b7f7584fb62bb31 : SUCCESS in 1h 30m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f530af6fbad4464b6e4a2c2b56154e3 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3af7e998cd754888b1361b3e4cc17793 : SUCCESS in 6m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c6e5deef6f14659b3f249039753ef58 : SUCCESS in 1h 56m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5718f028909eb09568da7acc027b1d89226a8280","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-06-10 17:44:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d2360d849357855ff300131228fe5a1ccec5b123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 17:45:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8f7fc4c30775ed72f47066e0117279175ea2a9ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 19:09:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d67f02dc74124083987c163f78cb97ec : SUCCESS in 34m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/799646679b8848e4b04e7768328acba5 : SUCCESS in 1h 05m 54s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c92a17db3c4b4d299be53b560983084f : SUCCESS in 1h 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d65109408174979886111d3df6035a7 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c6f15e43b994c878184b677e6cd8020 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b63522d093c48779a13987ccbf86ad3 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91533c9cabe44ab28a29a09ff91da23b : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d9101aa58d54fcfa7466ac7f50ee9ab : SUCCESS in 4m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbc68a12abd448968b87676fd819bae9 : SUCCESS in 35m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ed2108492fa4b0fa86115eba18ea430 : SUCCESS in 50m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0be0ef8a258845b98ee2ae1beb505657 : SUCCESS in 1h 04m 43s","accounts_in_message":[],"_revision_number":2},{"id":"1e61db8d14f38de4a179f63034ed14595f4e7a25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 19:09:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"831edfdcd9cef21c7b48003341843e35ceb7be6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 19:11:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ee3b4b8a3b461b84ca7d9c7a346047 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f67642c891184df9a2f5cd3552569c04 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"789c9c3b84e84796ac8211112fa29f415f560c88","revisions":{"d48d763ad01fafd0672e7618f6e88c42067c8d5d":{"kind":"REWORK","_number":1,"created":"2021-06-10 08:25:39.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/41/795741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/795741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/795741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/795741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/795741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/795741/1"}}},"commit":{"parents":[{"commit":"1c5d98ecab0c1f3a4643698808a78ae449fe1f14","subject":"Merge \"Fix _get_images_dir when pip-installed with --user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c5d98ecab0c1f3a4643698808a78ae449fe1f14"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 08:24:30.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 08:24:30.000000000","tz":0},"subject":"Enable Debian updates repo","message":"Enable Debian updates repo\n\nBackport below Wallaby switching bullseye to buster.\n\nCloses-Bug: #1931544\nChange-Id: I8460a4159c8da8ae0b856867e068d30fbffb4d34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d48d763ad01fafd0672e7618f6e88c42067c8d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d48d763ad01fafd0672e7618f6e88c42067c8d5d"}]},"branch":"refs/heads/master"},"789c9c3b84e84796ac8211112fa29f415f560c88":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-10 10:56:30.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/41/795741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/795741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/795741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/795741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/795741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/795741/2"}}},"commit":{"parents":[{"commit":"a6884085e1a5336f17c2ed64f67d8f885c454207","subject":"[CI] Add grafana to allowed-to-fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a6884085e1a5336f17c2ed64f67d8f885c454207"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 08:24:30.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 10:56:30.000000000","tz":0},"subject":"Enable Debian updates repo","message":"Enable Debian updates repo\n\nBackport below Wallaby switching bullseye to buster.\n\nCloses-Bug: #1931544\nChange-Id: I8460a4159c8da8ae0b856867e068d30fbffb4d34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/789c9c3b84e84796ac8211112fa29f415f560c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/789c9c3b84e84796ac8211112fa29f415f560c88"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
