)]}'
{"id":"openstack%2Fkolla~795747","triplet_id":"openstack%2Fkolla~master~Ic1e3be1b00f22eb2c12b4b67b80567466c60f329","project":"openstack/kolla","branch":"master","topic":"kolla-debian-bifrost","hashtags":[],"change_id":"Ic1e3be1b00f22eb2c12b4b67b80567466c60f329","subject":"[WIP] Build bifrost on debian","status":"ABANDONED","created":"2021-06-10 09:07:21.000000000","updated":"2021-11-04 11:30:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"eab354d331270e20577a6bad6358131359d1b4bf","_number":795747,"virtual_id_number":795747,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-10 10:29:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":-1,"date":"2021-06-10 11:08:52.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2021-06-10 10:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b60c641282e08f4557e327e82bcf1b3571f39339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 09:07:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c247bda9a89d60c273e4e6fd634d8661131a041d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 10:29:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/292e77c4544d4d90b595c8e17770db35 : SUCCESS in 29m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3f2d43f80ae042d78ff471fda3b95f14 : SUCCESS in 37m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/722244d775a24a1f8e1914fae753533b : SUCCESS in 1h 10m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/34e6e154ee4041d5a7fa4ac01cee314c : SUCCESS in 1h 02m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/478dee49c5c24269bc2fdc342c33cfda : SUCCESS in 1h 14m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ef844b2dd3624a9f8d9b305127e70cc3 : FAILURE in 44m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dadf3c9fc7154715a6e84b59ddadaacd : FAILURE in 17m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0b695cb472944e60baf917094ae46781 : SUCCESS in 59m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48d79da9f00b48e6b78adcace770f9c2 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0433290353045d18e1f08a70b9e64ac : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4674540b96d145fea955d457a9f04dda : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34c28396d94846c6b91cbb3a1758a025 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc762cf24bc45a0afd928813699fcd2 : SUCCESS in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b326fe5dedca459f810570ff8f3b95c6 : FAILURE in 22m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/29486a6cacb94645a5c976889948e32e : FAILURE in 30m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c166c00531804530a714bd8ff09ccaf3 : SUCCESS in 1h 02m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9dcc4241e0604847a279f4e1fa5fdb28 : SUCCESS in 55m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0cfbee67b6ef4415bb378adc15e08643 : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3ab0b79f6f59c74521cf28a35b20e87a5a6b8e4a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 11:08:52.000000000","message":"Patch Set 1: Workflow-1\n\nok, bullseye not yet supported in bifrost upstream","accounts_in_message":[],"_revision_number":1},{"id":"a0adac72a14db7a02e0ce4d19231a2ad54a780f8","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 11:21:22.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"736613f13983803b1259f66eedc7779eaa7fb9e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 11:26:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf87635d18f546ac83e915ddfb4d41ba : SUCCESS in 2h 13m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1af6a6c755ce43ce8763364e15c69b3d : SUCCESS in 1h 16m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/647f81280f8646139d80b0a425fca156 : SUCCESS in 1h 51m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49c85ae68e6641b0bdc9b83fe202574d : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c8ee61dbf684aff9b8a68ba2fb8c160 : SUCCESS in 7m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/993b8120e4784b14908607d5b949584e : SUCCESS in 1h 52m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eab354d331270e20577a6bad6358131359d1b4bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-11-04 11:30:53.000000000","message":"Abandoned\n\nmore complete variant: https://review.opendev.org/c/openstack/kolla/+/814783","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0901355d47e87ce2832e3113acd372a6ae22c9fd","revisions":{"0901355d47e87ce2832e3113acd372a6ae22c9fd":{"kind":"REWORK","_number":1,"created":"2021-06-10 09:07:21.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/795747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/795747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/795747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/795747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/795747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/795747/1"}}},"commit":{"parents":[{"commit":"1c5d98ecab0c1f3a4643698808a78ae449fe1f14","subject":"Merge \"Fix _get_images_dir when pip-installed with --user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c5d98ecab0c1f3a4643698808a78ae449fe1f14"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 09:07:05.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 09:07:05.000000000","tz":0},"subject":"[WIP] Build bifrost on debian","message":"[WIP] Build bifrost on debian\n\nChange-Id: Ic1e3be1b00f22eb2c12b4b67b80567466c60f329\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0901355d47e87ce2832e3113acd372a6ae22c9fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0901355d47e87ce2832e3113acd372a6ae22c9fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
