)]}'
{"id":"openstack%2Fkolla~795795","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I7430b72f5d8abe59a7151df73e6d59898d942636","project":"openstack/kolla","branch":"stable/wallaby","hashtags":[],"change_id":"I7430b72f5d8abe59a7151df73e6d59898d942636","subject":"Change network_mode default to host","status":"MERGED","created":"2021-06-10 13:51:13.000000000","updated":"2021-06-10 16:27:13.000000000","submitted":"2021-06-10 16:25:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795795","cherry_pick_of_change":795089,"cherry_pick_of_patch_set":4,"meta_rev_id":"189f7a6aeaafa3e14c9d59b9d96fc33c2f2d6eb7","_number":795795,"virtual_id_number":795795,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-10 16:25:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-10 13:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-10 14:01:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-06-10 14:01:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 13:51:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-06-10 15:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fad8859cdc92667081520c1a3045a24da3b2c98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 13:51:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"82b4fea7a002373f287f919376e79554f505ff1f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-10 13:56:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"59a0127024cb5e2c33fe1ac50b41948b6bcc6ce5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 14:01:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"33c9f939b540da60c8c0cdaf3bea44e9b3112858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/60c39339922e4d21b432d01c8a48f950 : SUCCESS in 43m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b60e218efeed48f791b5dbd5ee09fb45 : SUCCESS in 41m 38s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8b7f6a69e6e7473ab9575bbc8a388ad9 : SUCCESS in 1h 02m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7f0d6c3e666d46a38cf69b68f5cd42e9 : SUCCESS in 1h 03m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/53c5808444214f4aaeb20de66be98638 : SUCCESS in 1h 13m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7e9cf794c6654467b74eefb5979c9eb1 : SUCCESS in 33m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2f666672b7bc4649a7c2fab39513be15 : SUCCESS in 22m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f0a931ac709142c9a73d41d978526718 : SUCCESS in 56m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/462d24336fa94d45b29d30ba0fab1e01 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fcfae78f2f0497b9ac36a48118ad535 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c39c61c453f4f9194f42351aef75c9f : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ef67259f61e4df2893d8bd49bcbb63c : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00626cdd61404d68ab33c95b38c89a7a : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a32ce54cb124e5c8a72ad224588a082 : SUCCESS in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5df3ff10cfc14a50abd46e74543db26a : SUCCESS in 34m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aed9187004de4733ac97a4421a590c38 : SUCCESS in 41m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9ded2f9216b4aed9e4422c8142b5b61 : SUCCESS in 51m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aa179e17b9314602ac721d0d2d81168f : SUCCESS in 51m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/94161a85a5694babae0efb488389e2d0 : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":1},{"id":"98017f0bd1ffc72f92dfb87b93896f34e6a80299","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:10:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2051b8825e1f2778d97e8469b1c0089b701fe5a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 16:09:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/55d5afaf50ba44ec9c4fd972658138f7 : SUCCESS in 2h 16m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/94720ef4b0eb4dc1b2589eabd639aa5d : SUCCESS in 54m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d5351f01e3f14ecf8c19245b33778104 : SUCCESS in 1h 40m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5feff86294e54512ad0454850c4b029c : SUCCESS in 2h 03m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90528750c935719e6943926131edb764c2fc7e60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 16:25:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9504a83ea00e48dd81ec1a55d292d288 : SUCCESS in 36m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7f679bd6a8184323a7ad44726341ef20 : SUCCESS in 1h 00m 58s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/33acab8870c94f80a940c5fb73bc5e53 : SUCCESS in 1h 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d511b5401a05410abb0d47f87ccb6a89 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38c3b9eda444492a833a8ee32f440634 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfa147c617af40849092d0a7f1f9a2cd : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d34aca05768041b784d21a49116749f7 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d91b57ca0e35483d954b848784587399 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba0729c62f1b4474bd5cd30e2b938b46 : SUCCESS in 49m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6541fdba92b143d183649dcf1649a0e7 : SUCCESS in 50m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/da43ae06010e420483ca78c7adc03ac8 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":1},{"id":"24a553a32cf384dc2e1bda5accbafbe0414f12d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 16:25:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"189f7a6aeaafa3e14c9d59b9d96fc33c2f2d6eb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 16:27:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6cd9881eaf8466ca025510d617176e7 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b34826294ba4ae692479592860dbaed : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9f2f12cc9ab4f3e49f1f3f453f2fc2946cde441","revisions":{"f9f2f12cc9ab4f3e49f1f3f453f2fc2946cde441":{"kind":"REWORK","_number":1,"created":"2021-06-10 13:51:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/795795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/795795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/795795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/795795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/795795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/795795/1"}}},"commit":{"parents":[{"commit":"5e38030b0cb01f41db4e41ab2d3da5198b0c2fb5","subject":"[CI] Add grafana to allowed-to-fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e38030b0cb01f41db4e41ab2d3da5198b0c2fb5"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-07 11:56:48.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 13:51:13.000000000","tz":0},"subject":"Change network_mode default to host","message":"Change network_mode default to host\n\nKolla-Ansible deploys Docker without bridge by default [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla-ansible/+/751795\n\nChange-Id: I7430b72f5d8abe59a7151df73e6d59898d942636\n(cherry picked from commit d7252ba5829f005915ce8e7efc9ec252c06d8bc0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9f2f12cc9ab4f3e49f1f3f453f2fc2946cde441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9f2f12cc9ab4f3e49f1f3f453f2fc2946cde441"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
