)]}'
{"id":"openstack%2Fkolla~795811","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I8460a4159c8da8ae0b856867e068d30fbffb4d34","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1931544-stable/wallaby","hashtags":[],"change_id":"I8460a4159c8da8ae0b856867e068d30fbffb4d34","subject":"Enable Debian updates repo","status":"MERGED","created":"2021-06-10 19:17:48.000000000","updated":"2021-06-11 09:53:22.000000000","submitted":"2021-06-11 09:51:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795811-bug/1931544-stable/wallaby","cherry_pick_of_change":795741,"cherry_pick_of_patch_set":2,"meta_rev_id":"11bcb5e3fb50a10eca1137738298e97c3f368cb8","_number":795811,"virtual_id_number":795811,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-11 09:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-11 08:32:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-10 19:27:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-11 08:32:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 20:42:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-11 08:32:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"8d30509436c1ef12862edd7990c66ef2d464d06f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 19:17:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"304d25e2ee2c0959b0321a53d5dd0e4918672b77","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-10 19:27:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b417d3e323806a45b195add7f6bdd950507f81fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 20:42:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9b1421cc45874bebb2a30e8e8a42ff03 : SUCCESS in 24m 14s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/06ab9d6c5825414b9d803ff3b0567f3a : SUCCESS in 33m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4af8a08f8dc345b6a278022592d5de86 : SUCCESS in 1h 15m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9fcd129ca4f44806b7ac5f7fb3bb64ea : SUCCESS in 1h 06m 59s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c404772cdb0e425a8f7259323dd21be4 : SUCCESS in 1h 19m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ba713deb2da14fa089c0d1341d91ab8e : SUCCESS in 35m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/030538e55ce342d59bbd0f242df20426 : SUCCESS in 20m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a1e6902a40c74aada8c7ef800877f42e : SUCCESS in 1h 10m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0350a1f032c4670b4d1426aec408b8a : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d392bd1f0fbb465bbf06c2e7efedca9e : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a333d88f63264629b273f3cb69ac7061 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b49e6d3901cb42d0ad53ad2abfeee950 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c505f0ea94ea4365a7088e94301c295c : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d597c3c913e472c986b4f4984783b8f : SUCCESS in 4m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dc3e37bd59f6449e928aa7b3a7e36590 : SUCCESS in 19m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f63d202c81604819b7af642a34c0624d : SUCCESS in 35m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3488961ab5442dd9854f413d40e00b0 : SUCCESS in 52m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2b71e02755ff4a758b4b99285f9a4e53 : SUCCESS in 56m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e64627af539b4b529782040480254e3c : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a293d2f37ab038be2e0a2f48541f54f5ec8deb97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 21:32:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed331d192bbd4f64a606f0216c17e70f : SUCCESS in 1h 42m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/61721c5bf02346eeb8c7abeee40bb5c3 : SUCCESS in 1h 26m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6130514ce8a4666baf8f9a76f7cd9f7 : SUCCESS in 2h 10m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/24af42e4dca84a10bf68ca17083d2937 : SUCCESS in 1h 20m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75abee3f26fb0a3b29e1d00365ad177963791f00","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-11 08:32:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"857d3625774163872103d22fd5317a5bd299583b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 08:33:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2b1cbf59d0df6a56b4b224f725bfc2dab73d9ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 09:51:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/24d4e232923c489aad3ea5f67aeeb0f7 : SUCCESS in 34m 19s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f2a67de921bb433b947c27fbe6c34984 : SUCCESS in 1h 10m 05s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/126ab417f18c4c8aa1eda97a7cfac2c9 : SUCCESS in 1h 16m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a927e059a54ccaa99a61282674cb1f : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/034fc4fccd644260a23d527305aafbd0 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b427def217f44f418da38961ac904413 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54694ef1f39e4a42be2abe46601c9cbf : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f23db0035c04f90b6007a8b98a52c62 : SUCCESS in 5m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/284e033343f944a6b0fc0c89a015237a : SUCCESS in 36m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/24bf4dc0536d433c88109822ece23b71 : SUCCESS in 55m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a55aca3da934f52b11d1ecdd8e6fd39 : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ca3a33223824c7f8437ba1d5e352b4bbccf2b066","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 09:51:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"11bcb5e3fb50a10eca1137738298e97c3f368cb8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 09:53:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c69b12c8eb6b469ea7c497d45c2b2faa : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4be96af54be6450c9dd6978389de9066 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"746ae194eeaaea7b8ef30311acb424cb95b21da3","revisions":{"746ae194eeaaea7b8ef30311acb424cb95b21da3":{"kind":"REWORK","_number":1,"created":"2021-06-10 19:17:48.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/11/795811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/795811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/795811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/795811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/795811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/795811/1"}}},"commit":{"parents":[{"commit":"f9f2f12cc9ab4f3e49f1f3f453f2fc2946cde441","subject":"Change network_mode default to host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9f2f12cc9ab4f3e49f1f3f453f2fc2946cde441"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 08:24:30.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-10 19:17:48.000000000","tz":0},"subject":"Enable Debian updates repo","message":"Enable Debian updates repo\n\nBackport below Wallaby switching bullseye to buster.\n\nCloses-Bug: #1931544\nChange-Id: I8460a4159c8da8ae0b856867e068d30fbffb4d34\n(cherry picked from commit 789c9c3b84e84796ac8211112fa29f415f560c88)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/746ae194eeaaea7b8ef30311acb424cb95b21da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/746ae194eeaaea7b8ef30311acb424cb95b21da3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
